)]}'
{"id":"openstack%2Fopenstacksdk~1006191","triplet_id":"openstack%2Fopenstacksdk~master~Ibb90c3bfc82fc7ed1c3208cb5de017081ec42e2b","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Ibb90c3bfc82fc7ed1c3208cb5de017081ec42e2b","subject":"Bump up deprecation timeline for CloudConfig class","status":"MERGED","created":"2026-09-18 11:17:15.000000000","updated":"2026-09-21 13:23:56.000000000","submitted":"2026-09-21 13:22:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006191","meta_rev_id":"d816bedb25227a5a991ae6ccee962e363a45ce8c","_number":1006191,"virtual_id_number":1006191,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 13:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 13:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d423ba14a6afefd6d6df7a7ff7143be2556a645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 11:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73f96f69abc1fe2f2fb94e3c719673044b7cec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:17:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9f5edf62f847f699e0478f1944455f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffaa0d86b4ae45fa8f076e770e12b0e5 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e31613d4a20d419ca16082b1c5820a05 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16875b554e7f4b82a4c445e86f96dec5 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d4d1078fd31431187d5d1c99c240717 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c03b1c475a94cd9a17ac6b20b2bd631 : SUCCESS in 7m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/05fe39196cdb41a8b8757965ce5934ef : SUCCESS in 4m 55s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/13f77f178ec24fa3b64e84c71a5e9836 : SUCCESS in 55m 23s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6caaf454c11e4c1382b2721dbb9d0cf3 : SUCCESS in 7m 09s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/101682970c8c4341b9c0dcd780b079c9 : SUCCESS in 4m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/55359f74eceb4ce899ca3f23370e8478 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8e17bb15ec4993b77edf97b6ed91a0 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c2577c0e95f4a23bd4d943ffb39b108 : SUCCESS in 5m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/380766247f7845329d40fa6b5202bc6b : SUCCESS in 49m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/13ec4a502f3f48648018ccb34dd919d5 : SUCCESS in 1h 24m 41s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/502a557150aa4e48990a8b65294508f9 : FAILURE in 1h 37m 01s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/3855a2d01b2b49ba996213068cdd0a64 : SUCCESS in 20m 00s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b9d44ba00f9648d58d11cbc38398344c : SUCCESS in 36m 26s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/7ffcde51ed42451eb12d7974bc98019e : SUCCESS in 30m 36s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/a2dbafe85394410aa2b49ba12c5369b3 : SUCCESS in 32m 15s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/698754931e6a4a12860aa22d2ca2372d : SUCCESS in 31m 49s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d73799dcd12b4d70bf839d1ebb86d715 : SUCCESS in 48m 22s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6b5dbaec64bd4caab997c924714e6f95 : SUCCESS in 1h 18m 46s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6f869d65c14241b0b5fd5a297d7816dc : SUCCESS in 1h 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53ddace00d4e9cde03c82ed6ebc5a8f820f17540","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa4b7915899ce212ebc5b2fa02274c2429baf516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"646e698e5b6014ea17e09846d49b5fd7e645b1b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:22:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb92cc0af6c48f4bceedf51c52aea4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9384e56e206845cdb113bb13ef4a0222 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04409a7d63c3400a9b57dae253da3c3d : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f453787c530f4fb68d178486266b547b : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9dbc672bc0542e0bc31ae7ef59447f8 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5f7dced8a34a4900bb68819ab252288b : SUCCESS in 55m 03s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ac12d7654a25493a96b8610da39de5c0 : SUCCESS in 4m 35s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e6aeeb16759448e09433be43892b7ab0 : SUCCESS in 5m 26s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b88c062c5935430f8ffa86aaf0e8e5e9 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa7253a10984a42a43486e1f2b565b0 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93c2f6600174b108ae13f6ec4fe0745 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f61b8159de49419597fdcc6a1618fd65 : SUCCESS in 1h 02m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/861af5cebd4945438108321931d0b109 : SUCCESS in 1h 29m 05s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/acb574dffd73461caf7a35426c44efec : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d94d6d24d8947b32394cde99e2380b6656b6c457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:22:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d816bedb25227a5a991ae6ccee962e363a45ce8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:23:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dbb8dc53b014fb69ccb900fff3d8f4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21217d5cf2bf4d40a080539860ba394b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5fcc1b9072c39957f53239dd8b72f8d62d9be30","revisions":{"a5fcc1b9072c39957f53239dd8b72f8d62d9be30":{"kind":"REWORK","_number":1,"created":"2026-09-18 11:17:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/1006191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/91/1006191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006191/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:27:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:28:13.000000000","tz":60},"subject":"Bump up deprecation timeline for CloudConfig class","message":"Bump up deprecation timeline for CloudConfig class\n\nPer codesearch.o.o, nothing in OpenStack is using this any more.\nLets get rid of it sooner rather than later.\n\nChange-Id: Ibb90c3bfc82fc7ed1c3208cb5de017081ec42e2b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5fcc1b9072c39957f53239dd8b72f8d62d9be30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5fcc1b9072c39957f53239dd8b72f8d62d9be30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
