)]}'
{"id":"openstack%2Fopenstacksdk~1006192","triplet_id":"openstack%2Fopenstacksdk~master~I74870b1ceaeee9fe7aff89a66aa8b7b7c9bc24c7","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I74870b1ceaeee9fe7aff89a66aa8b7b7c9bc24c7","subject":"Bump up deprecation timeline for deprecated proxy, cloud methods","status":"MERGED","created":"2026-09-18 11:17:15.000000000","updated":"2026-09-21 13:25:08.000000000","submitted":"2026-09-21 13:22:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006192","meta_rev_id":"cfdba2394480ccc841f004f551d753979a29fe59","_number":1006192,"virtual_id_number":1006192,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 13:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 13:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"340683f393cc10f6fb35777547ba33ba8567a0d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 11:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327335c0b30a999a847a62c5aa13b01e406c5578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:09:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83120a05344a467d9f7e7ab50f65a289\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9fa64833e94f168346ea2851089e12 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a8cce5cd9924d548eaa82ec72f8af5a : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/521108ba2e3640c8892840ca52b3e858 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e71f8aa976b94fc0b773e0f12eb5c12e : SUCCESS in 4m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98f6d80f67c149e7b60ac6604215150d : SUCCESS in 4m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/baaffcde9f3d44148bbb3b84c9d00184 : SUCCESS in 8m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/40e4b2325c614016a9dc4ecaba5e5606 : SUCCESS in 1h 05m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/844aebe4781c4a9084d28d3d5abec667 : SUCCESS in 4m 39s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c774127ae9a4f3193beeb68d0a0e3ff : SUCCESS in 5m 35s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b5f78c727eac487c841824c4b1d3b8f6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4368bcf5122c4c56bcf2c11e595a5c65 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e87a89bf4f14f0381a37cfddd522c27 : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87829b0aa29e48a18296df9d1a516d66 : SUCCESS in 57m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8937c3c5f7714226b19e4d5ee36f5418 : SUCCESS in 53m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4588a97a8173456a8a99e2b2e8fe47b2 : FAILURE in 1h 28m 47s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/f1832b2f6d5d4a7d9f5850903906e7ef : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/26f6f84c07e9464a868e95901ab9c690 : SUCCESS in 38m 01s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/ea64b374ec3f4c1ab0fa1a69cdb51d46 : SUCCESS in 36m 02s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/3121b196442443799499b8a1516123d1 : SUCCESS in 30m 23s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/36c9bb39e86b4a0cb43ab4354126bc34 : SUCCESS in 30m 03s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b9b8e9abfcd54b4f962e0356dc63b734 : SUCCESS in 45m 09s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a4295a81fdc546b3abd484b8e67462d0 : SUCCESS in 1h 14m 59s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/74021ebbf06c47748d06d04828987276 : SUCCESS in 1h 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d415bee64e4a3f64f623a8c7bc2d5759f5b3084e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:44:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da45c6ce9a288f6ec0e4443568e65d9d52c4e5c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"452f54967d47c14f9df447466db5c71917c4f5c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:22:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6e9dd1d2834c938ee8776175e01a1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8fd0a59cafd4292b63a318eaff0c3f6 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d249e1b12be4852a8d0b0d401a300ca : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8b8d999db2d478ebc0e0637bf4cb4cd : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c14bf7ced73849e3a9cbff300d7eee09 : SUCCESS in 5m 02s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bed8d97a38ac44b5883f0fdd5e9035b4 : SUCCESS in 1h 05m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/aaca3ad693df4d3982adb38239895de5 : SUCCESS in 5m 12s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b971eeac59a24642806598f2a87df31a : SUCCESS in 4m 59s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/dd156653b744497e854d5a864d51191b : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a848626ce2041b9ac13d7affe340bd2 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fc515ef85d4482f96d214536b3b7f92 : SUCCESS in 4m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3bd767c76284531a1b17a7762f787ca : SUCCESS in 30m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cacb7376bf0447ec960cf2489ff0dcbe : SUCCESS in 1h 07m 21s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/2d32be3e399044718cd2a2f5c3c64f74 : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3819957f1e6908c0abde8c4cb2d6020c5a85ee0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:22:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cfdba2394480ccc841f004f551d753979a29fe59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:25:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b81db652d3d4960b56c19d2fc3344cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99f0ac3770b4182852819fbe11cdcb7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00528079cc0f9f289df2dd62ed6de1c62c943c1a","revisions":{"00528079cc0f9f289df2dd62ed6de1c62c943c1a":{"kind":"REWORK","_number":1,"created":"2026-09-18 11:17:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/1006192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/92/1006192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1006192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1006192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1006192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/92/1006192/1"}}},"commit":{"parents":[{"commit":"a5fcc1b9072c39957f53239dd8b72f8d62d9be30","subject":"Bump up deprecation timeline for CloudConfig class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5fcc1b9072c39957f53239dd8b72f8d62d9be30"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:23:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:28:17.000000000","tz":60},"subject":"Bump up deprecation timeline for deprecated proxy, cloud methods","message":"Bump up deprecation timeline for deprecated proxy, cloud methods\n\nMany of these methods have been deprecated for more than 6 months and\nhave clear replacements indicated. Aim to remove them in 5.0 rather than\nwaiting another few years for 6.0.\n\nChange-Id: I74870b1ceaeee9fe7aff89a66aa8b7b7c9bc24c7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00528079cc0f9f289df2dd62ed6de1c62c943c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00528079cc0f9f289df2dd62ed6de1c62c943c1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
