)]}'
{"id":"openstack%2Fopenstacksdk~1006193","triplet_id":"openstack%2Fopenstacksdk~master~Ic2bb00f9eb5beccdd1398c4edd4e15844bc58438","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 11:45:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ic2bb00f9eb5beccdd1398c4edd4e15844bc58438","subject":"Bump up deprecation timeline for deprecated cloud filter kwarg","status":"MERGED","created":"2026-09-18 11:17:15.000000000","updated":"2026-09-21 15:03:19.000000000","submitted":"2026-09-21 15:03:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006193","meta_rev_id":"2038d4731aba3ed596f1586cafe9bfccdf36ef18","_number":1006193,"virtual_id_number":1006193,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:45:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 13:09:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16dc1fa8a990d14a7c63d6ad53b6f251d8049993","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 11:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85acb0fa28d93d3b403fe594857e4e5412edf124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d19fc9bedff499a9234176aac425ca0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ef6eb2fd8a45389f57eae421528f12 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bb88dafaf5d4711b7a53f4d1c7ccd92 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e54e8907b7c54de49368cf4f25ef0a0d : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16e1631cdb4c4ed8b10730ee9a7c5176 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9546333e83e14045a08387d7bca600c2 : SUCCESS in 7m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6ef00c6ba8d64bdebf418a07a9c371af : SUCCESS in 9m 51s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fe18b5e16c734a5bb80055a792fe5a96 : SUCCESS in 1h 05m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/67cbd505785246db830d243258f0a233 : SUCCESS in 4m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c611bec94d84387b743b3a485d946ba : SUCCESS in 4m 46s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b6f4143c1a874efea5321e6a2f43aa29 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c084a404844cd7ae4cdba84ad0138d : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12e8392c623412e8c9320d712f285f6 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e4e3f21b0c64cb4a5a5bc8448ffaad8 : SUCCESS in 59m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ee2c73ec1c924033af873ea08c0308a4 : SUCCESS in 1h 25m 45s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d33f3aa46d894c5393f8e20d67dcbab3 : FAILURE in 1h 27m 29s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ced3fb34e1ad47d4a3c691ee889ddda9 : SUCCESS in 47m 37s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/a5677453dd5b4fe69baf1b457ae3b4ce : SUCCESS in 36m 15s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/87eed0220e4e4a939bff0805842deebf : SUCCESS in 28m 56s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/129a70bf97f2426bb9996262a960900c : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/53fad4beb3814b438731183297a60579 : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/260290effd954473839968f69699fe30 : SUCCESS in 19m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/724e7b04004f4fe589cc324a719d6ee6 : SUCCESS in 1h 18m 26s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/69ac329855df4d89937cb814b91a332d : SUCCESS in 1h 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f1932388d27a49c88723a9124d921d78cbadbc4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:44:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f804304ecb532421056ad1a2d9b77af8d09e2db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:44:59.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"85ea87f3f5a339323ff17f61a4479d363354c79d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:45:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"41515882f7de8f584b32125f751d99ed90a36df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:50:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba8acaac73ed5f10c9eac52e71d3b62119dc690d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:22:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55fd49f04597453a8f7803c752b32f34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f732fb55e5a43f5badb28496d4fdb69 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cc854d5fad245869384caa779298f6c : SUCCESS in 6m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/416e06df10fe41c49dd2df0abcb04d1d : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36a58dccfbb1434887fbe432778a600f : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/12165d84cb6445ab831def6c04c7b94c : SUCCESS in 56m 11s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/903634388d6b4cd7a75841dc6cab4e60 : SUCCESS in 5m 19s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8c0346b005d24102b9ec7cd8b3e770bb : SUCCESS in 5m 55s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/51431f4eb15f41aaab9af58fb43af62a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f170910a1a24e8e801b6757ace006b7 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4226a18d00e44d969e8982286b8a1221 : SUCCESS in 4m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f75509e5602a448aadc0a0eabced180e : SUCCESS in 54m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/eda7ba8dbb3147a39f9b7e7f345f87d4 : SUCCESS in 1h 20m 26s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b39a2157bc3148a784e9e066863106d6 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":1},{"id":"481be6d1352e4d3c04b709c9312022bd11fcef3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:25:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a735fb23efb40308314a5321291998a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beaa66afd9fc42bea0d036d8d158dcf4 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bff887195a084c9aa8b2bdd5836917de : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53115cc24f154dff85b833dbe1c3061e : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70e2aeb9907343ddbe9f0ecb484f66d2 : SUCCESS in 4m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2851f77349d747a9b39381c6a577de87 : SUCCESS in 6m 04s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8760b34a1bcc40f193650cd32c5da96e : SUCCESS in 9m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2733c625973345db88183ce3b718f03d : SUCCESS in 31m 18s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dd5138d4cd24427eae978ddf8c65f489 : SUCCESS in 4m 12s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c4057b960ce4bbdb63c9571822da1a9 : SUCCESS in 5m 16s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/43998c638a1143bcb8099971617cc1cd : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ea2d2769c9476b8e9f0832fc6a759a : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1c5e859f3f6440c9a0fc40a9623ee64 : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b69f499a55e40a68f80b289aca0b0e1 : SUCCESS in 57m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a25d526f584049b69cf162ab61a38664 : SUCCESS in 1h 28m 57s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d3648a7e3c9a43b988407a1fb3d65b6e : FAILURE in 1h 22m 34s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2529d33736cb4342947bb0d22aaae423 : SUCCESS in 29m 40s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/8e7ddad0113c407d960b5feda32c65c9 : SUCCESS in 40m 52s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/4de67ea009244196b1d90bd11bc75298 : SUCCESS in 20m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/83c56ae363df4a75a8a62387f5e71635 : SUCCESS in 20m 00s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/523e85a0c5354fdf996803e0a29fcf86 : SUCCESS in 33m 34s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/53273aa854674b1b8f38a17c9d823ab3 : SUCCESS in 31m 24s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d2075f23ed8a46bb802aee301d21b88a : SUCCESS in 1h 13m 46s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ffe8c7b06e854e30aa53e21082b93a37 : SUCCESS in 1h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6978f793e0f1fc6d6bdf378b9f02e3e4855e4480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:26:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d24e29aed904083b38d9d73a4b3690741cf4ee5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef3c09c5e4343d69c8fda517f583cc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94b028a4e1bd4e38b869a919e8684e4f : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72f686adefa34692b2b211d7649d85ca : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4580fa1e6d5f4f58bbd5dc261a3c3d5b : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/86d75f930b714d16a68ab86389f479e5 : SUCCESS in 7m 59s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/36086738bd3241f18df434df00c6f65b : SUCCESS in 1h 03m 17s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/604d8d0b553e41729132ce56fb4d791e : SUCCESS in 4m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d4c45fc0345841cb972a916f59ac3a4e : SUCCESS in 5m 53s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/72c745110de247339f7826624893062d : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221f2c04ed4a4afa927e7f0b8fc9ce11 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb559e8178ad4d0c9f38578d04daa650 : SUCCESS in 6m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4315e1942cc24cc5b3a135dccd1691b3 : SUCCESS in 58m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7b096411c7cb46c5a215f83e46c27208 : SUCCESS in 1h 26m 16s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cd9f948d9a224c5f817537b6d39b7f56 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2038d4731aba3ed596f1586cafe9bfccdf36ef18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"634b49afb18bcf72b9d09ea4b8d307860a03e676","revisions":{"6cb26759f1430dd3af1995cebce847c36ad30e22":{"kind":"REWORK","_number":1,"created":"2026-09-18 11:17:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/1006193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/93/1006193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/1"}}},"commit":{"parents":[{"commit":"00528079cc0f9f289df2dd62ed6de1c62c943c1a","subject":"Bump up deprecation timeline for deprecated proxy, cloud methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00528079cc0f9f289df2dd62ed6de1c62c943c1a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:25:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:33:05.000000000","tz":60},"subject":"Bump up deprecation timeline for deprecated cloud filter kwarg","message":"Bump up deprecation timeline for deprecated cloud filter kwarg\n\nAnother one that has actually been logging deprecations for some time. We\ndo not want to have to carry this into 5.x.\n\nChange-Id: Ic2bb00f9eb5beccdd1398c4edd4e15844bc58438\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6cb26759f1430dd3af1995cebce847c36ad30e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6cb26759f1430dd3af1995cebce847c36ad30e22"}]},"branch":"refs/heads/master"},"634b49afb18bcf72b9d09ea4b8d307860a03e676":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-21 11:44:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/1006193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/93/1006193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/93/1006193/2"}}},"commit":{"parents":[{"commit":"00528079cc0f9f289df2dd62ed6de1c62c943c1a","subject":"Bump up deprecation timeline for deprecated proxy, cloud methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00528079cc0f9f289df2dd62ed6de1c62c943c1a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:25:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:44:59.000000000","tz":0},"subject":"Bump up deprecation timeline for deprecated cloud filter kwarg","message":"Bump up deprecation timeline for deprecated cloud filter kwarg\n\nAnother one that has actually been logging deprecations for some time.\nWe do not want to have to carry this into 5.x.\n\nChange-Id: Ic2bb00f9eb5beccdd1398c4edd4e15844bc58438\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/634b49afb18bcf72b9d09ea4b8d307860a03e676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/634b49afb18bcf72b9d09ea4b8d307860a03e676"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
