)]}'
{"id":"openstack%2Fopenstacksdk~1006194","triplet_id":"openstack%2Fopenstacksdk~master~I4d03c5f15c89d72b88bb4935375a2fd3336bc36e","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 11:45:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4d03c5f15c89d72b88bb4935375a2fd3336bc36e","subject":"Bump up deprecation timeline for passing dicts to cloud APIs","status":"MERGED","created":"2026-09-18 11:17:15.000000000","updated":"2026-09-21 15:06:18.000000000","submitted":"2026-09-21 15:03:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006194","meta_rev_id":"d790cfcb9b254b54cfae9f061f95cbf82810746c","_number":1006194,"virtual_id_number":1006194,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 13:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6749d0454269c8973af99dbd1d9a878768c9afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 11:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8caaca2fd32e62514aa1b735f9ad7cb8c103066a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb57eb3543c541d78feb89fd3afb62f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff5bcdb8316467298f021863db683fb : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d4968773c63449e81825cf454245ed4 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae6378b7f701473fa01e5b6c71f93b6a : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c461ca0f549b45f595a7c60dd5b7da01 : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38e0cdfb639f45eeb2df8f68d9dbf949 : SUCCESS in 8m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6d874713672d4af5a216fbd5a8a7189f : SUCCESS in 7m 07s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/23435172bce043a89927bf8a0a5085ed : SUCCESS in 31m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e4f4515692274505a420b00031afd6c2 : SUCCESS in 4m 23s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bbc0d986c1a64469a1e165572e6f9eac : SUCCESS in 5m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0924b8858521494bbbe6bfe453862b6e : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef5e630527f4ac99c23550dfbe0e5a6 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b06390868e9475ab70e3dd3ccb349df : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b4b8e2decd341ccb3efdd22df94555c : SUCCESS in 38m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ad424821c86a489c8a504777e384f8d2 : SUCCESS in 1h 00m 08s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f35a4174ea8d4a7a855d316c501372fc : FAILURE in 1h 40m 29s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/78560699c89647fe86e81297385d17a3 : SUCCESS in 44m 00s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f15c7730b516405ea41be8efc3fc64ee : SUCCESS in 36m 49s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/70b6fb85a0e94efcb21d42bbfba6a814 : SUCCESS in 30m 39s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/a6701601d63043fa9b2892f8e275b246 : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/93b6da086b424ab389c38143d36b0f67 : SUCCESS in 31m 08s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/99098eaf485744098f331fbb69d00dcb : SUCCESS in 42m 15s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b00a8a14547e493abf40bb18eda3684e : SUCCESS in 1h 08m 42s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c56fc61271624a6e859b29c64a6cfb77 : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe31759601e8b52a0d8d686891ff348ef8486056","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:44:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f63f2a0054ffc43acd50c362a77aa6f8b485f42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:45:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0edfd0386b9073782006d1102b9119ea89009bc8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:45:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f4f883e8b747489bde916aa77d7fb78b8289ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b85f8817a0aea8e9a307e3396ed8495a0c2eeaf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:32.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstacksdk/+/1006193 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"5d2cfa114c2fb877d180d2adc8a0da31ade12fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:11:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4957f3786f948988543721b56cd6f7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99b252bd1f541788100f071695e6b19 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c7a65664c7042f489fb794773c2c67a : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e2400f0b3f64026aaa761b39dea429f : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2dd5f0db225a47f782d08a5c3ec79f96 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b36b4493da3044a6afee7ef149c6dca0 : SUCCESS in 6m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d045c29e58c54b198f46d2ed5c1c316a : SUCCESS in 7m 14s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f47b9f39ec1442eaa847942eede0ddfe : SUCCESS in 45m 56s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/117eb3b25f664ba7b478ebfcd340491e : SUCCESS in 2m 58s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/87e78231d7034ad0b2050c63a7ff4cd9 : SUCCESS in 6m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e7b887eb6082447ba2760e4254ede410 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef75feec695d4bc19dddde506fafc230 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724215e2b8634637ae35b4b5f8fe52a5 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b52efaebc56648b69337942fda77e027 : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/40bebd392790417fb82baed442d577b0 : SUCCESS in 1h 15m 29s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/73c8c2bf43864645aaf81b2b7f907153 : FAILURE in 56m 38s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/6a6684180cda496a92a371992dc74348 : SUCCESS in 38m 33s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2e8e97fee1ab4967b4d8b935b6cf44bd : SUCCESS in 26m 11s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/d23644b20543415f80aa89d1b9c11277 : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/4a4b8efa475040b7a042b05829e5147d : SUCCESS in 36m 18s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/cbc4d0b5d0f3473e86ec44ad1051c2d1 : SUCCESS in 31m 53s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5c08157bceb848f2af46af1b35620fd8 : SUCCESS in 30m 52s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/31b9f3c7119049848bb85902f8e3234c : SUCCESS in 1h 14m 30s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/aced31635444402fbd2cf117668fad3a : SUCCESS in 1h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1cd8a0b5784e1265854136e40581cbb8e72a82e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:26:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a6d5eda6916d1cc6642793b2a75b794909ef9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f30f09ffaaf4249b1dec04bf9704f99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dbdd4c6c61a4a478487e0a84dc8c88f : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f77e8becaec4073bfd0ff1dfbc20339 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31233d57345c4b99ac7903f4f2c2f3e3 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a66c57ad110a4273940f12e40c1f1249 : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5e2548973c5943a4ac7aed630738ec91 : SUCCESS in 1h 00m 31s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7a9b5f619c4142c581633fe8e090bbec : SUCCESS in 4m 57s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/32f66d4271ba4d1ba28cf09ce9ce67ef : SUCCESS in 3m 16s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/dd441420853e4453aa03c9ad2c1b93f5 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e84dbc8401447cb92cecdaf118403b : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f1568ee310d45849c933efbc9d3f821 : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ee245140ce843ba9c05ca3aa9f21ea1 : SUCCESS in 1h 00m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/90828b53823945269960405d20ad4781 : SUCCESS in 1h 21m 08s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/82a219135c2f486592512c47d44b2c21 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":2},{"id":"13fd4dbef9e68ce727ceefe53c777ee311708579","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d790cfcb9b254b54cfae9f061f95cbf82810746c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:06:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8846cf133721491388e5f7be7219656d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77456cefa41e4266943ed413f15c2136 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5280da57bfed16a2c679bd771704bffac6b9f4d","revisions":{"88c8f092478c34fbc1673884fe44609838113bd9":{"kind":"REWORK","_number":1,"created":"2026-09-18 11:17:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/1006194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/94/1006194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/1"}}},"commit":{"parents":[{"commit":"6cb26759f1430dd3af1995cebce847c36ad30e22","subject":"Bump up deprecation timeline for deprecated cloud filter kwarg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6cb26759f1430dd3af1995cebce847c36ad30e22"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:26:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:33:22.000000000","tz":60},"subject":"Bump up deprecation timeline for passing dicts to cloud APIs","message":"Bump up deprecation timeline for passing dicts to cloud APIs\n\nYet another long-deprecated thing we do not want to carry into 5.x.\n\nChange-Id: I4d03c5f15c89d72b88bb4935375a2fd3336bc36e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/88c8f092478c34fbc1673884fe44609838113bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/88c8f092478c34fbc1673884fe44609838113bd9"}]},"branch":"refs/heads/master"},"d5280da57bfed16a2c679bd771704bffac6b9f4d":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-21 11:45:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/1006194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/94/1006194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/94/1006194/2"}}},"commit":{"parents":[{"commit":"634b49afb18bcf72b9d09ea4b8d307860a03e676","subject":"Bump up deprecation timeline for deprecated cloud filter kwarg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/634b49afb18bcf72b9d09ea4b8d307860a03e676"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 10:26:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:45:06.000000000","tz":0},"subject":"Bump up deprecation timeline for passing dicts to cloud APIs","message":"Bump up deprecation timeline for passing dicts to cloud APIs\n\nYet another long-deprecated thing we do not want to carry into 5.x.\n\nChange-Id: I4d03c5f15c89d72b88bb4935375a2fd3336bc36e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d5280da57bfed16a2c679bd771704bffac6b9f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d5280da57bfed16a2c679bd771704bffac6b9f4d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6cb26759f1430dd3af1995cebce847c36ad30e22","ours":"88c8f092478c34fbc1673884fe44609838113bd9","theirs":"634b49afb18bcf72b9d09ea4b8d307860a03e676","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
