)]}'
{"id":"openstack%2Fopenstacksdk~1006219","triplet_id":"openstack%2Fopenstacksdk~master~Ib8003b029d90ada25a6442cac5fa1ce91fa00937","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Ib8003b029d90ada25a6442cac5fa1ce91fa00937","subject":"trivial: Typo fix","status":"MERGED","created":"2026-09-18 14:26:00.000000000","updated":"2026-09-21 14:55:19.000000000","submitted":"2026-09-21 14:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006219","meta_rev_id":"87104618a19d956d7d1023fe5d93589e009b563f","_number":1006219,"virtual_id_number":1006219,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 14:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 15:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8a892724b026fccf82a690196dedd07c14093d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 14:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7374e2bb354a4eab4a27b74154d16e5b3a87af39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 15:56:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d0b780d3d6455db79bf264763eb2e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260b9f98ee664a2bb40270e333bc1596 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa7c58b34ef84758be5f9c6c122280c4 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/139040baca094c7aa6a94ef7536c4873 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0480228188140fb97b83b08f7bee817 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8cbbf5d6c66e4527b35eb31d4f3e379f : SUCCESS in 5m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ee64c601f6a44f08a94bda35be9efc25 : SUCCESS in 4m 25s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6b7d87d0b1c24d55839c25cf8ed8676c : SUCCESS in 37m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4ed6fac1178a43609ddb277786423af9 : SUCCESS in 2m 54s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/947b49f03138458a87b4a5b8c96dcb33 : SUCCESS in 4m 27s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/5002f52da1d64317a1279210f98849f8 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e50f688d4d460daf8651e3e85ad5e3 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25aa976c654c4f86b51e542325077d0c : SUCCESS in 6m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0da8560e116f403c8d8d5d2ed0f65e0c : SUCCESS in 35m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e6b7cc859ecf405da19bd6a7a1a75250 : SUCCESS in 1h 22m 59s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d896761685794f73b1e39ef52f4c42a4 : FAILURE in 1h 04m 54s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/f7b29c12f21342c582827fb7c0132678 : SUCCESS in 40m 59s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4cdfb455284d4a33837c0fcb658a8291 : SUCCESS in 30m 15s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/a1ac2dfe211f434ebf19d4be60adee8f : SUCCESS in 31m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ef1725874ba44a29b81b35925b0420e7 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f63756f2348048ec972578276f596b96 : SUCCESS in 18m 32s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5d6b0129cc9d4ff78847c3c06310afdf : SUCCESS in 23m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a7d2e26d81b1470a84bec0fdaf4a7561 : SUCCESS in 47m 23s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0d283d2c78da4c9397925ae6558cf071 : SUCCESS in 1h 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8f2c3ea9fdc9a396389654c88cab2c280fa6544","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecb7f06ae2837c3668f6248d15add0b29dba2330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4346b3ac8118efaf05d59fb287ed1b68a8dc7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:54:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a6825602024fc7a0f3b7f330536000\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a6107ec91f4535b72c80f01ab7ab85 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd606ec8254b4cbe9e0bd47a9c0d95ad : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cba471fc3914b58a2ba069423af3f2c : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ebf483bd87394f809966c32552321abb : SUCCESS in 6m 56s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fd89694e95584c2f8a8a16a2ea41739f : SUCCESS in 1h 04m 19s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3515f50f4af94b3b8c1a01135b6edda7 : SUCCESS in 4m 08s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b71297b89fec46708714a97487373f96 : SUCCESS in 4m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/41204d7bee4e4c9dab4bdb69ed9ddcbe : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e4a1dd8db64077878ec29c2a5cfac4 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9543a056e4f84a97babd6aa86073725f : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b0ddd232ce84ec3bff2378f257e4e9e : SUCCESS in 48m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7b2168836f23432ab7b2f6013ac33d47 : SUCCESS in 1h 24m 19s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7fe149e9ae93410e810434331d5d3087 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2cbee5c7fa13c0714fd763de5bd9e2f8e9b12ea7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:54:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87104618a19d956d7d1023fe5d93589e009b563f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44339e5612b8470abf1819a43ef1baf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe9740d73154989b2edc1e640c5f8a6 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"013d0afa5cf77f41e019efa1a72a0b69450a5b35","revisions":{"013d0afa5cf77f41e019efa1a72a0b69450a5b35":{"kind":"REWORK","_number":1,"created":"2026-09-18 14:26:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/1006219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/19/1006219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/1006219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/1006219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/1006219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/19/1006219/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 22:39:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 22:45:27.000000000","tz":60},"subject":"trivial: Typo fix","message":"trivial: Typo fix\n\nChange-Id: Ib8003b029d90ada25a6442cac5fa1ce91fa00937\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/013d0afa5cf77f41e019efa1a72a0b69450a5b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/013d0afa5cf77f41e019efa1a72a0b69450a5b35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
