)]}'
{"id":"openstack%2Fopenstacksdk~1006220","triplet_id":"openstack%2Fopenstacksdk~master~I133cc7866fd094b13cc8e20a7a88bf143974887b","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I133cc7866fd094b13cc8e20a7a88bf143974887b","subject":"proxy: Group methods","status":"MERGED","created":"2026-09-18 14:26:00.000000000","updated":"2026-09-21 14:56:31.000000000","submitted":"2026-09-21 14:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006220","meta_rev_id":"171ab984f4474265bcb4a0930c96837787dbb71d","_number":1006220,"virtual_id_number":1006220,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 14:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:45:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:06:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b75344442da109deb216c031f6ef9276fbc73b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 14:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c88839d692d558b3d1c71a5002c2953c19279b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:06:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c808e409995645ff89592732ba0b1432\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc3a6a8bc8a40bdaa9c49108727bbb1 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c249c1bf11004dddb3594b3fa24f7ef2 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51f90bd7d8ed4cbe8e508960753635b9 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/946aab270beb4c5eb872e3019c88d191 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c4b05b1873a40cb997b4aa30ea2e01c : SUCCESS in 9m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/56b286aab0994c75b4e2ec127dfc91f0 : SUCCESS in 7m 14s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4200f14fd61845ad85177f6ca7bc8a3f : SUCCESS in 29m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/26bd8d460ae64877915f84e726493347 : SUCCESS in 3m 00s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9f24ce3c72bb4303a7c6cb522535c57f : SUCCESS in 8m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6c6fa4f83b5c4065bb0d56218d065664 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d157a94c0846f2ae5c3d18f59c21b5 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800649aa498849f0966cb37ac8c80f92 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84257848e82f436c9db812e4aad4fc41 : SUCCESS in 29m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce1e57dffb8f4798870cc11e300e5603 : SUCCESS in 1h 01m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/57fa71f569944abc9d62634715a32e91 : FAILURE in 1h 24m 28s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7c1fdd3cd05049549f165fc31c7049d6 : SUCCESS in 41m 35s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4ba01c12dfc444b4aadecc70580a7ff1 : SUCCESS in 32m 01s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/d3ad728bf33643b282a7fd2c8cff1cc8 : SUCCESS in 18m 04s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/14fa7c85957e4dcba0049b7f656b8a12 : SUCCESS in 17m 25s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/12997e988c184cb79b36054bc1e51a1f : SUCCESS in 32m 14s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/98d65cd3b7ab408b9af5dc480008b369 : SUCCESS in 33m 12s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/847cadcaceb04707b338723c6bbaf7c0 : SUCCESS in 1h 13m 18s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a086a7a73b10440982a2356ca3630ef5 : SUCCESS in 1h 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"627fa29783c902626c1c085d86e6b5972f55c4b3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:45:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dff7f8a97bfdb3ffda211bdb7fb5468c17aee17b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36bee310766a8d6526c61fe4528ba4cf1f6df8f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:54:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60fb74532c44453f9f2f7c1c6a1456c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4b41274968420bb9b2bf2a935375a5 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25261b7f96704aed9892b390172e7ff8 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd2b221a79a43409dd3419e07395fab : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e107441a2654f8b97e1619459b37b46 : SUCCESS in 6m 52s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ef8ce228c0e24c1ba0b0e1e7f8aeb20c : SUCCESS in 49m 28s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3c1009950fd54100b4d9eec5f6bc08ad : SUCCESS in 4m 00s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/49c9106eb08546d6a62906505f893e68 : SUCCESS in 6m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0e5bd184ad6e4e91b61771b6a4a4c418 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98cd1240fccb4f88bb6e7daae4779c69 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/184276ce405a49fc88d3d5164e4bae1b : SUCCESS in 3m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c31de3502e30447ca3ac788fb900bf4e : SUCCESS in 1h 02m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7cfd6f885db248fdb37676fae81866e6 : SUCCESS in 1h 21m 02s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/6e512173e1ba4cfe921bacebb761c1d2 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a67cde0a742d8181c95e67c7e6198a2d7c770ffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:54:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"171ab984f4474265bcb4a0930c96837787dbb71d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6251c00a7bf3428b9a02e53beea271e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8a2324242b426f8cd7fdc6201f0932 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d61e487a22bc9a6add99879293b6689744ad322a","revisions":{"d61e487a22bc9a6add99879293b6689744ad322a":{"kind":"REWORK","_number":1,"created":"2026-09-18 14:26:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/1006220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/20/1006220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/1006220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/1006220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/1006220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/20/1006220/1"}}},"commit":{"parents":[{"commit":"013d0afa5cf77f41e019efa1a72a0b69450a5b35","subject":"trivial: Typo fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/013d0afa5cf77f41e019efa1a72a0b69450a5b35"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 14:37:11.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 14:25:49.000000000","tz":60},"subject":"proxy: Group methods","message":"proxy: Group methods\n\nChange-Id: I133cc7866fd094b13cc8e20a7a88bf143974887b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d61e487a22bc9a6add99879293b6689744ad322a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d61e487a22bc9a6add99879293b6689744ad322a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
