)]}'
{"id":"openstack%2Fopenstacksdk~1006234","triplet_id":"openstack%2Fopenstacksdk~master~I033a5c5158c903885600ad520ecabea9dd732712","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 21:14:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I033a5c5158c903885600ad520ecabea9dd732712","subject":"config: Stop ignoring .0 microversions","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-21 22:46:06.000000000","submitted":"2026-09-21 22:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006234-placement-gaps","meta_rev_id":"699caf4b12c6f6f52ec7528d365eac2b3f4276b4","_number":1006234,"virtual_id_number":1006234,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 22:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 15:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-21 21:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-21 21:14:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:28:13.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"5974364553501392c4025cf93e20ea748ba7ed60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36d6bff0cfbf15a8a08e264cf4899646c9a8163f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368445e7221e451d8f10fe116edca6f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712cf9c9b5a040eea137bb62242a5fd0 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93f23f4e59544beaa68d6f7ceac6e978 : SUCCESS in 6m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d252b1eaa6c64b1ca149ab591a3f0ca3 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe0d8c2fcd34abc8be562eec2f85fef : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2eb9d837e1384e54bc8055649bf55427 : SUCCESS in 4m 23s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bc4f037a5aeb4336bb8a0736051e7ea6 : SUCCESS in 7m 47s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fb759696018a48559fb311a3c1571c1b : SUCCESS in 50m 58s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/acc37610e1584a1b96cbf6161cb8caee : SUCCESS in 5m 47s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6207f87d27424b7ab04ce07ec7ced3b5 : SUCCESS in 5m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/464bfcf9e8a74dd9a628810289814e6e : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cbb40d3b7a4a2d9373d49d7e28d102 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5722c040282c4d58808b678a3d26663d : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59440d833dc342c483c51f509bcedd48 : SUCCESS in 32m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d02f4a7bcec84d9d8066e877000caf01 : SUCCESS in 1h 29m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/3b017ae846cf4e89b99658688435b049 : FAILURE in 1h 30m 58s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a224a8779dc543e293293e8e862b5af3 : SUCCESS in 47m 52s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/51396f3a787d4c8196a5913782357ebf : SUCCESS in 36m 34s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/2b5ff114e7734decb75ea1cdae8de950 : SUCCESS in 16m 26s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/dadde63671274bcea32cb893c53d1acd : SUCCESS in 30m 18s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/49ac9a6060e843538f4ba7f8cceeddbf : SUCCESS in 29m 41s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1cd0f43835f1491baee5823931770258 : SUCCESS in 27m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/acfcc7bc1d3f4738981b895fde6e63cc : SUCCESS in 55m 04s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2b9992ce43cd4adcb7b1c438c35c3299 : SUCCESS in 50m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3102b45e22b0d87ad2f734934df7117810cb42af","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:28:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dde8b47350c33a2f1e34fe8627049c5ef89d6b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3123709d59882b92ec359cd1de06fd95728c7fc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a863078da84f36a1b1a0ff4e3d98bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb6ff0d6ab345048ade1800f251dc49 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb7fa7cd9fa9468ba12d91e4978c8c3e : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/488d862cce2140bd9872b85192121664 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d77d5fc6980b4daca16bb6ab2f4a6d0b : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d30a206676db49d38e256f62a9027751 : SUCCESS in 5m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ee18462ed690470280b4e4471ba3d846 : SUCCESS in 7m 12s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/37b826d99e404555b23f0b67a9af71d6 : SUCCESS in 1h 08m 58s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4ea49de5c0014c8794a666598f6b4d58 : SUCCESS in 5m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a3e0537a63274e148c154f4d423c534c : SUCCESS in 3m 47s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/28c05d78b4e64b92aaaa4df1f7c5dad2 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c39f041d0843f29ead30349d14a1a9 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/785fc5e9bd84417595664664d89fb11a : SUCCESS in 6m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d3366c7e828436cbedcaf8d76d74aff : SUCCESS in 44m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/50dbe50ee67943a1b7dcd3743d495aad : SUCCESS in 1h 29m 38s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/430e7bcc7ee0444987dc2918b25850ae : FAILURE in 1h 16m 00s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/48666b4f418c45b0b6ab325a8fb8db23 : SUCCESS in 34m 26s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f3c1df08e03c4f2bac66408bed67653e : SUCCESS in 25m 30s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/18dcb38617a84e8a80fb23d4c23b114c : SUCCESS in 22m 01s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/7c2db740a86e4234bf82a947f6008240 : SUCCESS in 17m 08s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/483adcb2938d4082ad4ed3f0ba7cbbff : SUCCESS in 31m 42s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/821f2374ad764a84b3074dc49cc20cd2 : SUCCESS in 49m 24s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/146f453132a04d63849f75189966ffcc : SUCCESS in 1h 07m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7724a1284e224ecd8136d77be4705826 : FAILURE in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba9b26454f22c9775bd2d934ad5dfad4116b3c16","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 21:14:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"10183dcf73de70ba2cc7d5d711269b22a9b5070c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 21:15:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f71d80509e1aced6722b4fc1486cedceafbac9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:45:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee834e5f23b744c8a9f6c42eb81d97f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e4bf116cfc44bca9f179d95d04390c : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bfcb5f1614a49068c1df979523507ac : SUCCESS in 6m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0001b17ceb6b48869b359425686eda01 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa85752767954991be9b8896c87f19c8 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/21e62aef8df7418d8295f504efea477a : SUCCESS in 42m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5601e23a81f64454bd8a58f3b1ff5e73 : SUCCESS in 2m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/aa21d7e1cf07473da9941e6da0797dc2 : SUCCESS in 5m 11s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/37e25c34b11741d096516e18910c73f8 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bbb381dc6e49cb968b7d9971767945 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e84c0e3dcc4e8aa834ae2921dd85be : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f28d6af78e14fa38ed817f3dced49ac : SUCCESS in 43m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6aa4cb11d15e4691a602b45a287bedd1 : SUCCESS in 1h 24m 36s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d29eab7003a84e2a9be588c071b1ff62 : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f5d9994ea9fe0de5964120317a18a6e09d9c80ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:45:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"699caf4b12c6f6f52ec7528d365eac2b3f4276b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:46:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4065fc81f09a48d89ea715dd6e588c96\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd9f978fc714cbaa733fcd97e8a27a3 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2","revisions":{"76b3143dc99f2dcf44192b9aad71ab431e6ada04":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/1006234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/34/1006234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 20:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"config: Stop ignoring .0 microversions","message":"config: Stop ignoring .0 microversions\n\nJust because nova started microversions from .1 does not mean other services\ndid. Neither ironic nor placement do this, for example. Strip this special\ntreatment of the .0 microversion and pass things through as intended.\n\nChange-Id: I033a5c5158c903885600ad520ecabea9dd732712\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76b3143dc99f2dcf44192b9aad71ab431e6ada04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76b3143dc99f2dcf44192b9aad71ab431e6ada04"}]},"branch":"refs/heads/master"},"eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/1006234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/34/1006234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/34/1006234/2"}}},"commit":{"parents":[{"commit":"0382628a572ff87f83cbf8fbfcf4b8ef5d5a0579","subject":"Merge \"zuul: Use proper boolean values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0382628a572ff87f83cbf8fbfcf4b8ef5d5a0579"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 20:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:46:55.000000000","tz":60},"subject":"config: Stop ignoring .0 microversions","message":"config: Stop ignoring .0 microversions\n\nJust because nova started microversions from .1 does not mean other services\ndid. Neither ironic nor placement do this, for example. Strip this special\ntreatment of the .0 microversion and pass things through as intended.\n\nChange-Id: I033a5c5158c903885600ad520ecabea9dd732712\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
