)]}'
{"id":"openstack%2Fopenstacksdk~1006235","triplet_id":"openstack%2Fopenstacksdk~master~Ib2b21e2bfee26d76e6e5bf1a50d2c161fc5ada1d","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 21:14:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ib2b21e2bfee26d76e6e5bf1a50d2c161fc5ada1d","subject":"placement: Fix trait create and exists","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-21 22:45:05.000000000","submitted":"2026-09-21 22:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006235-placement-gaps","meta_rev_id":"178f7667e15931d2aed2e3e3439f4e2ebf3a08a2","_number":1006235,"virtual_id_number":1006235,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 22:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 15:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-21 21:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:28:39.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"76d6e77e186a39ce97ed0a34b8cb476b1213855f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e46d5c8086936446225c132ce9b005192ce84c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10bb89c52b1c40b0b9b7705bc1700b2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c429e6537294387aa76855047463e1d : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1631e80f2f34da8b6fc2346fa2e47e7 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b543b43328e43379fa81d96d2e6319c : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/813bb355298249d7a196adb0e96b0c63 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/907b0565087c4c6d84f2bdeaaef10fde : SUCCESS in 7m 04s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/97805e439ee144a198b9c7697b34aca1 : SUCCESS in 7m 48s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9106379f97784eeca54957a08460d95d : SUCCESS in 55m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/62d8ca4302de47cb9c4569d7bd9c14af : SUCCESS in 4m 07s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/40e5dbd971644ed4a1fd0b16bc10e08b : SUCCESS in 4m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ab7e337346f74b578f5d49f86e702077 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70aeacdedc814bd6a2f1a6d694924f95 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c7b7a52bf234900902e5c7c443d004b : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bbce6f412ad463489fe6d898121ff46 : SUCCESS in 1h 04m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a9b60354b8024e4bb5f004dc738e0dc7 : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/383730d1309e4c8eacdd05e3dca96bab : FAILURE in 1h 28m 46s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/3f563bd75217443b8e5ebd8e91f5d923 : SUCCESS in 37m 40s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d109f04547064cddb9a97d4145fab204 : SUCCESS in 35m 45s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/cd6dee5bc0e94b63b3ee2af1e3a147d2 : SUCCESS in 29m 33s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/f04742ee363c4f8887d15896edfe988e : SUCCESS in 33m 37s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f0a8ccffe7754134a46339306a51928b : SUCCESS in 32m 34s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7122fc5d5de84137a51a3e820b87a0b2 : SUCCESS in 48m 15s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f8499e18bc5e4cd8af81d1dd2df2fc17 : SUCCESS in 1h 12m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/e76861df9f8c4b40aaa4044507df7da2 : SUCCESS in 1h 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa31e7a0b5fdf78f54d34f88d804745aeb3993b2","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:28:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c44dba648e37ebd1c1ea070c8bf3f7bf2d8fb1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7b15b0c25562ecaf3ff92010e52a7c755000b6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:44:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/141a8e19de384c08bc83fd17bd5e3980\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942f259f082d42838bdd0450d8b60ed5 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f873da833be2415495c6f8be10655e5c : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a14cc0c670645cb87f8466ca8951742 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8ed73895b5b4a7b86d6472c08dbec91 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3884c06873484cfdb42ce139040f302f : SUCCESS in 6m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a42edd1f580a40ec9bded51f80ceb78e : SUCCESS in 7m 21s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a9b62800748b49fc84c58fc6945fab56 : SUCCESS in 43m 19s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e77a7bf536404459b2da88453fd92c84 : SUCCESS in 6m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/45e15031221f43dea0b41e39180b9edc : SUCCESS in 5m 07s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1fd8a8db0be14ed9ba6914f28902db9b : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a303727151c64d918a4924867067cdf7 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8f8004c511d4bef84fd250c21131594 : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb1f160962cb421b9b0589e1ef1ed323 : SUCCESS in 57m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d9a54b37e5554733b2d539fe48541052 : SUCCESS in 1h 19m 09s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/fa1a57471e5d41a0b30ee43b9dcb691f : FAILURE in 1h 26m 59s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0a5ddf26f4ab4a0db526251af3bbcfad : SUCCESS in 39m 41s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b19b62fc131348ccbdf8862a3317c1e6 : SUCCESS in 24m 33s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/7e7a65b7c86a48bca85810773cd978d9 : SUCCESS in 31m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/39c662b1441d4be098c271d0433bc15e : SUCCESS in 27m 09s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/7569304af497473690f325b8decfacb6 : SUCCESS in 30m 15s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/4e91f50692434e0a93d5333edf7c9acf : SUCCESS in 28m 43s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a5f2be3cdc1d46b0add881214a45ac9c : SUCCESS in 1h 20m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/85837f8423324b9190e0dcf3461fcb03 : FAILURE in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aef36e7ae668c0297de516ed9453701f1ae10649","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 21:14:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d79b0bff23c02535f03091b6cd2b43dfa396090d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 21:15:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24022e348c4649d36efb52e1e580b5f776d6cb76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:45:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84381165640545738ffda1116c4d8ab6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93313d1f4f34f71ba96fa304b85d35f : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7baab0002ce9490aa45523b00e19dade : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce261e7f16204d83ba61c5cafa647908 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09784ea182f24209a9ec2c28209a6850 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/91c78efa4afa41d78b4fd097f2100ba3 : SUCCESS in 51m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5c77a8cb1b924b63a093c9aa3d2d0bc0 : SUCCESS in 3m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/08ea77e822954252a88fb7977fc1424f : SUCCESS in 4m 45s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6d44fe8812d44ff59e037c01d4a9cc8c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a761fb9105465fb3ccdb05bac6e9db : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e74d3ed787a45fca494a053c4328b54 : SUCCESS in 6m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9e717fb9c994760a41d83525f3853a1 : SUCCESS in 1h 01m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/54b5ee82bbce46cfbf07ee8a67c83993 : SUCCESS in 1h 07m 18s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/72e42ef89bd540dabe48efc88ff890c3 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":2},{"id":"178f7667e15931d2aed2e3e3439f4e2ebf3a08a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:45:05.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5f5c8d818110549b1b53ddee9aade4b71bfc4f6","revisions":{"00b37a0a8b9c7ad6e05e97223e40b44b56a7532b":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/1006235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/35/1006235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/1"}}},"commit":{"parents":[{"commit":"76b3143dc99f2dcf44192b9aad71ab431e6ada04","subject":"config: Stop ignoring .0 microversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76b3143dc99f2dcf44192b9aad71ab431e6ada04"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 22:41:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Fix trait create and exists","message":"placement: Fix trait create and exists\n\nPlacement\u0027s PUT /traits/{name} endpoint expects no request body (the\ntrait name is in the URL). Override create() to discard \u0027name\u0027 from the\ndirty fields before calling super(), and set create_method \u003d \u0027PUT\u0027 so\nthe base class uses the right HTTP verb.\n\nIn addition, change exists() so that it passes query parameters.\n\nChange-Id: Ib2b21e2bfee26d76e6e5bf1a50d2c161fc5ada1d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00b37a0a8b9c7ad6e05e97223e40b44b56a7532b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/00b37a0a8b9c7ad6e05e97223e40b44b56a7532b"}]},"branch":"refs/heads/master"},"f5f5c8d818110549b1b53ddee9aade4b71bfc4f6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/1006235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/35/1006235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/35/1006235/2"}}},"commit":{"parents":[{"commit":"eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2","subject":"config: Stop ignoring .0 microversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb6350ef6b0ebd5d6ad7b34c74df799165ee3be2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-17 22:41:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:46:56.000000000","tz":60},"subject":"placement: Fix trait create and exists","message":"placement: Fix trait create and exists\n\nPlacement\u0027s PUT /traits/{name} endpoint expects no request body (the\ntrait name is in the URL). Override create() to discard \u0027name\u0027 from the\ndirty fields before calling super(), and set create_method \u003d \u0027PUT\u0027 so\nthe base class uses the right HTTP verb.\n\nIn addition, change exists() so that it passes query parameters.\n\nChange-Id: Ib2b21e2bfee26d76e6e5bf1a50d2c161fc5ada1d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f5f5c8d818110549b1b53ddee9aade4b71bfc4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f5f5c8d818110549b1b53ddee9aade4b71bfc4f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
