)]}'
{"id":"openstack%2Fopenstacksdk~1006237","triplet_id":"openstack%2Fopenstacksdk~master~I09e6443938531c68bb66f9c2ba563aa2afdb318a","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-09-21 22:46:12.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 00:17:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09e6443938531c68bb66f9c2ba563aa2afdb318a","subject":"placement: Add support for allocations","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-22 00:18:34.000000000","submitted":"2026-09-22 00:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006237-placement-gaps","meta_rev_id":"1ef60271f925cbf1f1eeead96766c75772ace3d0","_number":1006237,"virtual_id_number":1006237,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 00:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 22:46:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-21 22:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:33:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"bff84d5b5a2671f0a62bf547ce50b71ed4e0ea7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7fdb21a2f2935390a92adeb3f9568c0b1a92fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:56:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b599626a423c4f2cb1deb3014aa4e140\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26af9f15fd6a46cc9eae48fdd3565151 : SUCCESS in 7m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/856824ce74a04d72a5d7e86ab682af13 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5658791c2dd34265aebe910508097614 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4992409531244a9a863e80a70f8e2dc8 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8f8de87311349a6bf409dcc6da68742 : SUCCESS in 6m 41s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/452de1f39c0b4a1a8cc99d96f06e95c2 : SUCCESS in 7m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/86701c1097e04f5b8004e501605dd5e2 : SUCCESS in 1h 06m 12s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e459122e27c74dd5aa69a632cdb50586 : SUCCESS in 5m 50s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ad5536952f074e9394b06d58722fe7e2 : SUCCESS in 5m 37s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/91c6f6485f1f4081a9c3dea79ef94379 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164639eb4bf248b8ba0c6d3acff6f73c : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/705cfa1b2a6046b69d87a7cb8f6acad3 : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2af3e1a4faa84362ac072aa00c6b52b8 : SUCCESS in 1h 02m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a7552672b7544f61888690bc8653a16e : SUCCESS in 1h 22m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/0b6622722cf540d1a4a0264387a6c2ba : FAILURE in 1h 35m 07s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7de279d112e24f90bab9cb0a74302a30 : SUCCESS in 46m 21s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5fc03adea8dc42e48fc7ec9774e47524 : SUCCESS in 38m 21s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/fc1606ab2b4f4ee9bcec5ad6f7c10a96 : SUCCESS in 32m 04s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/94619a82b81645a8b75105974308e271 : SUCCESS in 35m 14s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/61268c9e967141c19e4e0a766c40a056 : SUCCESS in 36m 36s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7d604117633145668585c79edb7cd283 : SUCCESS in 49m 39s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/00af9de087ad4a1ea6929939b06a216f : SUCCESS in 1h 17m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1e2af78ddb0247c881fa1cb4017a5d7f : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"044414a2ec0db4d0a3f6fdb2e200bbae674ba3e4","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:33:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0363c34acf6bdc95264d377645d69aee19282d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9b38be483ceea7c701dd88a38bc879fd22b10a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:42:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d351827b94fe47818a35fdb32e8c87f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853ad145b09445a28dc6def4062b28cc : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7755f125791345f6af8af09e4733a345 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd3f494e1bbb407eba8d51fd1b2be17a : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/027f6860fd00431cbbc2422cb003a725 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b42e8b6d12e4d238a890959f135eaf1 : SUCCESS in 8m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/af191042d0014864b9975938adce2c0a : SUCCESS in 9m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/283f4198a52f4722b62ccbad43ce92ee : SUCCESS in 57m 35s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a31cddc079bf412d922d1baf0333306b : SUCCESS in 6m 06s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fbb5e07902684258a8ec6277473426be : SUCCESS in 6m 56s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/8c591c17c8e241b18afbc2a8d8a5c3ba : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356f9ea6031e43ec8bc5624949c11f3a : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c840650a8d9476caef9ad11ee4b3366 : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4148c20367f44c8b87b9a3426bef6c07 : SUCCESS in 1h 02m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2ac4ace679ac4d8397d5b6df9d4d7c6b : SUCCESS in 1h 18m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4d0f2ef0846c4d4f86387ed0f62a3dbb : FAILURE in 1h 33m 00s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a5de169c9fd34fe9a620bae3d1ed0893 : SUCCESS in 30m 38s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ee6ec849fcb94ec7a19156abc362fd3a : SUCCESS in 34m 45s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/fef328cb558b48f19e8baa8b25f317e0 : SUCCESS in 20m 35s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ab599b35c3704e5bb504914d0b0400ed : SUCCESS in 21m 00s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b1d12dfcec944d6689ce955113d94f10 : SUCCESS in 30m 40s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0ab45523686743ed8f02a0cdab9c38e0 : SUCCESS in 26m 21s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/533a8bfd7a384ec8b0de768cdebd0316 : SUCCESS in 57m 15s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/63544dc176f349ba91993c81ed9ebbf3 : FAILURE in 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5af9c2a22a8fd4a51351f13229c89934d3947fa","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 22:46:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e2fe3f9db1e683a85715a2ebade108efeaca1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:46:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de1045bb58ffcef4146a768e69749e7240f512f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:17:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7245b4c0ae94ccba627017b3de05b54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef213b16bf364d4d8e7c64346bd2f540 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/206daef6232b478ca8a699a537597cdf : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee666288c2444f0d8247cd1c06a34e66 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa02d1e33f554083bfd9dddef397f058 : SUCCESS in 6m 45s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/44781229e9574099ac70c9b8be8039ec : SUCCESS in 56m 51s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5dacd78dfa2d454fb3ca17995378099f : SUCCESS in 4m 09s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/45ab1147315c49a79f63a66341ea0925 : SUCCESS in 7m 40s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6a9d77382e5342729f57b4f9cfbda9e5 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40895fa2cf6a4d64835b54b21fdff2cf : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b184b11d41324816a27c08f21ab4f4f1 : SUCCESS in 4m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6af92b2159be473ab3c6dc11cb3028a8 : SUCCESS in 58m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f87775d163dc4f4b8036366daa272d90 : SUCCESS in 1h 25m 27s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/347cfa9f65304211a15e412010fced9f : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c2b0b33aad1a0ce82829833649af2b69b519bf34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:17:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1ef60271f925cbf1f1eeead96766c75772ace3d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:18:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3114abb50884c21a9c90ebf59154cc6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2748e9cbd62a407fa5d76bd9034bda2d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ed09f090644a08ce3c5ed565dc5d4848970b478","revisions":{"5c55e51ba57e880149b81fc4da36a10f9b961622":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/1006237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/37/1006237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/1"}}},"commit":{"parents":[{"commit":"f1fb8eaf896641e8ab59d088c715120730781dfa","subject":"placement: Allow empty commit on resource class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1fb8eaf896641e8ab59d088c715120730781dfa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-25 14:46:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Add support for allocations","message":"placement: Add support for allocations\n\nChange-Id: I09e6443938531c68bb66f9c2ba563aa2afdb318a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5c55e51ba57e880149b81fc4da36a10f9b961622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5c55e51ba57e880149b81fc4da36a10f9b961622"}]},"branch":"refs/heads/master"},"7ed09f090644a08ce3c5ed565dc5d4848970b478":{"kind":"REWORK","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/1006237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/37/1006237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/37/1006237/2"}}},"commit":{"parents":[{"commit":"a0a2eefac605014c991b994b5cd97d785e58f6b0","subject":"placement: Allow empty commit on resource class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a0a2eefac605014c991b994b5cd97d785e58f6b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-25 14:46:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:48:48.000000000","tz":60},"subject":"placement: Add support for allocations","message":"placement: Add support for allocations\n\nChange-Id: I09e6443938531c68bb66f9c2ba563aa2afdb318a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ed09f090644a08ce3c5ed565dc5d4848970b478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ed09f090644a08ce3c5ed565dc5d4848970b478"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
