)]}'
{"id":"openstack%2Fopenstacksdk~1006238","triplet_id":"openstack%2Fopenstacksdk~master~I6837a60c9e00d43e512c598407788dae7f8adbc5","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 00:17:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6837a60c9e00d43e512c598407788dae7f8adbc5","subject":"placement: Add support for bulk create of allocations","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-22 00:19:43.000000000","submitted":"2026-09-22 00:17:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006238-placement-gaps","meta_rev_id":"d76782cc9f64a51d15a568bf8218479f9c445bcf","_number":1006238,"virtual_id_number":1006238,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 00:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 22:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-21 22:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:34:03.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"4cd10e56c07e794a3896bb214e4d8cfb16566242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc06daf22121bc075eaa12c30bf2b5c474f72243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:46:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18eb2bc5951a42d09d14d809ed20a753\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e478a5010e4abcaf67996087bb1df8 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96d44add95be4c24a95b096432484705 : SUCCESS in 7m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03a291ee688e46d197d1bc770741db84 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0236522d98b944539df01c1fae0e16c5 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07b225b4c4404ef19a3fccac36bff8a4 : SUCCESS in 7m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ee568071133a4099b57f3c1045c5737c : SUCCESS in 5m 08s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bff97eab39a9454196a25e37bdcba147 : SUCCESS in 33m 21s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7648e6d57734489192ed3dc4ee268fa5 : SUCCESS in 4m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2da4c5cf1e304d069c49b40e0085bb29 : SUCCESS in 5m 32s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1daf5583ffdb49af801499146f323247 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8333781f36b44ab8d43b3c44b9677cc : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a153d3b221438b87d5ede39d6872e6 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a99d57321fb42a1a2ca5df1e17e4a47 : SUCCESS in 32m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/84a04b0c07214dc88cc20acdbfc834d3 : SUCCESS in 56m 42s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/86963c1d2ce8405695475c92a7560370 : FAILURE in 1h 25m 03s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/6ab27f0f6a3d49e7a40b95f8973c5c2a : SUCCESS in 49m 29s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d29fb95481604a0fa6211b7528dd7924 : SUCCESS in 38m 27s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f53907b56c2145068860c21f1e02c2d8 : SUCCESS in 30m 48s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/4e57d2d9c68d4c5eb78162bb4beab5e3 : SUCCESS in 19m 56s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/88fd3fbadc0f491182f5ed3ff210fac8 : SUCCESS in 20m 43s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/290594dab2164373ac0f8317b8fa8765 : SUCCESS in 27m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fccf800cf5fa40e480aaba6bf389e58a : SUCCESS in 42m 40s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/08aec0b1ffcd4405bfad7f319e3ade96 : SUCCESS in 59m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ab76c946ce1e2222630849a36f381980ed4119","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:34:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b92558bbdf3bb35d04dc96191ee14d21b6c1804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8fc45fdd91bedcaba6f0efc606e702f202437033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/036c436a769d490a9b1d0e38e82fab5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a9c15735534bef975fec4c700e21bc : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28514151c9204759b0652b594ed3a9a0 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f762ca49b6c411f87b2b795d3054b95 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75b17f7f72c14025a418fe913e929f05 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f733a3b9bd04c2885f57c8609dc01bf : SUCCESS in 5m 43s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6e6c69dd9ebc49419ff734edc959edb7 : SUCCESS in 6m 16s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d77cb56e34bd41e781938212793ea618 : SUCCESS in 1h 07m 35s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/76410f99fc284cfe8a8c4c278321338b : SUCCESS in 3m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ef67c1bc8d2c40829165117e3c19b869 : SUCCESS in 4m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1d9be8f416d347dabc54c161156af44c : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed92818a8e7645cd9e085626b79d1524 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5ce7df3bb54be3824ebe6d72c177f7 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07e6cc6b4c4f4577bba4714da15bfd06 : SUCCESS in 44m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/19a343eca4da454aaf716feb5ccf059c : SUCCESS in 1h 00m 07s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ffcfcd2537174df0a885ce526b29c84c : FAILURE in 1h 26m 14s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e839ae3dd3c142b3a848d4ae77216190 : SUCCESS in 44m 10s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/012b6058722844c79cee60ddf0d3dfff : SUCCESS in 39m 22s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/0add609e19bb4e329763d69a8dddf459 : SUCCESS in 31m 33s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ec57df4822ff4eb89f2e59ad683d348d : SUCCESS in 18m 02s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0ffe7fcf586f47058dc50e8ffe1c5ddf : SUCCESS in 22m 05s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/79559098fbff48a780b576c1f324a092 : SUCCESS in 43m 44s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dc4dfedbdaa74bdc891ec802cdd11c2b : SUCCESS in 1h 12m 11s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c7e3765b3f9d4a83a863f4424b234cdd : FAILURE in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc7d66cb1c959bab550a5828ae4f77883f0c96f8","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 22:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba8a9c0cf939f84151f541abdeb82f72ccdf4ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:46:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0943f994e404fd9eb5da59413f9426e3759be19d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:17:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a587f4a3a95849a0b88730db98056ecd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beed2aa6df74458c9780568a015b8850 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59031f6044cb4b7594233e4fbc61db21 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/777705e370e6486b80c0401c407dd60f : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/460d91b4a0a346f286204ef733e6847c : SUCCESS in 6m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/16e4050a580a4ee9a9540f47a2ba6fd3 : SUCCESS in 53m 55s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/22f781c82a654ea9b5cc4d3f0bb489ee : SUCCESS in 3m 27s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1a802fc3c3044a5d99567192b9907636 : SUCCESS in 4m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/2da6b37d5a5a4633a661eba4e0754c63 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef5857c0b4a4be28df9ac4feceea9a6 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22879128aac343f8a3b053631e2ad3db : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5ce5604a324d10a5a1cd2f053b16d1 : SUCCESS in 48m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/63468e318e4a4bbd9dfe922f41842545 : SUCCESS in 1h 26m 33s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/37276893f3f14d3c9caf361d188e1d7e : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":2},{"id":"617016da188407529e5895e303aecbea8ca090ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:17:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d76782cc9f64a51d15a568bf8218479f9c445bcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:19:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22545a22719d4225be2103167957cfe0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56952cab10440efb62ef16103f2ab67 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b894cf7687be3ab161cfb66c75ef1b5064fa7f76","revisions":{"d06074f16f1ff6f8776380da2117e5c0768241f5":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/1006238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/38/1006238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/1"}}},"commit":{"parents":[{"commit":"5c55e51ba57e880149b81fc4da36a10f9b961622","subject":"placement: Add support for allocations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5c55e51ba57e880149b81fc4da36a10f9b961622"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 16:01:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Add support for bulk create of allocations","message":"placement: Add support for bulk create of allocations\n\nThis is kept separate since it\u0027s a pretty \"unique\" API.\n\nChange-Id: I6837a60c9e00d43e512c598407788dae7f8adbc5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d06074f16f1ff6f8776380da2117e5c0768241f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d06074f16f1ff6f8776380da2117e5c0768241f5"}]},"branch":"refs/heads/master"},"b894cf7687be3ab161cfb66c75ef1b5064fa7f76":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/1006238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/38/1006238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/38/1006238/2"}}},"commit":{"parents":[{"commit":"7ed09f090644a08ce3c5ed565dc5d4848970b478","subject":"placement: Add support for allocations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ed09f090644a08ce3c5ed565dc5d4848970b478"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 16:01:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:48:49.000000000","tz":60},"subject":"placement: Add support for bulk create of allocations","message":"placement: Add support for bulk create of allocations\n\nThis is kept separate since it\u0027s a pretty \"unique\" API.\n\nChange-Id: I6837a60c9e00d43e512c598407788dae7f8adbc5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b894cf7687be3ab161cfb66c75ef1b5064fa7f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b894cf7687be3ab161cfb66c75ef1b5064fa7f76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
