)]}'
{"id":"openstack%2Fopenstacksdk~1006241","triplet_id":"openstack%2Fopenstacksdk~master~I5fb0e1381112212d74962d9dcccb4d40b3dd245f","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 18:37:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fb0e1381112212d74962d9dcccb4d40b3dd245f","subject":"placement: Add support for deleting resource provider traits","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-22 18:40:00.000000000","submitted":"2026-09-22 18:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006241-placement-gaps","meta_rev_id":"5b65e9fec0e11a9f2c7cef3392cd588a6e6af78e","_number":1006241,"virtual_id_number":1006241,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 18:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-22 17:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-22 17:05:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 17:05:33.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"41ed1ef81cb48fa58c2a794659f29c3d6e253f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e4ced53ddc86adf3550c20112463db4ea635c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:53:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a127b6c3494a0480d4ef635c3d87b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3271dbc2cc42461bbedc2a81ee19519c : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca1c530e49b048429d8317af4d9ac1a9 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a784dc7b7c664b9187e190a6dacb08e3 : SUCCESS in 7m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c34f946a6684410c9b9017b72b45b19c : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ca30b588f474e9e8914b78e5b149892 : SUCCESS in 7m 31s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1ea5c197f3344963954d41f3329bf495 : SUCCESS in 7m 10s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7a72f0858f8a46a9a48be95521c943d3 : SUCCESS in 1h 05m 34s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4317c090a58f491197082216104a249b : SUCCESS in 6m 41s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ba77e0dbb4ed401193766cdd375bce8c : SUCCESS in 5m 45s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f81edd731c0344279fdcebe0f2ccae59 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f160a4df09ba4ed2839539911635c51b : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce60f08010d42a6b8f6c093bbb5a79c : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/701455bb6584472ab3f819b2ac1c0f3b : SUCCESS in 59m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/359a24eb5aa74961a3e9f4de30fa6611 : SUCCESS in 1h 20m 53s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e5ba0e02699947de9402f47faf2f850d : FAILURE in 1h 35m 50s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b2d0269f4be649cc90d43783d47f0f4d : SUCCESS in 43m 58s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e39dcdefd6fd4c4fbb05921b11d8a588 : SUCCESS in 35m 18s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/4f6338c9d03e413aba17e3cc4031fe7b : SUCCESS in 33m 00s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/54fe43b413724aff9fd7d308c42e6ffb : SUCCESS in 30m 40s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/70646f9a0b994c67b51c53721c275650 : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/907de26cd1984fcfa1696d69d88ff3be : SUCCESS in 45m 42s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3b36086f1f574164abcbfcd285ae5070 : SUCCESS in 1h 16m 47s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/17e1892a3cdc49099c8a0436661ee3cc : SUCCESS in 1h 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b28b32a325dd540cb5b857d94aeb9f016b03bd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"913a2721adcb1a07a0e428c51632e8778e76fe8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:29:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8637a22c25ea43ca92a43111fadf7312\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a3ed65514c4d39b4d6f2e22921c760 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75bee17d01b44a49b3278e33953a8b1e : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c157fcd40bcd414881fddc20eaaaa65a : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dba1e2b1543e458f97194904501da9ce : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f6b5d5b8e7347ff8ae6b54a848bf42d : SUCCESS in 7m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d7bf5a0234524734b7c9a57bb38ef282 : SUCCESS in 7m 23s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/65c750bf187744be87947d98d5e1e5d4 : SUCCESS in 1h 04m 39s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2bf752b5da6e42ed922e04af53f2a5bd : SUCCESS in 4m 05s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8252efd8fe0446299f593ffda7540168 : SUCCESS in 5m 34s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/67eae37a9d81426aba7eca12ff4acd4e : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c9359de0c0411aa6453525511b7cb8 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ff2abcf57454afd9d0e5872f4c06f90 : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c25fb08a139402db2c297a834bcbd32 : SUCCESS in 1h 01m 20s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cc464f582abd42829b96930a6dc825f4 : SUCCESS in 1h 18m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/6f3a386f163b4a19a9a35615288ac501 : FAILURE in 1h 03m 09s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d45c78b77f68425db2bcd8c6a263d6e3 : SUCCESS in 41m 00s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/97920b4a82a8409698a6675c0d576048 : SUCCESS in 28m 11s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/827855d91c124b68bcdd4bf602687d17 : SUCCESS in 20m 51s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/856c04306a564ecbb069545ddb1e7d3c : SUCCESS in 28m 24s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ad00d2b0ca4d4a5cb3a48a3053b24f5a : SUCCESS in 28m 39s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7ae818d7820e4fd397c1acde0287ee9e : SUCCESS in 47m 41s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8596ccd07986488c98af225792281a2a : SUCCESS in 1h 20m 34s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/385c133c30964082b9c02b75f9d0bfc3 : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86f5d35e125ec4006bebfc9f35550fca9be151d4","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-22 17:05:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dec345ffb69fd4e5edfcdff27a560f49d5bf46fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:06:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e5cfefbc1fbb0a91d6025571dd76e580f570e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:37:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60fe106b0c74a6b9be38f33b42ee15b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f1124d6ece4956a1872da9efe3506a : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5fb0b9e96384873b6315137669966c7 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eba767a2eaf94c4289d50fe14775676e : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/636cdcdb061d4c159486aef7fc74d91c : SUCCESS in 7m 10s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8016d150d693456989b23ab6ac20fdfe : SUCCESS in 47m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/07d62f37d97e498c8b7dd9b0ed2c995d : SUCCESS in 2m 50s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/709cc32ec89642e8a6f524921d738640 : SUCCESS in 3m 36s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/72253310597947518947e9adf95bdc26 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be05155fb1c04f43a2a747588186c4a5 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8a4d024f962495ab64dd865d671edeb : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69e6ae37c3bd4ed3897918f7b6a178fd : SUCCESS in 53m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1b0d8f4c9b854603a21995b251c721fe : SUCCESS in 1h 09m 37s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1d96441309644502af9e92173e0b4d44 : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":2},{"id":"49b9c75febe2f1d13d20e27dd61e70109f74f3dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:37:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b65e9fec0e11a9f2c7cef3392cd588a6e6af78e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:40:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/995a1a9f68324bb2804143850b068e9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1253567a71849adae5a37764d0a57fa : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee1a639f7f7f52e431b1462992e0b32a0cfe6425","revisions":{"f3143def0658306a915eab572275beef1da3709a":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/1006241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/41/1006241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/1"}}},"commit":{"parents":[{"commit":"88cb2e6a41a5ac72818be226c5112497926f1a64","subject":"placement: Add support for allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/88cb2e6a41a5ac72818be226c5112497926f1a64"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 14:47:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Add support for deleting resource provider traits","message":"placement: Add support for deleting resource provider traits\n\nThis was missed in change I42d823beb608d6172141832686995ee4eccf8fd5.\n\nChange-Id: I5fb0e1381112212d74962d9dcccb4d40b3dd245f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f3143def0658306a915eab572275beef1da3709a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f3143def0658306a915eab572275beef1da3709a"}]},"branch":"refs/heads/master"},"ee1a639f7f7f52e431b1462992e0b32a0cfe6425":{"kind":"REWORK","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/1006241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/41/1006241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/41/1006241/2"}}},"commit":{"parents":[{"commit":"e8f4ab18b9011a1aae88f4134c517c92c7ff0e27","subject":"placement: Add support for allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e8f4ab18b9011a1aae88f4134c517c92c7ff0e27"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 14:47:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:49:35.000000000","tz":60},"subject":"placement: Add support for deleting resource provider traits","message":"placement: Add support for deleting resource provider traits\n\nThis was missed in change I42d823beb608d6172141832686995ee4eccf8fd5.\n\nChange-Id: I5fb0e1381112212d74962d9dcccb4d40b3dd245f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ee1a639f7f7f52e431b1462992e0b32a0cfe6425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ee1a639f7f7f52e431b1462992e0b32a0cfe6425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
