)]}'
{"id":"openstack%2Fopenstacksdk~1006242","triplet_id":"openstack%2Fopenstacksdk~master~Ic37bbf03d77987d65b0c79213a4efbf915052b5e","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","hashtags":[],"change_id":"Ic37bbf03d77987d65b0c79213a4efbf915052b5e","subject":"placement: Add support for bulk update, delete resource provider inventories","status":"NEW","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-21 17:35:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5faddddc1b795bbf9d4e93b979d61387b3cc49d8","_number":1006242,"virtual_id_number":1006242,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 17:35:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78f217136fb6be02281fd9088fd79ae98e416a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d9de26b2120bc625b167b38689e444ac64d7c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5308235bf8f24033a53acbfd70fc0892\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e5947d815a47debc4bce9691339cbd : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8001f30d4d34409b6e013d00d26f0fd : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f61e63003114e579a0541ffefe26e9f : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d50e1c31909b49568a39dc2232453d3e : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00d287246cd6447692cbcb64297c1fc8 : SUCCESS in 7m 24s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6e0ce37dd2fa4f52ac8e9f67fa988a86 : SUCCESS in 6m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/198629dca6854f4cae2a04dd9da7a59b : SUCCESS in 1h 03m 59s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/257a79adc7204ca9a3e085878183d3c4 : SUCCESS in 4m 36s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9647391c11f5473290a0ed5d9cb56872 : SUCCESS in 5m 15s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/80a10553f59d4c4789f82300373746a7 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0812c70a66364469aaa6502e5b996639 : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd4f667c336346dfbd98befe49d21966 : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99e1e469b2e34570a780ae9ea8874720 : SUCCESS in 32m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/68a633194a5046ec8e6cd8b716330e22 : SUCCESS in 54m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/43b9cb21a678429c843b31f717d7c543 : FAILURE in 1h 17m 16s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cbdd39a561684503b261055948c2987c : SUCCESS in 41m 05s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ef31bc01cd974a0d8b5c6321132250cf : SUCCESS in 37m 27s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b24f7a4df49b45f29905186f87dc9a5f : SUCCESS in 30m 06s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/936803f5fe11425083aa1a39ec540707 : SUCCESS in 30m 46s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/88fb231e48c148e3a332d4ba09bd2718 : SUCCESS in 33m 47s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ce4f44a7819a4185964a8effd388dfc8 : SUCCESS in 23m 34s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/66eab4bca4c447b0b081894bb06fcc01 : SUCCESS in 1h 12m 15s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/bf9925c19cf342cb960f6b04650a31df : SUCCESS in 1h 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e93f18487bf4f728b857c7fa9e787b75d4a437dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5faddddc1b795bbf9d4e93b979d61387b3cc49d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:35:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d73b37233d4575a3c46efc7f852512\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6fd26d9f3f428fa4eb613058454b03 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b4d246743fb4b4f97c50d7f4be571f6 : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/598402ee74574aa0adaa2daee88b7497 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c5b7879f8454677b4cafee4fa022c5f : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0a152a36c3d4a2d84b2447996bbe70f : SUCCESS in 7m 33s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ad09ee1b193543c8aed04142774cdd7c : SUCCESS in 7m 25s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/673325c9f9ff4a959b39046e44df8581 : SUCCESS in 1h 05m 17s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0929240e3da84118922a809081dbb990 : SUCCESS in 6m 07s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/68452130d0f346d7aa87659ee8926425 : SUCCESS in 6m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/fa1fcbb37ecc4b88b7603a4d366a9723 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b8cbb088624393963695844b5edc0a : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8336e7fa46a249769d9e0976480ca46b : SUCCESS in 6m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/416475b3197a41a4a98413f1c3648cc8 : SUCCESS in 1h 04m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/21acec0c1ec94e5c8bbf4b6fae945d24 : SUCCESS in 1h 21m 10s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e263a77830e2456cb7f8273f774ae1bc : FAILURE in 1h 36m 30s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/caebcd8cfd464ba08aa0380da964abdb : SUCCESS in 41m 31s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/60de69e2427a4169be9147ddab3ac5da : SUCCESS in 26m 47s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f0e19c7dc0cd408ba8c4ab47e4c003f5 : SUCCESS in 36m 15s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/78ffc7134816400c9157dee69715d91a : SUCCESS in 18m 16s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/7192ca2cb25e4c09886337888ba9a904 : SUCCESS in 32m 55s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e38286c0c1ae4afba230dc5242a06ce8 : SUCCESS in 48m 13s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4203607a14ba4d0e909b957b1060b113 : SUCCESS in 1h 15m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/e81b0bdae2c4447e970af29f3382d6ad : FAILURE in 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ce659ca5a9ccfd8440b43c18fb8130da502f154","revisions":{"abd9b738152a41cef412e6884efa9fe85e7021e4":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/1006242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/42/1006242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/1"}}},"commit":{"parents":[{"commit":"f3143def0658306a915eab572275beef1da3709a","subject":"placement: Add support for deleting resource provider traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f3143def0658306a915eab572275beef1da3709a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 14:57:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Add support for bulk update, delete resource provider inventories","message":"placement: Add support for bulk update, delete resource provider inventories\n\nThis was missed in I5ea5d0a477147e1a4e30b428f17b44807253deaa.\n\nChange-Id: Ic37bbf03d77987d65b0c79213a4efbf915052b5e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/abd9b738152a41cef412e6884efa9fe85e7021e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/abd9b738152a41cef412e6884efa9fe85e7021e4"}]},"branch":"refs/heads/master"},"5ce659ca5a9ccfd8440b43c18fb8130da502f154":{"kind":"REWORK","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/1006242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/42/1006242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/42/1006242/2"}}},"commit":{"parents":[{"commit":"ee1a639f7f7f52e431b1462992e0b32a0cfe6425","subject":"placement: Add support for deleting resource provider traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ee1a639f7f7f52e431b1462992e0b32a0cfe6425"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 14:57:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:50:07.000000000","tz":60},"subject":"placement: Add support for bulk update, delete resource provider inventories","message":"placement: Add support for bulk update, delete resource provider inventories\n\nThis was missed in I5ea5d0a477147e1a4e30b428f17b44807253deaa.\n\nChange-Id: Ic37bbf03d77987d65b0c79213a4efbf915052b5e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5ce659ca5a9ccfd8440b43c18fb8130da502f154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5ce659ca5a9ccfd8440b43c18fb8130da502f154"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
