)]}'
{"id":"openstack%2Fopenstacksdk~1006243","triplet_id":"openstack%2Fopenstacksdk~master~I1b2ff077f72e352f80bbfb6b7525a4297ad900c3","project":"openstack/openstacksdk","branch":"master","topic":"placement-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-23 04:55:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b2ff077f72e352f80bbfb6b7525a4297ad900c3","subject":"placement: Add support for usages","status":"MERGED","created":"2026-09-18 15:12:08.000000000","updated":"2026-09-23 04:57:51.000000000","submitted":"2026-09-23 04:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006243-placement-gaps","meta_rev_id":"f03875c1f12b6b0474d10d76c6d39ee929cc418f","_number":1006243,"virtual_id_number":1006243,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 04:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-23 02:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-23 02:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 17:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 02:07:11.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"128658dc0fefe1e22dfef4dfa2778f9e9b00af01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a54d1660f0aaf298df556c052f9a89f15c6cc4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3488cc3ead6542229198902299dadcbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46757c6d4fd143b58d5a1f89158a5834 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e242dfc9ed39446d9991513b87a42df4 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5eafeef383444ebab8df1250ba7bd56d : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b36ac0d3e8a495587c939c901f0a39a : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70bc0fd0a47f4fcf91d238f8e56003f4 : SUCCESS in 8m 44s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/f5c84cfc2f4047c08f767b457865a59b : SUCCESS in 7m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f4119775548b4f78ac65464ac717c9ee : SUCCESS in 1h 04m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/19454885184c47f5b71e42bd301f3c5c : SUCCESS in 4m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dca0b68f1f1744e9a5d21add0228d43e : SUCCESS in 5m 30s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/3cac8c87a63e44e884c4cb36585d4380 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22be3b7854be4e85afaff0d435c953f5 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1d5de7fcf014b6ea3337cb4caed4302 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/461f8cc28ef94eccb5b8868d5866b11e : SUCCESS in 1h 06m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7f66d214df854f85a89a25cc497363f0 : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8f180333eedb4157b0da408fee1d1d8d : FAILURE in 1h 38m 57s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7e15d2efbde04168be2ff8b66b59d3eb : SUCCESS in 41m 33s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/81002d48131d46a1bbcf6a95bcb1b732 : SUCCESS in 30m 01s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/8cb0a057567b47a5a62580accdd52ccb : SUCCESS in 36m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/47e5cc580b1146b892e6ad94cec8974c : SUCCESS in 19m 59s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/5c3bf9bfea2e4e8bb2c6f28dd477dc80 : SUCCESS in 16m 20s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/196ab8b58ad846d5bb3895da4c7ca9bc : SUCCESS in 55m 34s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5bf1d0c57a7c4b29b2ca59b86e5b2451 : SUCCESS in 56m 25s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2a02347fd47d468a84dda4f15e720a83 : SUCCESS in 1h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c9ae318ecca1ab4449a39e8f5f43aa5becaf332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 15:50:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"59687cdb77c4076fc368a3a0ada487fc5cd8e9b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:24:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829113552d474281b4d518cb9c59e940\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddca7f33c184b4bae306e162fd17022 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4962b4c4afc24c698501c5fd5e589c19 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0b003daba4d4cc394503f462b46f6c3 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af7d37bb50ee447391d12e68048976f7 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/273958ff5d4a45d0b69c15040125b795 : SUCCESS in 4m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c48d044cb2c646c78d559fb7f898df77 : SUCCESS in 7m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/11e8ec059a414d49a58b5d1d1ed8209c : SUCCESS in 38m 02s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/846620bb2a2b426b9ac40ef0a68bc618 : SUCCESS in 4m 22s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1ff2f1d81cdb4927acb8245da8412a65 : SUCCESS in 4m 14s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/15c72daf0cbd402d8b77776430365b7a : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a16eb835c0a4bc583898a32109f350c : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09de2377d3a34dee8337489c732e20cd : SUCCESS in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2665f9264a5b463d92590275e354ac01 : SUCCESS in 56m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/11abf6372ebd40e2af02ce011a7cbbf7 : SUCCESS in 1h 23m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/589d0f9914e940b8a128c63035e7f8a7 : FAILURE in 1h 22m 15s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e6ecc0f515fc4deb9726fa8a450fde19 : SUCCESS in 43m 12s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2e295c51a2db438ca005e176fdef9bfe : SUCCESS in 35m 41s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/97223bc738fd42719b687af14479a846 : SUCCESS in 50m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/99c3e85dae064dfaa9e532f88c0e5ea1 : SUCCESS in 20m 19s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d8b6c85b92b640c5ae010964c1d45c43 : SUCCESS in 18m 38s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/dc377fe42b8047aa907f0686c56d1faf : SUCCESS in 45m 00s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9ca3165240444580b856d3110994fb29 : SUCCESS in 1h 12m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a9def9d0da9e49828b8afbdbaebe0dbf : FAILURE in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be63212a495ab72e2ec3fb2fb08c96cdf9fc7e5d","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-23 02:07:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f57698497872ca63d678229733e17f380768d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 02:53:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b33a95e16f99f1e6e702b7e48af99bc58b5d796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:55:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7ad7f2e8244e288538a7bba9627926\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41e10c0365e4eeca0f0f37e4357da25 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e332959e3b1462aba9c174cccbc4a39 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de1d27ea6a4c4a9188a93ed9576e593f : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a8caff12645d4b568f0917c7a0948718 : SUCCESS in 58m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b15ceb0a669c4ba5814b2934dff5e8af : SUCCESS in 4m 06s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e958fb332a3347ef957ba0bfdcdbc319 : SUCCESS in 7m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1270108598d54e1ab627450d6f90b2bb : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ebc07a4dbd405a8788c821afb2df85 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ca05ac1426b435aa174c7da3148b962 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/379d727e69e8464ab3d1dec5ea17e17a : SUCCESS in 1h 01m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/39f7367824a9430d87fc70aefbe86fad : SUCCESS in 1h 17m 50s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/50fc82377f5243e48898d96deeb1be9f : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3379a4598390a4ef09e94296769c1158cfbefd26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f03875c1f12b6b0474d10d76c6d39ee929cc418f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:57:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4a3d547c3a4d27b92458fb99dd33ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf1fb6412384f529d3d07e95ade1ddd : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f86dd156133a0ddc5873a899754179787d33e356","revisions":{"ccb8b4fba9ab59d0b26bf1d87100e49b0fe68d5b":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:12:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/1006243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/43/1006243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/1"}}},"commit":{"parents":[{"commit":"abd9b738152a41cef412e6884efa9fe85e7021e4","subject":"placement: Add support for bulk update, delete resource provider inventories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/abd9b738152a41cef412e6884efa9fe85e7021e4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:21:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:39.000000000","tz":60},"subject":"placement: Add support for usages","message":"placement: Add support for usages\n\nChange-Id: I1b2ff077f72e352f80bbfb6b7525a4297ad900c3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ccb8b4fba9ab59d0b26bf1d87100e49b0fe68d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ccb8b4fba9ab59d0b26bf1d87100e49b0fe68d5b"}]},"branch":"refs/heads/master"},"f86dd156133a0ddc5873a899754179787d33e356":{"kind":"REWORK","_number":2,"created":"2026-09-21 15:50:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/1006243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/43/1006243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/43/1006243/2"}}},"commit":{"parents":[{"commit":"5ce659ca5a9ccfd8440b43c18fb8130da502f154","subject":"placement: Add support for bulk update, delete resource provider inventories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5ce659ca5a9ccfd8440b43c18fb8130da502f154"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:21:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 15:50:32.000000000","tz":60},"subject":"placement: Add support for usages","message":"placement: Add support for usages\n\nChange-Id: I1b2ff077f72e352f80bbfb6b7525a4297ad900c3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f86dd156133a0ddc5873a899754179787d33e356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f86dd156133a0ddc5873a899754179787d33e356"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
