)]}'
{"id":"openstack%2Fopenstacksdk~1006267","triplet_id":"openstack%2Fopenstacksdk~master~Icf7641fbaef864bac4fa614bdc51450c297fab37","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Icf7641fbaef864bac4fa614bdc51450c297fab37","subject":"docs: Add missing service guides","status":"MERGED","created":"2026-09-18 16:11:39.000000000","updated":"2026-09-21 15:05:07.000000000","submitted":"2026-09-21 15:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006267","meta_rev_id":"86e628a8a63f24cf47f86b05110a2cda3aea4ade","_number":1006267,"virtual_id_number":1006267,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 17:58:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad240646ab479ed0c9525eb91d74a008e27bc451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:11:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"673c33e68931286c35a2ef552171dcfe4deb9b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:58:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24d10922bac447fea3eaf3d61384b706\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e8a9fcef3bc47bc83a1d8b33b01afce : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0623ca00462b46b982444acadc6e90df : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f775cf620f9c4616b59aa28f220ef240 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f503d6a40de9482ea50b3460a192338c : SUCCESS in 7m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5fc9ad3a9754dc2ade57e5160844e15 : SUCCESS in 5m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b826f1a6d15b4caea5f2075b8a287abd : SUCCESS in 4m 35s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/38d0346e54cf4a45b52589cdfdcdb930 : SUCCESS in 36m 25s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f706ddff69b4406fb4b8a72417169fa3 : SUCCESS in 4m 02s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a4e378a30ee347188d8807b673ad93b5 : SUCCESS in 2m 53s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/dc890d477cf241ea92f4a43cc77b3d34 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f178301d8ede428ea84ad01c3ba1ec60 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e03b8f5f9c2d42cfb4062af1d4092748 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d50c7527685b4978ad552873c73aae43 : SUCCESS in 50m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf0c4e7f63bb442db92e113ccaa44c01 : SUCCESS in 1h 23m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/53d223b8398a4317958dcdefafbb196c : FAILURE in 1h 24m 29s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2ce6a03ab1414b32b17113aee6bbe5c2 : SUCCESS in 43m 05s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/995e637905f94401b3d9b8b3adc04aec : SUCCESS in 34m 32s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/010d6a12aaa147a0a6adc703e71ff30d : SUCCESS in 29m 51s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/59c60573ad5e4e41952a2d252fc0b102 : SUCCESS in 28m 23s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ee10f214e4764a45a1fd195172dd886a : SUCCESS in 19m 21s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/65a59cc6cfc3406fa94252f8f56c2f48 : SUCCESS in 18m 40s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b5ab08943e734e538be6c3eb190ac453 : SUCCESS in 1h 10m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a29527a0d5e34618a5837286bb7aee3a : SUCCESS in 1h 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d7ded8a13bd0da4e2ed350fdc89adacb0bec9f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:41:39.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"90501ddefdbbd1937ad912bec93e7859c89b0910","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:41:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35ec471a0ea2476d77ab249e836915c7e2396ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:19:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4454f8cd7c3e4d6bb7c4000903ddad9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d34c924edb4f3981d232434c5fa625 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8ade20aa7244e7fa36b11d988da9d40 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2770a87eb6144b6a0fead423afcc5c2 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ede8ad65e29644d099279c16a6476e12 : SUCCESS in 6m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56f4a8ba5e6c4f01a188b2cba22c8fdf : SUCCESS in 4m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7bea0bb6c9f1441dafc9e155fbbc6704 : SUCCESS in 9m 19s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2c1511a068284e81810c94fd5245abf8 : SUCCESS in 55m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2c540be7352341cc97a24d0144631d59 : SUCCESS in 6m 33s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6b0342647ad5407ebe1d2a65b369da0d : SUCCESS in 7m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/44a641fb19ee4e68aab72366a675abe0 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b56fc713e04aeea28283fbe161bc47 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f2a132f7e33423a8d376c6b3cc82012 : SUCCESS in 6m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95a08b7a8f394c87b99a63015d8d1e00 : SUCCESS in 1h 01m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d06dc8dc6fe644aca1335e07e99d83f7 : SUCCESS in 53m 37s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/703c15f7b6514a6b9b06bf378c600ecb : FAILURE in 1h 30m 38s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9ff71641846742908f7ef4f7dffbec59 : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ceec943a8e3d4dcbb4ccb30ae1aaa72e : SUCCESS in 21m 59s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/67b7aa76c05e45818eb3320b7585d800 : SUCCESS in 15m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ebe004d08b5c422ba69d94a18c836b9a : SUCCESS in 32m 15s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/88f66c3ed0d64cb7a25cd8e7f86cb46c : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/643ecdc447434ce2989abf0489f6ebe0 : SUCCESS in 40m 14s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/748782cab2cb4f68b6743ad09914de92 : SUCCESS in 1h 20m 17s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8503240bf21e4d8d94f806b1afc47d21 : SUCCESS in 1h 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d11999301cda38aa8db6a32695fe6a9af3fd9015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:20:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b9fab4fc002004af030b518f3a127458caeccc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc74387601b4ba3912f64c1e1bbd2f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a912c90c5648dab3be1f4261da6889 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66ef175d30254fc8b71c2194ab9121bc : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63483ca734d34ce392541f57beac9eca : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2421543201ff498ca9b325d3b3157a1e : SUCCESS in 7m 30s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/abf1eba1be5148f89ff694866f8f713e : SUCCESS in 34m 17s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/53b333c3cf1041adb8c0c3a0e5d1bee2 : SUCCESS in 4m 37s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/54dfb52d843a491a93d414088095642c : SUCCESS in 6m 17s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/66f1e126706c4740973951938905eadb : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a605777c60bc41cd981826467deac921 : SUCCESS in 16m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b50fe733c142fa938f3c49ee7c364a : SUCCESS in 6m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9519d538734a4d6fad96a24d746d2fed : SUCCESS in 1h 02m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4b9d78e3e4fd419b95c64f18f83e192d : SUCCESS in 1h 32m 31s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/bec6143337784daebd79b8ed81ef599a : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":2},{"id":"65764c3846c9a69910a716aa7e80f350f4387f53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:03:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"86e628a8a63f24cf47f86b05110a2cda3aea4ade","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:05:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3aaf2aaa41e43f08e4738df844a2e97\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52996f0578dd4f798d353f220123e66e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8c90601e0d6827eee9cd3d33397da46d68f07f6","revisions":{"9c5cc68baa97f43390ac2f159c341f23d783efd2":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:11:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/1006267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/67/1006267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/1"}}},"commit":{"parents":[{"commit":"29620beef08f3bdc6dd73d5b3f42c69ba60c2f4a","subject":"docs: Add proxy, auth plugin guides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/29620beef08f3bdc6dd73d5b3f42c69ba60c2f4a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 22:01:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:38:41.000000000","tz":60},"subject":"docs: Add missing service guides","message":"docs: Add missing service guides\n\nChange-Id: Icf7641fbaef864bac4fa614bdc51450c297fab37\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9c5cc68baa97f43390ac2f159c341f23d783efd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9c5cc68baa97f43390ac2f159c341f23d783efd2"}]},"branch":"refs/heads/master"},"e8c90601e0d6827eee9cd3d33397da46d68f07f6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-21 11:41:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/1006267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/67/1006267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/67/1006267/2"}}},"commit":{"parents":[{"commit":"29620beef08f3bdc6dd73d5b3f42c69ba60c2f4a","subject":"docs: Add proxy, auth plugin guides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/29620beef08f3bdc6dd73d5b3f42c69ba60c2f4a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 22:01:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:41:39.000000000","tz":0},"subject":"docs: Add missing service guides","message":"docs: Add missing service guides\n\nChange-Id: Icf7641fbaef864bac4fa614bdc51450c297fab37\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e8c90601e0d6827eee9cd3d33397da46d68f07f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e8c90601e0d6827eee9cd3d33397da46d68f07f6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
