)]}'
{"id":"openstack%2Fopenstacksdk~1006268","triplet_id":"openstack%2Fopenstacksdk~master~I77ff355b27df8f687879189671f2e64d99c42c66","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I77ff355b27df8f687879189671f2e64d99c42c66","subject":"trivial: Remove dead TODO","status":"MERGED","created":"2026-09-18 16:11:39.000000000","updated":"2026-09-21 15:12:54.000000000","submitted":"2026-09-21 15:11:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006268","meta_rev_id":"967eb6f1b13c6d557da1f6233c0f36821f2619a1","_number":1006268,"virtual_id_number":1006268,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 17:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6c162974829ea693c8c419f6b8ff45038128d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:11:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b97f3de2624c4e180c9556a3566b9ad1db175ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:50:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0377025856644a882df22c2a5942040\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3794dbacc64e9a9bcc62349b5403ca : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e740fe245d6c4ddf94bfd77ed2f5f9eb : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cd45583876a4ef495c408047237f7e3 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3e789b0a7bb46ecbb057df46b1068e6 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a528f594eb864d0993ccb38b87a0c8c0 : SUCCESS in 7m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/eb454bf4673f47138f0d37a7df516041 : SUCCESS in 5m 53s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/220b4dbd078d4ba18bf749653bab0142 : SUCCESS in 1h 05m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2c379c88cf414297b366d018fce86b4b : SUCCESS in 3m 44s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/582b506405bc4e9891b077a128d54041 : SUCCESS in 4m 26s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/958be92d1c134079b8903625759c0582 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd80965306e748d2851e24f6f832528a : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3d355abbc784a9b902ecd816792240c : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d6907608f084f4c86b84696edde78b4 : SUCCESS in 35m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6df918fdb4ee4a9c8d9d0c22ab64465c : SUCCESS in 1h 18m 25s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2d0b2633af7644f286e108ab9669c3cb : FAILURE in 1h 29m 25s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/409a4ea9da6b49d5be4f2691da374548 : SUCCESS in 41m 47s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/637626c983fc4e84a7462421f164de71 : SUCCESS in 29m 59s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/3824856991c04aaca761d9397fc05335 : SUCCESS in 22m 35s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/d6ae3c02010149aebf65365dde1cbfdf : SUCCESS in 29m 40s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ed93c0f9d0554712bac0a0c6992053cd : SUCCESS in 31m 49s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cd40610c8f494e83a64e7df2124d9b4d : SUCCESS in 42m 35s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/de98c4b0c5594c5d9b761f3922c1d6d5 : SUCCESS in 45m 43s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c46ad85d828c4c8f96b23409cf735fa5 : SUCCESS in 1h 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3b3e83f7c6eb807081607d3dbffefe644a91734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:41:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"06fe7dbfe2cf1c0fb58829bee18c8bea5fd278fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:41:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c8d8d0a2f63bbaabbf80f01440945d95813261cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:42:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"defdb9d544defed93e1441db13a012a8fa475bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:26:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e206ee5aaddd43abbf37ae1a613a0e2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3773cf7b284d4d84946ee5e34d5355 : SUCCESS in 7m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8533984a83bb463f92b9ee1f5d29e569 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b121961ba18342cfa064bb5dde1e0227 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9ab96808f934c2cbabfd819a4cefbc8 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42e7e26b4b2f42b989f0b46c5cf8a487 : SUCCESS in 6m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5d7c0fad9187415e8c855cca49086637 : SUCCESS in 7m 21s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/53c707a6d3ac471880a0887281f4bf9f : SUCCESS in 1h 05m 12s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bc4e0e0867ce41f9b1de0b35b6948036 : SUCCESS in 4m 01s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b2da36c6cf3e4eb7926111e841849ac4 : SUCCESS in 6m 44s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7691de3acf894f9a8a8fad4d8e91f543 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e22869064f4dfda4f410d8802188b4 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f168927afab04dd5a5f49411dc1b9110 : SUCCESS in 3m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5adf5edecf3844deba3ad9b993d37b7d : SUCCESS in 29m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/023ef493dd044605a567ce8ced4992ff : SUCCESS in 1h 23m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/129ecbc4c15e45d99b36148c41f3ec45 : FAILURE in 1h 37m 01s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7e0fd39ec8204bb88a6ee45f0d36aadf : SUCCESS in 40m 50s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/785adf1df09847188f0efbadb5701db1 : SUCCESS in 38m 42s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/4f77c11c945647088cde5625ba94b9e5 : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/2601279631f84395bafeac7a8ec9e099 : SUCCESS in 27m 34s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/50c87cf8da314591b9434818397651bd : SUCCESS in 30m 20s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d50ced6d9e204499bf7c52534d1d35d0 : SUCCESS in 42m 56s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cf87f38857da4bf493518bde1e7c0f6b : SUCCESS in 41m 30s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cfbbb6050a7c4e77a7be2cfa33cf4a1b : SUCCESS in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14148edaafd719b320ebd60f081af00c9107af54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:27:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb8d8992d7a8fe69036244d2419fe77127ec44b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41634386caec4aa9b082ec362073dc02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a9a2e8a25a4424bd6be34777c8b0c1 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ae75526378444a18b08c7cc10c68bde : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ed9212b040d41a5b1d1401fa32f06e2 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94729d2d7aaa4965a7fdfc9b851e6889 : SUCCESS in 7m 42s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/efccfcd636b043ebb859887a2773a853 : SUCCESS in 1h 06m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e73404479f2d4e2ba13349dd9a11efe7 : SUCCESS in 4m 58s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/de8b84e7966840c180e35d8c907252c1 : SUCCESS in 5m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/938781566c1f4691aaef53695450b201 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f4f6d7f0e74638a3ee05fef3191b4f : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e7e89978bb14f1786954da07b656db9 : SUCCESS in 6m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae8c15ca11874d45bae11b801db538f8 : SUCCESS in 1h 02m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/797d9196dd154836b1286e50bfb14e4c : SUCCESS in 1h 29m 18s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3ec4ade2b648457c870bbcf383728ad6 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"bf8e77ae12017263effdf477fd040bd406112696","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"967eb6f1b13c6d557da1f6233c0f36821f2619a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:12:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829634071d8743dbadf2cbea49e234ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4825bd8a39d4937a2c3d6fe549673b9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98569713552bf455b768c43143551141fa70ab2c","revisions":{"46bf801917320972f62f0e620138c91d59e387a9":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:11:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/1006268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/68/1006268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/1"}}},"commit":{"parents":[{"commit":"9c5cc68baa97f43390ac2f159c341f23d783efd2","subject":"docs: Add missing service guides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9c5cc68baa97f43390ac2f159c341f23d783efd2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 22:49:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:38:41.000000000","tz":60},"subject":"trivial: Remove dead TODO","message":"trivial: Remove dead TODO\n\nThe issue in subunit was fixed a very long time ago.\n\nChange-Id: I77ff355b27df8f687879189671f2e64d99c42c66\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/46bf801917320972f62f0e620138c91d59e387a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/46bf801917320972f62f0e620138c91d59e387a9"}]},"branch":"refs/heads/master"},"ee7ff0367c8b11330b260f1bd854d4ff2b237db3":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-21 11:41:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/1006268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/68/1006268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/2"}}},"commit":{"parents":[{"commit":"e8c90601e0d6827eee9cd3d33397da46d68f07f6","subject":"docs: Add missing service guides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e8c90601e0d6827eee9cd3d33397da46d68f07f6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 22:49:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:41:47.000000000","tz":0},"subject":"trivial: Remove dead TODO","message":"trivial: Remove dead TODO\n\nThe issue in subunit was fixed a very long time ago.\n\nChange-Id: I77ff355b27df8f687879189671f2e64d99c42c66\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ee7ff0367c8b11330b260f1bd854d4ff2b237db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ee7ff0367c8b11330b260f1bd854d4ff2b237db3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9c5cc68baa97f43390ac2f159c341f23d783efd2","ours":"46bf801917320972f62f0e620138c91d59e387a9","theirs":"e8c90601e0d6827eee9cd3d33397da46d68f07f6","merge_strategy":"recursive","contains_conflicts":false}},"98569713552bf455b768c43143551141fa70ab2c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-21 11:41:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/1006268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/68/1006268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/68/1006268/3"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 22:49:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:41:56.000000000","tz":0},"subject":"trivial: Remove dead TODO","message":"trivial: Remove dead TODO\n\nThe issue in subunit was fixed a very long time ago.\n\nChange-Id: I77ff355b27df8f687879189671f2e64d99c42c66\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/98569713552bf455b768c43143551141fa70ab2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/98569713552bf455b768c43143551141fa70ab2c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e8c90601e0d6827eee9cd3d33397da46d68f07f6","ours":"ee7ff0367c8b11330b260f1bd854d4ff2b237db3","theirs":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
