)]}'
{"id":"openstack%2Fopenstacksdk~1006278","triplet_id":"openstack%2Fopenstacksdk~master~I6f57621599d70403501e770635ee3e595903ec16","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I6f57621599d70403501e770635ee3e595903ec16","subject":"zuul: Remove unused jobs","status":"MERGED","created":"2026-09-18 16:28:43.000000000","updated":"2026-09-21 13:03:17.000000000","submitted":"2026-09-21 13:00:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006278","meta_rev_id":"1dc5f9e462f5707a226039396546011f1f31078a","_number":1006278,"virtual_id_number":1006278,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 13:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 18:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92bade1dcb5139a70dc7e52f3e8aa8f21b01034b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef514025563b3c25723235a08ecd0af2eb07cbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:06:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f8eb455edb4d2e8a8ade74ab022be1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3b04a3e5544e4fa11716ae4a2f7828 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/874dcf4ce9244328aceaafb56033d5ed : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df53b32c59364ac0b57412629152d6ab : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f367ac1b14af4e69a7e63dd1bb4cd772 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43b8637350764ae9890567bf1d0671ad : SUCCESS in 6m 16s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/23fa8d64199b4b5587b5a03d2f515b87 : SUCCESS in 7m 24s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2956ad3967df4d09967f83b12d9aba41 : SUCCESS in 32m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ae2a6dd106e0445da5051c31b6413e16 : SUCCESS in 3m 25s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fa9b04f0a8e4456395ea544f2807dc7d : SUCCESS in 4m 19s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b059fc776c9c45c6a1724775567bc846 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b265894e864ecb937648439ee2c694 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3173a7bc3477461ba25e2ba3b504e382 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93613f80a4794673aaf6676260711497 : SUCCESS in 36m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/55fbc50b337542768ebe3daf6026473d : SUCCESS in 58m 27s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/0a7ab94ccdb84d6d8eda3ded3f02f786 : FAILURE in 1h 26m 23s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/aa8386bf5744486ab081a59e00b42bdb : SUCCESS in 23m 32s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c6a623cc5edd41e88ff15842993cb44b : SUCCESS in 22m 42s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/78afb9f3743e44d7aa5591a173307029 : SUCCESS in 31m 36s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0f6dba05929841f4a8adb398f612e4f9 : SUCCESS in 17m 22s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8ec4278c504148ecbd53de5716a4e342 : SUCCESS in 18m 47s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9c715085744d47af99a0cb40574173cf : SUCCESS in 22m 10s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/06350d372a444bf4ad197030158abcce : SUCCESS in 43m 49s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0a7be38d64d84c41be61b01fe7e5b056 : SUCCESS in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4529bbf46880f124e7f838cc323e217136a63441","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:42:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b641648b88516d8f9c89f0c814322cc451014de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:43:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b120dd02b05695d9f508df6e6ee5cbb8263abae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:00:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae6fe68cd494d448b169cbece255329\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf9508deb5f40f099f798219ef5a364 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/939e1b0686d6421c82a7ae59a55c5d58 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07d71504156c45d0ada5b45dd06c8657 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/405f84a118bb487787dd42a2c962fef9 : SUCCESS in 7m 33s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c8c985ac694144f39cde8a23fe70c15d : SUCCESS in 1h 05m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6f3e7d5ae8df4ab3bece782c75945bd8 : SUCCESS in 4m 24s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e7649a1655d34005aecec73bcda89bba : SUCCESS in 3m 56s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/c8f312d66ac84242828945bbebb1d54f : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9af625bfa484993a8447c34d0f9eb52 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/450f19234fc34167bf32d70f6f2111cc : SUCCESS in 6m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/048c0358b7a540f29a378cd4a7f29136 : SUCCESS in 43m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8bb04b75b3af4c63830b42eaf440a3b2 : SUCCESS in 1h 14m 23s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e30c66e2bd464c6687a471d8b074546c : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ea54484112cc62475195319e50a9933fc94b2be5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:00:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1dc5f9e462f5707a226039396546011f1f31078a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:03:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7affa8f27d8b452eb7a4a18e19e3c4ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cb0acf464d465e863c99a7cfe686f3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb5a6698f40e699c4b30c22d779d840583bc7751","revisions":{"eb5a6698f40e699c4b30c22d779d840583bc7751":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:28:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1006278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/78/1006278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/78/1006278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/78/1006278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/78/1006278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/78/1006278/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:20:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:21:07.000000000","tz":60},"subject":"zuul: Remove unused jobs","message":"zuul: Remove unused jobs\n\nThese have been unused since I99abe18e3d91d52fc580a3d75146d04ec3de775a.\n\nChange-Id: I6f57621599d70403501e770635ee3e595903ec16\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb5a6698f40e699c4b30c22d779d840583bc7751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb5a6698f40e699c4b30c22d779d840583bc7751"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
