)]}'
{"id":"openstack%2Fopenstacksdk~1006279","triplet_id":"openstack%2Fopenstacksdk~master~Iff40617dc4dcf87eb7034b1d47f2c298f31e0e42","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 11:42:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Iff40617dc4dcf87eb7034b1d47f2c298f31e0e42","subject":"zuul: Remove more dead jobs","status":"MERGED","created":"2026-09-18 16:28:43.000000000","updated":"2026-09-21 15:11:14.000000000","submitted":"2026-09-21 15:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006279","meta_rev_id":"df46402c10c8a0e26441591cca5a174ea38aac07","_number":1006279,"virtual_id_number":1006279,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:42:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 17:41:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5e89f0af86c3d7cf71ebba35c49226d98a44dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6752026c6c3b84a162b38c5f5bc5592f263b07d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:41:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b4347639f504b4d95be201e8e37622e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec5634da3294b81a195e3457b5a6e2d : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a93ec6f837b4a5fad0230824731ff65 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/876316a094214456afe06614bc930f7a : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f614cea808a4426a4712b0e12f53282 : POST_FAILURE in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c91ade4f7354e0ebda734d27be4defa : SUCCESS in 6m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/718b0c97cb5b48758d966ced6633d923 : SUCCESS in 7m 06s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2c0e2bb7effe419882665645e99ba9f6 : SUCCESS in 51m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/06f5b39179eb46c1bba9b1903baf2459 : SUCCESS in 4m 37s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/af79b7afcef149598fba05855506d4cc : SUCCESS in 4m 15s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/872267165e7640ceacf21bd4f0375e40 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126e49bbfa3e4f189f3c57835b73831c : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7f660a0f8d48b1bf78261c4c5ec956 : SUCCESS in 3m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ec0c4278467408e9063dcc3eff3fbbc : SUCCESS in 32m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/281424bb87864c9e86a920471c8b23e5 : SUCCESS in 55m 06s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/3050ef354ff54f15a96d781bcce24c98 : FAILURE in 57m 51s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/658451ad56314ca1a1c588edb7a535ff : SUCCESS in 24m 01s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/44733610c5f74eee996f0915583f2721 : SUCCESS in 22m 44s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b4a8d150f1424f59a396e2c9b96a13cb : SUCCESS in 18m 25s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/cb536cfafe5749c4896213373265d9a8 : SUCCESS in 17m 32s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/2bf22b23deef418380909502b855071c : SUCCESS in 17m 04s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5d0fc42528944c778427365243b9a328 : SUCCESS in 41m 28s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/362b29cf367843bdb6259c5bbf744dac : SUCCESS in 1h 00m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c8f8d27792364a62b838a99daf1172db : POST_FAILURE in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ac9909e99c3f2cb3690cea95805c7e804cc8963","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:42:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c48212b8eee91d11d861f574fe36b623ef896a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:32:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf1a13eaf0545f0b52059cc706beb9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844e6d3d23454b8e85b111cf0b77e2ee : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d135a120f954dd290100c959d2124cc : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/542b59c0997241adab4e21845f156072 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc2a998f97794be99486f04ec4ba8ccd : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97e5cd6e034a4df5a51339a13535ca86 : SUCCESS in 7m 16s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/909b9f285dff4289aa881026cb9129f7 : SUCCESS in 8m 16s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/595fac6a2b9c4f019e16d3ae96c6b48f : SUCCESS in 50m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/95a90e9630d647caa20492fe11a40a6d : SUCCESS in 4m 58s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5e9865aa349a4574af77658f00c09242 : SUCCESS in 5m 08s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/87e8ca114e1d49b6b49eb4f772e0960a : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7637c71feadd4e75b91b3c137fbfdff5 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f6e5825fae40e8a4028b47616279ee : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8dce33d4a5e44d8bfe6be1228975208 : SUCCESS in 46m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/68848e371e734f29b57fb74aef0fcd8d : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/54e21514ad2b42ec89a8c67b444b7353 : FAILURE in 1h 38m 31s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/f90ade6ce6714e88889d20a1e017cf92 : SUCCESS in 53m 16s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/6af3b21097954add85fcc4959d01d4fa : SUCCESS in 41m 31s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/c348a0fecd284fa79491b3887a599976 : SUCCESS in 31m 31s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/eda89670dabb4952a8bef482b4d9f545 : SUCCESS in 29m 22s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d22f0562d9ee4648a6f25e952e241331 : SUCCESS in 29m 45s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/297f8e42bd6641deab925e779e99b507 : SUCCESS in 41m 13s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9dce61dc2e494797b3b5dd67904df93c : SUCCESS in 43m 06s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/28e948ef85c34a33aa813c4a82e1639b : SUCCESS in 1h 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"837eb46c0281ac8588963cbe26ab5cff729a81c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:34:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3149cd5a16470e0f8668c246c987628a3c57a6bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab3f5ca0f90741d893634f3acf58e2da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db64a259ba5e4f53976f91423fadc1c3 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a0bc6bcc20b436aa12e60343806bb5f : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11dad06dd9934708ad760e39f123ce05 : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da72c8e74831448fb239e83d2b88c553 : SUCCESS in 7m 07s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a0b870aeb5e54682a19bbfa70a64c3dd : SUCCESS in 50m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c40b927c906d42898d3e376e524fc1b8 : SUCCESS in 5m 47s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5e3906d7b35746b0aa9443e8fbc667fd : SUCCESS in 5m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9e8a60d45476431aa21917e9ed5d8a31 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b383ba852e6e4a628461b498c6a49926 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e8fa978fcb46418b53e090d1d3608a : SUCCESS in 7m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/391f493191fc4bb4bda7d1a0c259f24d : SUCCESS in 1h 01m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/42a3b9fef4764f588917809d16a5f2f4 : SUCCESS in 1h 13m 09s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c3eded01860a4ecda3909aace9ec4f64 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":1},{"id":"df46402c10c8a0e26441591cca5a174ea38aac07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f331761c8ccef1b2dc4b1f62c805ad46a9692905","revisions":{"f331761c8ccef1b2dc4b1f62c805ad46a9692905":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:28:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/1006279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/79/1006279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/1006279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/1006279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/1006279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/79/1006279/1"}}},"commit":{"parents":[{"commit":"eb5a6698f40e699c4b30c22d779d840583bc7751","subject":"zuul: Remove unused jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eb5a6698f40e699c4b30c22d779d840583bc7751"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:23:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:24:20.000000000","tz":60},"subject":"zuul: Remove more dead jobs","message":"zuul: Remove more dead jobs\n\nThese have not been running since they were moved into this file.\nNowadays, most of the jobs they depend on are dead. Remove them.\n\nChange-Id: Iff40617dc4dcf87eb7034b1d47f2c298f31e0e42\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f331761c8ccef1b2dc4b1f62c805ad46a9692905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f331761c8ccef1b2dc4b1f62c805ad46a9692905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
