)]}'
{"id":"openstack%2Fopenstacksdk~1006280","triplet_id":"openstack%2Fopenstacksdk~master~I1e42d7de19ec684a8e7d2669ecfea07db29837b8","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I1e42d7de19ec684a8e7d2669ecfea07db29837b8","subject":"zuul: Use proper boolean values","status":"MERGED","created":"2026-09-18 16:28:43.000000000","updated":"2026-09-21 15:14:35.000000000","submitted":"2026-09-21 15:11:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006280","meta_rev_id":"cf1aa81c1c21886d59b32baaed75bf81167baa76","_number":1006280,"virtual_id_number":1006280,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 11:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 11:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 18:06:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"962360be7b3248a30a6938c2fec38925df096c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b665bfadad6e4c5f367565ce9af67525272bfa58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:06:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3a11e74e6e4e48896d5798bfa9511c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b51991925f24eb8aabffd5db40eb09d : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/755f4f60c55f4e11837ee57981d7e2fa : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21986ef38cbd40569c5f3de4cfda1457 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/341dff1ff25340f4867829fcf15624ab : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e8a2589ffc24ab38b3b5dc306630164 : SUCCESS in 6m 43s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bec5ce7c9d5940d7acb24efa8236d231 : SUCCESS in 7m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9abbc9e8355b4c79993a2afbd4bc4ab3 : SUCCESS in 1h 04m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/79f9b9317c51430c81a85779238eb0c4 : SUCCESS in 4m 17s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/53e1b3ff74ab44e5b9b1f047508ef898 : SUCCESS in 5m 25s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1965e0fd638a4bdb9ab03f8b34920848 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ae1c4513874b85b73291195e3c6876 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/719e4a65d67942eab247e0d8b8453916 : SUCCESS in 3m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/742974eaf8c54184abada3f49d6f9488 : SUCCESS in 1h 10m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23a3fa8aa1f84ecab8a5fe042b84f169 : SUCCESS in 58m 15s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/406cd8d0181a413c89aa978605e8d21a : FAILURE in 1h 25m 10s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/6b982baf49414eab97161d74f74f01ed : SUCCESS in 26m 14s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1b0a62cda67346478293699cac876ac3 : SUCCESS in 34m 44s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/ae5c702861064979bd10eded9765d329 : SUCCESS in 32m 03s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/74267d3aef0c4ad1860ae1cde735cd88 : SUCCESS in 17m 53s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b42c2b63045d4a5998adf6b3f7cc28f9 : SUCCESS in 34m 26s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/4331be95b4d04dd4a753b648d3a7be37 : SUCCESS in 43m 22s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dd44b2baa87947bea0d4411fb0b09d2b : SUCCESS in 48m 38s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/15257857b8ea452b93cc29c9d4a25374 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ceb61631c55ae8794acf945488e1d1e6df1ee03f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:42:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1023d035de96cf169300f81b7ba0827fd6416092","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d76e6db251a397744e7a1a939ccc341fed0b9df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7976db4e29c4e9984cbf03d9c713f70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b777553f62a4ac196dcecb3985d3b84 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc181083017f40cf905e04019fe589c0 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4ab5a45164a4e9fa552fee451d89572 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da0aa5476b7d49159a6cb99fc553aaf1 : SUCCESS in 7m 21s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f980c252bb084e4d9cf8fe071b1eea79 : SUCCESS in 44m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f497d9ebe7f24c15bcc0cc96a34c9b82 : SUCCESS in 4m 34s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4d26bc5bb21e4fd7987ff912691d6b34 : SUCCESS in 4m 07s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d72caef468594cd6acfc2eeabbd156f8 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cdd8bbafc8497283bf60d31499f4b8 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8170f004e5504da0ae11c60d4259f0da : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/585e2de9f8694ae9b5fed59324e9f835 : SUCCESS in 44m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/33baec4406cc438f8fd735014dfa451c : SUCCESS in 59m 29s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/51c437abc3f743b4bc8c36f8cfefddf2 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":1},{"id":"afd5ce91630cc171da2f72191cda8d6652ad8858","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:11:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf1aa81c1c21886d59b32baaed75bf81167baa76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:14:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74527663cf0d4292b294a13bb33afbdd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5735820bb67a448b9d38beade8d80abf : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca726d28ef8df93a87aa1bad62561b97358205e5","revisions":{"ca726d28ef8df93a87aa1bad62561b97358205e5":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:28:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/1006280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/80/1006280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1006280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1006280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1006280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/80/1006280/1"}}},"commit":{"parents":[{"commit":"f331761c8ccef1b2dc4b1f62c805ad46a9692905","subject":"zuul: Remove more dead jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f331761c8ccef1b2dc4b1f62c805ad46a9692905"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:16:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:27:13.000000000","tz":60},"subject":"zuul: Use proper boolean values","message":"zuul: Use proper boolean values\n\nWe also remove a dead project from the required-projects list and drop\nsome unnecessary newlines.\n\nChange-Id: I1e42d7de19ec684a8e7d2669ecfea07db29837b8\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ca726d28ef8df93a87aa1bad62561b97358205e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ca726d28ef8df93a87aa1bad62561b97358205e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
