)]}'
{"id":"openstack%2Fopenstacksdk~1006281","triplet_id":"openstack%2Fopenstacksdk~master~I5eef8259c7d4fa568188b2f75ce7100455df9a00","project":"openstack/openstacksdk","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-18 18:14:32.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5eef8259c7d4fa568188b2f75ce7100455df9a00","subject":"zuul: Re-enable post-review queue","status":"NEW","created":"2026-09-18 16:28:43.000000000","updated":"2026-09-18 18:14:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3f776d664a7560099d41a538fbc2868cf4261597","_number":1006281,"virtual_id_number":1006281,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 18:14:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 18:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fbaebf1583177109376d99586e590c3c2ed3d958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 16:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f776d664a7560099d41a538fbc2868cf4261597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:14:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07877978ca3c468e8e1116a3cd7979f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8052d88666be4e3f86cd15cdfca5c802 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bacb40973272423fa9a7600f2dd85884 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a46a0a09be7a4abdbe2d581bd413c4a9 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea7b1138863848b2872a8deb81ede043 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e3a262fa23a4a619f92407afcfeb3e9 : SUCCESS in 7m 20s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c11e8557754b407299e9ef270f81375e : SUCCESS in 7m 47s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6b849537855e4b18869379ef95400325 : SUCCESS in 50m 22s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4282fe0a35a546419803903f24ae584e : SUCCESS in 3m 54s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bafbfbc5740f4b0ebef54a6ecd1acdae : SUCCESS in 5m 11s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ea990b36abc8461ab292640ae4f76342 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdda04fcc06c4c6887831907593ac34f : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c3df081acc34c32a0c06db349e78412 : SUCCESS in 6m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05dfaf25300f44b49fb3516ae953da41 : SUCCESS in 1h 05m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5bcc4b209b3b49de9ff6876c115e99ec : SUCCESS in 1h 27m 41s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9333e4a34af047558bf9b9f58831c984 : FAILURE in 1h 33m 44s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e6e00e9230934ed0b11e50f9295f4162 : SUCCESS in 42m 19s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e4d48643eb114451ae916946287e8171 : SUCCESS in 29m 10s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/0d4d6a09f6eb4243bb12a8094c597444 : SUCCESS in 28m 22s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/b03bfdf196d04d0fa9c11d79b253e2c8 : SUCCESS in 32m 55s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/90d65c2d110a4d43b416e0c16f6c2d22 : SUCCESS in 33m 18s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3ab1cb9656f24f8a88563d8d85e57655 : SUCCESS in 44m 41s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f756ad14f1a6465fa8d5e2000a491b5b : SUCCESS in 1h 18m 57s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/46128d39423d4e27a447a1cb5cb4e7f8 : SUCCESS in 1h 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5973759b4ba6550083ca5028d26aa6fc1f087318","revisions":{"5973759b4ba6550083ca5028d26aa6fc1f087318":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:28:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/1006281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/81/1006281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/81/1006281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/81/1006281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/81/1006281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/81/1006281/1"}}},"commit":{"parents":[{"commit":"ade81e35883b716134608e4260f29e314f468a21","subject":"zuul: Make openstacksdk-functional-devstack-networking-ext voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ade81e35883b716134608e4260f29e314f468a21"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:27:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 16:27:13.000000000","tz":60},"subject":"zuul: Re-enable post-review queue","message":"zuul: Re-enable post-review queue\n\nChange-Id: I5eef8259c7d4fa568188b2f75ce7100455df9a00\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5973759b4ba6550083ca5028d26aa6fc1f087318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5973759b4ba6550083ca5028d26aa6fc1f087318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
