)]}'
{"id":"openstack%2Fopenstacksdk~1006291","triplet_id":"openstack%2Fopenstacksdk~master~Iea2e48465f232710742885ccf6910b43c9c90790","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-21 16:13:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-21 19:13:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea2e48465f232710742885ccf6910b43c9c90790","subject":"Fix typo","status":"MERGED","created":"2026-09-18 16:57:24.000000000","updated":"2026-09-21 19:17:28.000000000","submitted":"2026-09-21 19:13:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006291","meta_rev_id":"95706622d76ee56ecd0582db407fcfd58ce3d7d9","_number":1006291,"virtual_id_number":1006291,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 19:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 14:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-21 14:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 16:58:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-18 18:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:24:09.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"0bb2a4c5d82ab0dd0cde2d5985c6ae2631f3a448","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 16:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fc75050b853d98752fb7f65898e60ce7e404e52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 16:58:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c1c035215bea66acec7c8966d3fd88d78ee3fefa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 16:58:22.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"80103cf6e084d55ee63f37102c4cea20d53e86ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:31:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd44c9d034d4b348e9aa653cea4b436\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79fb3fa4511c42d4ab9e9523e0c59136 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5405f96e13cd4b078f74cbb574bdf9d5 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d19836f867724d57b1f2f5a19e5890ef : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/610acda38bcd4a2aa3fe843a0afd89c9 : SUCCESS in 5m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03dd72945ff24662842d60cb12ef73c8 : SUCCESS in 8m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/cf456fbb941b471ba6f1299ebfc640be : SUCCESS in 7m 01s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/86ef4a79d09f492a9add231d02b2ec46 : SUCCESS in 57m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/15a1d08823a24056bedc3a58c321f6f8 : SUCCESS in 3m 52s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d143c1c70062410f81b7e40561be7eae : SUCCESS in 4m 43s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f79ed75c030f408daef449a09f2b27e7 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdc5d76a8454c3aaf76110d2a3ffb4e : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c691ef5ce254e8e837453a9f6abda2e : SUCCESS in 3m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95751353a4ab424fa9afb3014359388b : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/751d016363124cceaa9babf7acb778b1 : SUCCESS in 1h 25m 43s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8da9de7052904633891c22db973218d9 : FAILURE in 1h 20m 24s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5444366a7c414a81bcb791086157ced7 : SUCCESS in 37m 37s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b11c5be583844a85943455595df6f253 : SUCCESS in 36m 49s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/2ac8ee66a8bd4eb08c7203d120600bb1 : SUCCESS in 29m 59s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/89c5e6f3c6d84e8fb18967f6f2bbbf32 : SUCCESS in 31m 47s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/49a9b52875de4f1fb653825eef9c9786 : SUCCESS in 25m 57s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0271976e8cf444879d5887f83a1c19ba : SUCCESS in 41m 43s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/96a07f764d6241558354c2c59e05296b : SUCCESS in 1h 17m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/add771ea50f34d538e3092dee8e3b1bf : SUCCESS in 1h 09m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa0a3835ca48671f13a76c4c9251d413e51a0d86","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:14:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbf82f9a0f7065ad921611611cff4f217f81c810","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:22:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dabc2d84df1cd21e839db51691cdef546d179f51","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:24:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6dc89383cfd900eacf105d0c02c11a0ceec7f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:26:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10ab96d1133a31afc72e6060b4e05c989cd995d5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-21 15:02:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f3daa7c356872b3096d76e4ce7113afe9ab062c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:55:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aebf0778ea9143ffb4fe8b1184ef39a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a2e7d0b28b4930af38ab0d4a08162a : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c76b0eacee446cbb8491fee24b7853b : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/417073ffe3ef4db8ab9439ee20a72594 : FAILURE in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/103ea5382fff495bb8f7b41416f1385d : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/546d82c0e9534da58e717abfec4525ba : SUCCESS in 1h 09m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e43ca8293c5f4ec09dafbf5664c80532 : SUCCESS in 4m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8659db9ea24749f5b95429a0088dcdc2 : SUCCESS in 5m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/15768828c50643dc94d9dc0df03fec73 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14812b8756a143ad8383227006b404c3 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5ed4aeaf9a471589175399c9b23523 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb89647c986d4040be7a0a3cd420ad75 : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5ed711d3c834bec81a8dd8fe6da15af : SUCCESS in 1h 27m 31s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e3cf8a1b5ccc41b885b4b436cd590771 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a6c6724cb63f3ae1dcf5f8afcef1a73d0fdf5bb4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 16:13:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1dfd48316dc2887ff1fa1412ca871a1eb0e1b11e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97cb26edda2c40cca49fc386f087ac19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815fcac198cf485eb17faa9b7b4de217 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55136ffb34564fa7b1bd627fe3bb599d : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1a5b04d51fe436bbda6c08a4510d65a : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef29f4a48f944a14bdb230b42e1fd226 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/253ebe1cc75247a5a9df54ace7087037 : SUCCESS in 6m 34s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bb7d585d101c4ea68f46dac23016c61f : SUCCESS in 7m 40s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8526264fd73c49a588302bb9fb12f5e3 : SUCCESS in 59m 01s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cf13b534df6d4f969f0aa593e9c37150 : SUCCESS in 4m 08s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2134c43262954bc9aa870c3defb9da55 : SUCCESS in 2m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/94d40ad5ea4e411d85733f06d0a8d853 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5a2d0461a744a39f6002f6b8d938ed : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1601d9a8fe74d92bbd0f9a7639ddcbf : SUCCESS in 5m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfa7581a5f1f44e3aa3708f7d8ba0d18 : SUCCESS in 36m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d7608f3ae33e4c17afc81369e4b99b51 : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/b0d678d986b348aab1e7323617ae956c : FAILURE in 58m 05s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7641273fd70346ffa0d950f36810aba4 : SUCCESS in 24m 34s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/594d8fda3d39403c8308f8d8451c843a : SUCCESS in 34m 04s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/dc02e1c71d03430d8aff2892edf5cec0 : SUCCESS in 21m 46s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/df068aafaeda49fd943f462a1e019e77 : SUCCESS in 17m 38s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/4799b458e4ac42d19ae018bae11f9411 : SUCCESS in 19m 49s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f5c50fbaec194611bc41080ea6c4ea0f : SUCCESS in 42m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/aa506c437e5c4fda988eb6f1406a9ee2 : SUCCESS in 1h 17m 42s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/289278455ee249e693a3a78b10c24578 : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf2b474d93ad352d75320b91a3db90472f395119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb2a6113a259da55b041ed799c4b98c8df5dc0aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:13:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb641bbab5a4ec3b5fff792beeb9e6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61de8d0abc574afd94cca898758f0cbd : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/594ae769f3504b8a91ca49c711766e98 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc22aa269e6141ad9a6b38369a0fed5a : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ae6183696f84fbd956786ae41af8c71 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0259b7926645481b83ce058ea8b36947 : SUCCESS in 35m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a5f631be2e474b26a7ccd18fb77347cf : SUCCESS in 5m 55s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a141706ef2c94459a891ecc1d794533f : SUCCESS in 5m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9c9c2c00b6644a38b81909e22201bf53 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a322bbdbce941bda66ee35e8c0ac029 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf6128fcaa2444da5baab3821ce93c4 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa8b2e9b5c6747b599a4ff93d0a932da : SUCCESS in 35m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d9538f36808d433b8746290cb36d692d : SUCCESS in 1h 20m 03s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c43f42848dd043e4b6318dd2ae60bb6d : SUCCESS in 23m 58s","accounts_in_message":[],"_revision_number":2},{"id":"8224403ae2cbc21a9a7a636045a3a9d9c27ea78b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:13:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"95706622d76ee56ecd0582db407fcfd58ce3d7d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:17:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/329c337520a647ba977081e7c187cd4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9e4630b012455fac87878d17b8e974 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5d03ef73c414e74b895583b688f55a818ac62ae","revisions":{"8c7ced609b629ebdc314a0db9c33cb2cf0625ca3":{"kind":"REWORK","_number":1,"created":"2026-09-18 16:57:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/1006291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/91/1006291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 16:57:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 16:57:24.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Iea2e48465f232710742885ccf6910b43c9c90790\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8c7ced609b629ebdc314a0db9c33cb2cf0625ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8c7ced609b629ebdc314a0db9c33cb2cf0625ca3"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"a5d03ef73c414e74b895583b688f55a818ac62ae":{"kind":"REWORK","_number":2,"created":"2026-09-18 16:58:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/1006291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/91/1006291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/91/1006291/2"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 16:57:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 16:58:15.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Iea2e48465f232710742885ccf6910b43c9c90790\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5d03ef73c414e74b895583b688f55a818ac62ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5d03ef73c414e74b895583b688f55a818ac62ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
