)]}'
{"id":"openstack%2Fopenstacksdk~1006703","triplet_id":"openstack%2Fopenstacksdk~master~Icd93debabf56de1414b99c60d11345cbdb54e81a","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 18:11:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd93debabf56de1414b99c60d11345cbdb54e81a","subject":"tox: Run mypy over package","status":"MERGED","created":"2026-09-22 11:15:35.000000000","updated":"2026-09-22 18:13:25.000000000","submitted":"2026-09-22 18:11:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006703","meta_rev_id":"e22bd1633ba525bfc54982c37cc1466ba164c1d4","_number":1006703,"virtual_id_number":1006703,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 18:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-22 17:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-22 17:04:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 12:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 17:04:24.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"6b95fc08395f5d6b768293d7e09bc2a15d69fff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-22 11:15:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19665447fff8465bc40fe68d8694b15fb78dc0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 12:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35e9dff77ca456b84d2cf76a103a2d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f805dd4d2e364a6eaf33e8e6611c2e61 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e342f28201e34ae79db2a91e0adac5bb : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e60db5cb3fe45ed89421c3018e80883 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a8aa5f92c914b35855a8a850c6ad740 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a4e113403b1466fa3c2b017c5ba7d5d : SUCCESS in 6m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/98b24d0e68464e6e80711629ce2945f8 : SUCCESS in 6m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/939bfb5fa9664482b446ca7974e351e0 : SUCCESS in 33m 03s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0f38846f86f94d77b8db0c039c21c759 : SUCCESS in 3m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f34a7aaa6ffe4f0797c22fe3934793df : SUCCESS in 4m 32s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/33742466ef6244aa9847627d8cf9cbf0 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d671bb2eb840f9a24dc1b350595a3c : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1622ff9f512c49ac977e4667f2e1d08f : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b75bcd04c6404cd3971bfc5715801421 : SUCCESS in 57m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c3002daba1b14d1093f85da90fb8f3e7 : SUCCESS in 1h 27m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/da81a1e304864a25921f523ffed92377 : SUCCESS in 1h 28m 50s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ee67111f59cb499580d01db4ff3acf27 : SUCCESS in 42m 13s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e672c7ebeb5e4d53bbdca21a02aab1de : SUCCESS in 34m 44s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/e8354f5ce86744dc85d939b04480bc36 : SUCCESS in 27m 32s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/1f196607463546bf98a2397782ee9cf0 : SUCCESS in 29m 00s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/72c32769b95145b6bf83702aefd630b4 : SUCCESS in 29m 53s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b21864636e754d2ba70b5af24764b329 : SUCCESS in 42m 27s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/09f7bf7e45864bd18e529234a993bd92 : SUCCESS in 1h 06m 23s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/780bb97a37ad41129155a812325b926b : FAILURE in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12b2cd2660979902968f95dc832e83da85bcac40","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-22 17:04:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b80eb32a7a8b68c3bef426335d4142581f19ba40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:05:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37d09260421e7a134903c251fc8659794f1eb90d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:11:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b47f67e8b9d4782a08063993fc932ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c1a296fa294df38ed33263e810681f : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c7f0e5e839c4429a58c56ad211edb68 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d134a8cbb6e4773bc0ead9630c6432f : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bddd104685446ceb747d4d27534da0e : SUCCESS in 4m 58s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e40448fafe4c41c2b21130fb4beee925 : SUCCESS in 40m 08s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/94f0952843d74ce094b7630428dbe5dc : SUCCESS in 4m 07s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/be0a97762cf74ff190de36bd9c41cbec : SUCCESS in 3m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0eac447cf7c443ddb034b38ef30a2365 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d30b94e19849e99086907a78fb096a : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ed3c0688da45d595a87ca98bda6d67 : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22b8b82d4411471e9eba95ece605b4e6 : SUCCESS in 42m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8b6b2eb28216426b86d8920b3fa83620 : SUCCESS in 1h 05m 22s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/305bbf5da261466180a30b1e1490d082 : SUCCESS in 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"778d9fd3c6b1f3ca1a62f0ed8d239f85338c9bfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:11:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e22bd1633ba525bfc54982c37cc1466ba164c1d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:13:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd5087e3efb45bb84b524dcd3e03d5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4cd709240494cd188b536c7dccd12e3 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/753a48c7a50e431797015b0c3555d118 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14b4923b168c627d5706186a3aa10aaf37d45f4c","revisions":{"14b4923b168c627d5706186a3aa10aaf37d45f4c":{"kind":"REWORK","_number":1,"created":"2026-09-22 11:15:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/1006703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/1006703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/1006703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/1006703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/1006703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/1006703/1"}}},"commit":{"parents":[{"commit":"0382628a572ff87f83cbf8fbfcf4b8ef5d5a0579","subject":"Merge \"zuul: Use proper boolean values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0382628a572ff87f83cbf8fbfcf4b8ef5d5a0579"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 22:53:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-22 11:15:23.000000000","tz":60},"subject":"tox: Run mypy over package","message":"tox: Run mypy over package\n\nWe never want to run mypy with this configuration over a different\npackage.\n\nChange-Id: Icd93debabf56de1414b99c60d11345cbdb54e81a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/14b4923b168c627d5706186a3aa10aaf37d45f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/14b4923b168c627d5706186a3aa10aaf37d45f4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
