)]}'
{"id":"openstack%2Fopenstacksdk~663429","triplet_id":"openstack%2Fopenstacksdk~master~I7622af9213df2e5c954157f28dc6ec21b188f0fb","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I7622af9213df2e5c954157f28dc6ec21b188f0fb","subject":"Swap default of use_direct_get","status":"ABANDONED","created":"2019-06-05 21:06:01.000000000","updated":"2023-03-24 10:06:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d313578724df3ed11df88a95632b0816c0051de2","_number":663429,"virtual_id_number":663429,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-05 22:58:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 22:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6385b0573d023c54fcc26c49bb5ccced36d2a4c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-06-05 21:06:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0f9451175306f148cbc4974fc483f9b3dfb029b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 22:58:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/29/663429/1/check/openstack-tox-lower-constraints/be7b97d/ : FAILURE in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/29/663429/1/check/openstack-tox-pep8/91aaa2d/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/29/663429/1/check/openstack-tox-py27/f988bc0/ : FAILURE in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/29/663429/1/check/openstack-tox-py36/ce34eff/ : FAILURE in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/29/663429/1/check/openstack-tox-py37/747c319/ : FAILURE in 3m 08s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-tips/f807c1e/ : FAILURE in 51m 39s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-tips-python2/1b3a7cf/ : FAILURE in 1h 04m 11s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/29/663429/1/check/openstacksdk-tox-py36-tips/75caa83/ : FAILURE in 3m 32s\n- os-client-config-tox-py27-tips http://logs.openstack.org/29/663429/1/check/os-client-config-tox-py27-tips/73b68bb/ : SUCCESS in 4m 54s\n- os-client-config-tox-py37-tips http://logs.openstack.org/29/663429/1/check/os-client-config-tox-py37-tips/bbb1e3f/ : SUCCESS in 5m 53s\n- osc-tox-py27-tips http://logs.openstack.org/29/663429/1/check/osc-tox-py27-tips/05810de/ : SUCCESS in 6m 18s\n- osc-tox-py36-tips http://logs.openstack.org/29/663429/1/check/osc-tox-py36-tips/5b5d182/ : SUCCESS in 7m 35s\n- openstack-tox-docs http://logs.openstack.org/29/663429/1/check/openstack-tox-docs/94317d4/html/ : SUCCESS in 3m 53s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/29/663429/1/check/openstacksdk-ansible-devel-functional-devstack/be47c4e/ : FAILURE in 42m 41s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/29/663429/1/check/openstacksdk-ansible-stable-2.6-functional-devstack/a5657fc/ : FAILURE in 43m 17s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack/3c9ec3b/ : FAILURE in 1h 07m 40s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-networking/3b8e571/ : FAILURE in 1h 36m 11s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-senlin/593702e/ : FAILURE in 52m 52s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-magnum/ac0f928/ : FAILURE in 57m 51s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-masakari/7ed29cb/ : SUCCESS in 32m 24s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-ironic/946bd52/ : SUCCESS in 34m 25s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/29/663429/1/check/openstacksdk-functional-devstack-python2/d610559/ : FAILURE in 58m 14s\n- osc-functional-devstack-tips http://logs.openstack.org/29/663429/1/check/osc-functional-devstack-tips/86bca0e/ : SUCCESS in 48m 29s (non-voting)\n- nodepool-functional-py35-src http://logs.openstack.org/29/663429/1/check/nodepool-functional-py35-src/d26461a/ : SUCCESS in 42m 32s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/29/663429/1/check/bifrost-integration-tinyipa-ubuntu-xenial/441a3b5/ : SUCCESS in 21m 20s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/29/663429/1/check/metalsmith-integration-openstacksdk-src/5bf18d6/ : SUCCESS in 52m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d313578724df3ed11df88a95632b0816c0051de2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-24 10:06:13.000000000","message":"Abandoned\n\nToo old change.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4267d5aafdca9a6c3da0e6416c571aa6bb17672f","revisions":{"4267d5aafdca9a6c3da0e6416c571aa6bb17672f":{"kind":"REWORK","_number":1,"created":"2019-06-05 21:06:01.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/29/663429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/29/663429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/663429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/663429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/663429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/29/663429/1"}}},"commit":{"parents":[{"commit":"8780232b7a56e560c2671ce037a8cbd74f9f6704","subject":"Push use_direct_get support into resource layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8780232b7a56e560c2671ce037a8cbd74f9f6704"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-05 15:58:28.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-05 21:05:56.000000000","tz":-300},"subject":"Swap default of use_direct_get","message":"Swap default of use_direct_get\n\nSince we support this all the way down now and nodepool has\nupdated its usage, swap the default value.\n\nDepends-On: https://review.opendev.org/663368\nChange-Id: I7622af9213df2e5c954157f28dc6ec21b188f0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4267d5aafdca9a6c3da0e6416c571aa6bb17672f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4267d5aafdca9a6c3da0e6416c571aa6bb17672f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
