)]}'
{"id":"openstack%2Fopenstacksdk~663435","triplet_id":"openstack%2Fopenstacksdk~master~I73a4056142d7aa396f6929ddfac2edfe5e06fffc","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I73a4056142d7aa396f6929ddfac2edfe5e06fffc","subject":"WIP Plumb service disabling into Connection adapters","status":"ABANDONED","created":"2019-06-05 21:23:04.000000000","updated":"2019-06-06 14:35:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4f3e1a8f8df489a885dfaf16e11588500555be41","_number":663435,"virtual_id_number":663435,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-06-05 23:44:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":-1,"date":"2019-06-05 22:03:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 21:46:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-05 23:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a41c04d8340b47bb25f9b634336234b54b92872c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-06-05 21:23:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"facd1f00712d7bac80a2234a523eff880db9aed5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 21:46:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2c58b8675ca206139cd10d34aabdf8ed0ea97a5e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 22:03:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"dddb27d14b5b66fb2b29fa9bf202d74f7bd90cbe","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 22:32:43.000000000","message":"Patch Set 2:\n\nI resolved my issues in the next patch fwiw","accounts_in_message":[],"_revision_number":2},{"id":"5305a2cf633cf9dc78c63e448ab7249b62cd8744","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 22:47:26.000000000","message":"Patch Set 2:\n\nSquashed into https://review.opendev.org/#/c/663439/","accounts_in_message":[],"_revision_number":2},{"id":"7eed4777e4327eb1c67f23e16f611e4df176dd0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:44:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/663435/2/check/openstack-tox-lower-constraints/197380b/ : FAILURE in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/35/663435/2/check/openstack-tox-pep8/c737fe8/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/35/663435/2/check/openstack-tox-py27/ff46908/ : FAILURE in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/35/663435/2/check/openstack-tox-py36/9b59190/ : FAILURE in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/35/663435/2/check/openstack-tox-py37/0977bb4/ : FAILURE in 3m 50s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-tips/3e5c7c0/ : SUCCESS in 1h 01m 22s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-tips-python2/2e2a696/ : SUCCESS in 1h 02m 41s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/35/663435/2/check/openstacksdk-tox-py36-tips/3940f79/ : FAILURE in 3m 38s\n- os-client-config-tox-py27-tips http://logs.openstack.org/35/663435/2/check/os-client-config-tox-py27-tips/29ccf87/ : SUCCESS in 5m 12s\n- os-client-config-tox-py37-tips http://logs.openstack.org/35/663435/2/check/os-client-config-tox-py37-tips/f13f93d/ : SUCCESS in 5m 04s\n- osc-tox-py27-tips http://logs.openstack.org/35/663435/2/check/osc-tox-py27-tips/c00c329/ : SUCCESS in 6m 25s\n- osc-tox-py36-tips http://logs.openstack.org/35/663435/2/check/osc-tox-py36-tips/971f35b/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/35/663435/2/check/openstack-tox-docs/b86c11e/html/ : SUCCESS in 3m 38s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/35/663435/2/check/openstacksdk-ansible-devel-functional-devstack/53e179f/ : SUCCESS in 39m 55s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/35/663435/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/17d2ccc/ : SUCCESS in 34m 55s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack/a2d88ef/ : SUCCESS in 1h 00m 45s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-networking/35df523/ : SUCCESS in 1h 24m 12s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-senlin/c9da5ef/ : SUCCESS in 59m 55s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-magnum/322b8fc/ : FAILURE in 1h 00m 52s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-masakari/0e5f015/ : SUCCESS in 37m 09s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-ironic/e5d04ca/ : SUCCESS in 33m 36s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/35/663435/2/check/openstacksdk-functional-devstack-python2/8dd90d0/ : SUCCESS in 1h 07m 34s\n- osc-functional-devstack-tips http://logs.openstack.org/35/663435/2/check/osc-functional-devstack-tips/bb852b1/ : SUCCESS in 57m 58s (non-voting)\n- nodepool-functional-py35-src http://logs.openstack.org/35/663435/2/check/nodepool-functional-py35-src/2225807/ : SUCCESS in 40m 12s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/35/663435/2/check/bifrost-integration-tinyipa-ubuntu-xenial/29dc68e/ : SUCCESS in 17m 07s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/35/663435/2/check/metalsmith-integration-openstacksdk-src/62cfc39/ : SUCCESS in 51m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f3e1a8f8df489a885dfaf16e11588500555be41","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-06-06 14:35:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2797727ee87d2195f360caaa4bce67f6cf9fd72","revisions":{"5f1b1965ecbb50ad7b812756821bd44f64ebfc71":{"kind":"REWORK","_number":1,"created":"2019-06-05 21:23:04.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/35/663435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/35/663435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/1"}}},"commit":{"parents":[{"commit":"a5dfa85d4e7ff5bd5a40d37519670392f05ab6fe","subject":"Get rid of unused _OpenStackCloudMixin.get_region","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a5dfa85d4e7ff5bd5a40d37519670392f05ab6fe"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-05 21:22:01.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-05 21:22:01.000000000","tz":-300},"subject":"WIP Plumb service disabling into Connection adapters","message":"WIP Plumb service disabling into Connection adapters\n\nIf someone sets has_{service_type} to false in their config, we\nshould remove the adapter for that service and replace it with\nsomething that throws errors when people try to use it.\n\nChange-Id: I73a4056142d7aa396f6929ddfac2edfe5e06fffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5f1b1965ecbb50ad7b812756821bd44f64ebfc71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5f1b1965ecbb50ad7b812756821bd44f64ebfc71"}]},"branch":"refs/heads/master"},"a2797727ee87d2195f360caaa4bce67f6cf9fd72":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-05 22:03:44.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/35/663435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/35/663435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/35/663435/2"}}},"commit":{"parents":[{"commit":"7338aae7586a38b8ace81896ebed478d9f583862","subject":"(Broken) from_conf test paths for oslo.config exceptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7338aae7586a38b8ace81896ebed478d9f583862"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-05 21:22:01.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-05 21:33:41.000000000","tz":-300},"subject":"WIP Plumb service disabling into Connection adapters","message":"WIP Plumb service disabling into Connection adapters\n\nIf someone sets has_{service_type} to false in their config, we\nshould remove the adapter for that service and replace it with\nsomething that throws errors when people try to use it.\n\nChange-Id: I73a4056142d7aa396f6929ddfac2edfe5e06fffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a2797727ee87d2195f360caaa4bce67f6cf9fd72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a2797727ee87d2195f360caaa4bce67f6cf9fd72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
