)]}'
{"id":"openstack%2Fopenstacksdk~672190","triplet_id":"openstack%2Fopenstacksdk~master~Ia936cd41b87dd9422b00e96ee541f2c3842dc636","project":"openstack/openstacksdk","branch":"master","topic":"add-set-boot-device","hashtags":[],"change_id":"Ia936cd41b87dd9422b00e96ee541f2c3842dc636","subject":"Add set-boot-device to baremetal","status":"MERGED","created":"2019-07-23 00:03:32.000000000","updated":"2019-07-25 02:20:17.000000000","submitted":"2019-07-25 02:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672190-1564021217814-b36f6c23","meta_rev_id":"0b67fa587190e8c5237e142c0599c6839661efe0","_number":672190,"virtual_id_number":672190,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-07-24 23:40:35.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-07-25 02:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-24 11:38:11.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 12:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-24 11:38:11.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 12:44:27.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-07-24 11:38:11.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-25 02:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b67b9c777588d7ae68163c3b8973b0ef0fe0967","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-23 00:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f98a2ccc97777e5a99e7431a328bb2cac7cc49dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 02:09:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/1/check/openstack-tox-lower-constraints/fd2107b/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/1/check/openstack-tox-pep8/9aa8ce3/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/1/check/openstack-tox-py27/30d5857/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/1/check/openstack-tox-py36/e6faabf/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/1/check/openstack-tox-py37/41c4045/ : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-tips/94365c0/ : SUCCESS in 1h 00m 30s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-tips-python2/83a8d9c/ : SUCCESS in 1h 01m 59s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/1/check/openstacksdk-tox-py36-tips/4d1dcd5/ : SUCCESS in 4m 17s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/1/check/os-client-config-tox-py27-tips/8a8871b/ : SUCCESS in 3m 15s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/1/check/os-client-config-tox-py37-tips/daed60b/ : SUCCESS in 4m 19s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/1/check/osc-tox-py27-tips/b24851c/ : SUCCESS in 4m 32s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/1/check/osc-tox-py36-tips/c273163/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/90/672190/1/check/openstack-tox-docs/38bdeb1/html/ : SUCCESS in 3m 26s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/1/check/openstacksdk-ansible-devel-functional-devstack/1db1bbb/ : SUCCESS in 38m 30s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/1/check/openstacksdk-ansible-stable-2.6-functional-devstack/5486a63/ : SUCCESS in 41m 16s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack/a2d1526/ : SUCCESS in 53m 44s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-networking/9e4c34d/ : SUCCESS in 1h 18m 53s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-senlin/18a25b7/ : SUCCESS in 1h 06m 08s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-magnum/5a2a8d2/ : FAILURE in 50m 38s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-masakari/dec5981/ : SUCCESS in 32m 20s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-ironic/1a7310a/ : SUCCESS in 46m 17s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/1/check/openstacksdk-functional-devstack-python2/29db085/ : SUCCESS in 54m 06s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/1/check/osc-functional-devstack-tips/03bc6d9/ : FAILURE in 35m 06s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/1/check/nodepool-functional-openstack-src/0b01735/ : SUCCESS in 33m 24s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/1/check/bifrost-integration-tinyipa-ubuntu-xenial/6a63ef5/ : SUCCESS in 19m 33s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/1/check/metalsmith-integration-openstacksdk-src/a6054cb/ : SUCCESS in 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a819fed7e5f768354e56d5e1b11fcb2fb72b7cc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-23 09:27:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67167317d891f8f7ea2f55a33a6b337d059e5322","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-23 09:43:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba0b34b28ca9e94033b88b6caa5ccd64ee0ec9b1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-23 09:44:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47713934b28e8b91179689a39c4b3fd2fcbccee7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:10:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/check/openstack-tox-lower-constraints/57a5601/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/check/openstack-tox-pep8/5eb8873/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/check/openstack-tox-py27/9af28bd/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/check/openstack-tox-py36/387286c/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/check/openstack-tox-py37/0bb57ee/ : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips/b9d82d1/ : FAILURE in 35m 43s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips-python2/95fae10/ : SUCCESS in 1h 05m 04s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-tox-py36-tips/30b99df/ : SUCCESS in 4m 16s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py27-tips/77b1a7c/ : SUCCESS in 4m 56s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py37-tips/adc351c/ : SUCCESS in 4m 32s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py27-tips/5a031e3/ : SUCCESS in 4m 13s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py36-tips/6f9039e/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/check/openstack-tox-docs/ae0e8ce/html/ : SUCCESS in 4m 45s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-devel-functional-devstack/d5e6def/ : SUCCESS in 39m 25s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/c1540b8/ : SUCCESS in 43m 06s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack/1861f11/ : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-networking/ac8345f/ : SUCCESS in 1h 20m 30s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-senlin/f8432d7/ : SUCCESS in 54m 57s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-magnum/fcdd6a5/ : FAILURE in 58m 03s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-masakari/d95dcc4/ : SUCCESS in 40m 31s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-ironic/3ff9ad8/ : SUCCESS in 38m 23s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-python2/5172f5a/ : SUCCESS in 1h 05m 28s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/osc-functional-devstack-tips/77281e6/ : SUCCESS in 1h 00m 45s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/check/nodepool-functional-openstack-src/9d974f9/ : SUCCESS in 32m 32s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/check/bifrost-integration-tinyipa-ubuntu-xenial/c40b6e7/ : SUCCESS in 25m 03s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/2/check/metalsmith-integration-openstacksdk-src/182003d/ : SUCCESS in 52m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86045df48fa19232bfd93d5707f00e404eb03b71","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-23 11:16:01.000000000","message":"Patch Set 2:\n\nrecheck connection aborted","accounts_in_message":[],"_revision_number":2},{"id":"cd76fa99ffb3b128da55ab607e69acb2283d19a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:39:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/check/openstack-tox-lower-constraints/25ab27d/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/check/openstack-tox-pep8/b3fb99c/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/check/openstack-tox-py27/08a7520/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/check/openstack-tox-py36/c5433fb/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/check/openstack-tox-py37/d24e4fb/ : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips/a561f9f/ : SUCCESS in 57m 24s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips-python2/6298228/ : SUCCESS in 1h 02m 00s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-tox-py36-tips/cdd48aa/ : FAILURE in 5m 56s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py27-tips/7801337/ : SUCCESS in 5m 30s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py37-tips/3bcf2e5/ : SUCCESS in 5m 43s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py27-tips/255f843/ : SUCCESS in 6m 48s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py36-tips/8ecdf5a/ : SUCCESS in 7m 08s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/check/openstack-tox-docs/b31520b/html/ : SUCCESS in 4m 40s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-devel-functional-devstack/578b801/ : SUCCESS in 43m 51s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/20514ca/ : FAILURE in 40m 56s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack/f63b25d/ : FAILURE in 36m 54s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-networking/4677980/ : SUCCESS in 1h 19m 42s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-senlin/41c26e3/ : SUCCESS in 57m 50s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-magnum/26d1644/ : FAILURE in 1h 06m 01s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-masakari/308c4f5/ : SUCCESS in 34m 49s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-ironic/c841005/ : SUCCESS in 37m 27s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-python2/42b6159/ : FAILURE in 39m 44s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/osc-functional-devstack-tips/d082f04/ : SUCCESS in 1h 02m 32s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/check/nodepool-functional-openstack-src/2f1a7bb/ : SUCCESS in 34m 16s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/check/bifrost-integration-tinyipa-ubuntu-xenial/e1a7c52/ : SUCCESS in 18m 30s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/2/check/metalsmith-integration-openstacksdk-src/d7e5fb8/ : SUCCESS in 50m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6210d3424fea56fe2fc551808ba172c16b8de26c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-23 12:44:27.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad22fef8c12720e3df24541c407b61bd1b51d610","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:14:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/check/openstack-tox-lower-constraints/e41560c/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/check/openstack-tox-pep8/dccbf94/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/check/openstack-tox-py27/19d3019/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/check/openstack-tox-py36/0bc7c5d/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/check/openstack-tox-py37/14c9a49/ : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips/17ce1ea/ : SUCCESS in 57m 57s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips-python2/b9b424a/ : SUCCESS in 1h 01m 26s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-tox-py36-tips/f54573e/ : SUCCESS in 3m 52s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py27-tips/08c47bb/ : SUCCESS in 3m 49s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py37-tips/aed5e8e/ : SUCCESS in 3m 45s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py27-tips/4281d8e/ : SUCCESS in 5m 00s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py36-tips/6fdf40c/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/check/openstack-tox-docs/d79a426/html/ : SUCCESS in 4m 37s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-devel-functional-devstack/dad0691/ : SUCCESS in 44m 48s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/2d2186e/ : SUCCESS in 48m 18s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack/779dfa0/ : SUCCESS in 1h 24m 39s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-networking/b383093/ : SUCCESS in 1h 20m 36s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-senlin/dccdb9f/ : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-magnum/67fedfd/ : FAILURE in 1h 06m 33s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-masakari/9624603/ : SUCCESS in 38m 02s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-ironic/7015aaf/ : SUCCESS in 36m 09s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-python2/3459ab1/ : SUCCESS in 58m 46s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/osc-functional-devstack-tips/fe85bb2/ : SUCCESS in 56m 34s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/check/nodepool-functional-openstack-src/29563f7/ : SUCCESS in 34m 16s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/check/bifrost-integration-tinyipa-ubuntu-xenial/492388e/ : SUCCESS in 19m 47s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/2/check/metalsmith-integration-openstacksdk-src/4db9f38/ : SUCCESS in 54m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8ee244f427e6bd295ad214d279cc7f33e21c32c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-24 11:38:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5f1343b74a6b4719402c12fac15324df4342ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:38:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ef46c429b5859203eb723732b5b267ffd3562e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:23:13.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/gate/openstack-tox-lower-constraints/ab6a86d/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/gate/openstack-tox-pep8/b5d6991/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py27/9da22be/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py36/38cef3d/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py37/57b0075/ : FAILURE in 5m 03s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips/1c2c562/ : SUCCESS in 1h 09m 20s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips-python2/64614c2/ : SUCCESS in 1h 00m 29s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-tox-py36-tips/e848d1b/ : SUCCESS in 4m 54s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py27-tips/1096b58/ : SUCCESS in 4m 30s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py37-tips/2b917eb/ : SUCCESS in 5m 22s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py27-tips/8749e51/ : SUCCESS in 5m 25s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py36-tips/69218c3/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/gate/openstack-tox-docs/9a6a030/html/ : SUCCESS in 4m 46s\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack/fc4daa6/ : SUCCESS in 1h 05m 23s\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-python2/0411490/ : SUCCESS in 56m 43s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-networking/a87f724/ : SUCCESS in 1h 41m 54s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-senlin/abfea83/ : SUCCESS in 51m 22s\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/gate/nodepool-functional-openstack-src/ce863d1/ : SUCCESS in 37m 34s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/gate/bifrost-integration-tinyipa-ubuntu-xenial/a3749a7/ : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2cb5889f20f085da8d0f13679fd4825f8e2644d8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-24 13:27:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"554616a083647421aba7688eef0b82d374cecea3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 14:56:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/check/openstack-tox-lower-constraints/569dbf2/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/check/openstack-tox-pep8/0bf2bbe/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/check/openstack-tox-py27/79eceb9/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/check/openstack-tox-py36/dce9d01/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/check/openstack-tox-py37/62a76df/ : SUCCESS in 4m 23s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips/9d4b9d3/ : SUCCESS in 57m 51s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips-python2/98d1b0d/ : SUCCESS in 1h 04m 40s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-tox-py36-tips/e6b6d2f/ : SUCCESS in 3m 57s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py27-tips/35dfacc/ : SUCCESS in 3m 38s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py37-tips/7311cf1/ : SUCCESS in 5m 12s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py27-tips/9beff5d/ : SUCCESS in 5m 45s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py36-tips/e514823/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/check/openstack-tox-docs/a3650e7/html/ : SUCCESS in 5m 07s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-devel-functional-devstack/87eb6ce/ : SUCCESS in 46m 02s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/3f066ec/ : SUCCESS in 41m 59s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack/5d10265/ : SUCCESS in 55m 28s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-networking/a0f7001/ : SUCCESS in 1h 23m 31s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-senlin/18d199f/ : SUCCESS in 58m 49s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-magnum/78700db/ : FAILURE in 58m 32s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-masakari/e4589b1/ : SUCCESS in 36m 53s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-ironic/8f80fab/ : SUCCESS in 35m 21s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-python2/068ca2c/ : SUCCESS in 1h 01m 48s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/osc-functional-devstack-tips/4839635/ : SUCCESS in 56m 58s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/check/nodepool-functional-openstack-src/2e00a67/ : SUCCESS in 33m 00s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/check/bifrost-integration-tinyipa-ubuntu-xenial/bb25105/ : SUCCESS in 30m 37s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/2/check/metalsmith-integration-openstacksdk-src/0fb3bf6/ : SUCCESS in 56m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b87d70c5f3ce355e73f6cf9b17e5e547b7923aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 14:56:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed70abd16bbc95091197baa80105d775129532d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:19:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/gate/openstack-tox-lower-constraints/91b3f15/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/gate/openstack-tox-pep8/7e9c82d/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py27/b5946eb/ : FAILURE in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py36/a6c6d24/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py37/e6ddcb6/ : SUCCESS in 6m 36s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips/2b31755/ : SUCCESS in 1h 00m 04s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips-python2/860de0d/ : SUCCESS in 1h 02m 15s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-tox-py36-tips/420ad1b/ : SUCCESS in 4m 46s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py27-tips/64cb053/ : SUCCESS in 4m 09s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py37-tips/5e79f78/ : SUCCESS in 6m 16s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py27-tips/865628c/ : SUCCESS in 5m 36s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py36-tips/978fdb3/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/gate/openstack-tox-docs/23ea484/html/ : SUCCESS in 3m 52s\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack/f4738a8/ : SUCCESS in 1h 00m 56s\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-python2/1266c16/ : SUCCESS in 59m 04s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-networking/cb27bbc/ : SUCCESS in 1h 19m 06s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-senlin/bec46b0/ : SUCCESS in 55m 32s\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/gate/nodepool-functional-openstack-src/159669f/ : SUCCESS in 35m 12s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/gate/bifrost-integration-tinyipa-ubuntu-xenial/20c9c97/ : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a09ae85c9ccde2bada4837b3b2c81f0d454981f3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-24 23:40:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5c575348a2409077ac10cd91c2efa17477705f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 00:58:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/check/openstack-tox-lower-constraints/2f268ec/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/check/openstack-tox-pep8/4fb37d6/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/check/openstack-tox-py27/3763e72/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/check/openstack-tox-py36/2814419/ : SUCCESS in 2m 53s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/check/openstack-tox-py37/44c8ce4/ : SUCCESS in 3m 20s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips/5914de7/ : SUCCESS in 52m 43s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-tips-python2/d1eafbf/ : SUCCESS in 55m 21s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/check/openstacksdk-tox-py36-tips/f7794ce/ : SUCCESS in 3m 40s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py27-tips/f8cb48c/ : SUCCESS in 3m 35s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/check/os-client-config-tox-py37-tips/c660830/ : SUCCESS in 3m 37s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py27-tips/9908871/ : SUCCESS in 4m 43s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/check/osc-tox-py36-tips/1cfcc25/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/check/openstack-tox-docs/9b3f233/html/ : SUCCESS in 3m 39s\n- openstacksdk-ansible-devel-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-devel-functional-devstack/ebadddf/ : SUCCESS in 42m 41s (non-voting)\n- openstacksdk-ansible-stable-2.6-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-ansible-stable-2.6-functional-devstack/86a657d/ : SUCCESS in 39m 00s (non-voting)\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack/82e29a2/ : SUCCESS in 53m 31s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-networking/f492c64/ : SUCCESS in 1h 16m 29s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-senlin/55f211d/ : SUCCESS in 52m 50s\n- openstacksdk-functional-devstack-magnum http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-magnum/21c583c/ : FAILURE in 56m 05s (non-voting)\n- openstacksdk-functional-devstack-masakari http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-masakari/847f13b/ : SUCCESS in 31m 35s (non-voting)\n- openstacksdk-functional-devstack-ironic http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-ironic/85209dd/ : SUCCESS in 39m 50s (non-voting)\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/check/openstacksdk-functional-devstack-python2/2a4b432/ : SUCCESS in 56m 15s\n- osc-functional-devstack-tips http://logs.openstack.org/90/672190/2/check/osc-functional-devstack-tips/9a0ef53/ : SUCCESS in 1h 00m 07s (non-voting)\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/check/nodepool-functional-openstack-src/7f8c314/ : SUCCESS in 28m 55s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/check/bifrost-integration-tinyipa-ubuntu-xenial/6c57251/ : SUCCESS in 20m 38s\n- metalsmith-integration-openstacksdk-src http://logs.openstack.org/90/672190/2/check/metalsmith-integration-openstacksdk-src/82e8f22/ : SUCCESS in 47m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45d4c8cffdb23e9d4478ceed067c3f46987e521a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 00:58:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"028837f1eb654af368b4bef0ebec1708ce7810ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 02:20:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/672190/2/gate/openstack-tox-lower-constraints/0e10c7a/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/90/672190/2/gate/openstack-tox-pep8/b98e530/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py27/31f8ec8/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py36/77efc71/ : SUCCESS in 2m 57s\n- openstack-tox-py37 http://logs.openstack.org/90/672190/2/gate/openstack-tox-py37/67d28da/ : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips/f994a15/ : SUCCESS in 57m 57s\n- openstacksdk-functional-devstack-tips-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-tips-python2/fd25796/ : SUCCESS in 53m 29s\n- openstacksdk-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/openstacksdk-tox-py36-tips/a5afde4/ : SUCCESS in 3m 12s\n- os-client-config-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py27-tips/7e67351/ : SUCCESS in 3m 15s\n- os-client-config-tox-py37-tips http://logs.openstack.org/90/672190/2/gate/os-client-config-tox-py37-tips/1f18244/ : SUCCESS in 3m 35s\n- osc-tox-py27-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py27-tips/26dcae1/ : SUCCESS in 3m 47s\n- osc-tox-py36-tips http://logs.openstack.org/90/672190/2/gate/osc-tox-py36-tips/8f3cf59/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/90/672190/2/gate/openstack-tox-docs/2dcbc8d/html/ : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack/663a4df/ : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack-python2 http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-python2/39f26a5/ : SUCCESS in 1h 04m 12s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-networking/543929e/ : SUCCESS in 1h 16m 37s\n- openstacksdk-functional-devstack-senlin http://logs.openstack.org/90/672190/2/gate/openstacksdk-functional-devstack-senlin/4065189/ : SUCCESS in 50m 19s\n- nodepool-functional-openstack-src http://logs.openstack.org/90/672190/2/gate/nodepool-functional-openstack-src/b278264/ : SUCCESS in 32m 27s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/90/672190/2/gate/bifrost-integration-tinyipa-ubuntu-xenial/a7f4739/ : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"693b653428e597e2adcc4950fa5930ce3aeb7f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 02:20:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6548a961b9a4a5e621eb238226d47b6632ca9188","revisions":{"dcc5bc8943c459f20a2f86b7cb59597edc413ba5":{"kind":"REWORK","_number":1,"created":"2019-07-23 00:03:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/90/672190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/90/672190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/1"}}},"commit":{"parents":[{"commit":"6a2668724551cd60589d018bd8962840e9f9a9a4","subject":"Correct endpoint_override discovery for service with dashes in their type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a2668724551cd60589d018bd8962840e9f9a9a4"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-07-23 00:01:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-07-23 00:03:30.000000000","tz":120},"subject":"Add set-boot-device to baremetal","message":"Add set-boot-device to baremetal\n\nThis patch adds the set-boot-device function to the baremetal\nmodule.\n\nChange-Id: Ia936cd41b87dd9422b00e96ee541f2c3842dc636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/dcc5bc8943c459f20a2f86b7cb59597edc413ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/dcc5bc8943c459f20a2f86b7cb59597edc413ba5"}]},"branch":"refs/heads/master"},"6548a961b9a4a5e621eb238226d47b6632ca9188":{"kind":"REWORK","_number":2,"created":"2019-07-23 09:43:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/90/672190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/90/672190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/90/672190/2"}}},"commit":{"parents":[{"commit":"6a2668724551cd60589d018bd8962840e9f9a9a4","subject":"Correct endpoint_override discovery for service with dashes in their type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a2668724551cd60589d018bd8962840e9f9a9a4"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-07-23 00:01:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-07-23 09:43:46.000000000","tz":120},"subject":"Add set-boot-device to baremetal","message":"Add set-boot-device to baremetal\n\nThis patch adds the set-boot-device function to the baremetal\nmodule.\n\nChange-Id: Ia936cd41b87dd9422b00e96ee541f2c3842dc636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6548a961b9a4a5e621eb238226d47b6632ca9188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6548a961b9a4a5e621eb238226d47b6632ca9188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
