)]}'
{"id":"openstack%2Fopenstacksdk~713551","triplet_id":"openstack%2Fopenstacksdk~master~Icfb92c267a10c7526e91d678e00fc522058a92bc","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Icfb92c267a10c7526e91d678e00fc522058a92bc","subject":"Add Octavia quota to the SDK docs","status":"MERGED","created":"2020-03-17 22:55:44.000000000","updated":"2020-03-18 00:29:10.000000000","submitted":"2020-03-18 00:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713551-1584491264805-cb471cbf","meta_rev_id":"a1e18d436198710b9be12462bbd397493c9ee4a5","_number":713551,"virtual_id_number":713551,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 00:27:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-03-17 23:05:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-03-17 23:05:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 23:05:30.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-03-18 00:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc3b70315193aecb2a2cd664fec25339f25f4818","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-17 22:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1715121d5aa9aaf2202e0433c0b43f2e018e5290","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-17 23:01:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0ae727518e996f46aeaf8e556d84991867780cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-17 23:04:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b9cc38bdb8b95260feea20611972e059c1cd445","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-17 23:05:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d4d3f7b1840b37309e854dd30c6da82deba1a90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 23:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5681eb6a6be4142a1f503efbe510a3f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fbd2d31c2244debbff460ff0a4b66f : SUCCESS in 8m 55s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/cd1723a2f3d64f419e2bc8f34bcba60c : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e4bd9fd2463acd81fd128fb5b4f3a033383d43f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 23:43:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c472ce53700de9bfdf7069809f815d65e4c10dd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 00:27:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"97078c2e6150ec645ae751d0a836fb502553b7d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 00:27:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f157fff6e14d91acb86347247c8a51 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be8a60a7d764c04acdd5d164dddae47 : SUCCESS in 8m 52s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/6872d345bc104eef9c8495ff9561cb4a : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a1e18d436198710b9be12462bbd397493c9ee4a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 00:29:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770d9d289064406882d29842ef59ebc3 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36fe79f93c562775621b2160ee337e4bf795cbd7","revisions":{"582c88d8df7de1e5710694f13e8cb2d620aa8f54":{"kind":"REWORK","_number":1,"created":"2020-03-17 22:55:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/51/713551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/51/713551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/1"}}},"commit":{"parents":[{"commit":"d041357f99272ed9c85211c20d0d1adbd0f3e58f","subject":"Merge \"Add description field to portforwarding NAT rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d041357f99272ed9c85211c20d0d1adbd0f3e58f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-17 22:54:33.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-17 22:55:43.000000000","tz":-420},"subject":"Add Octavia quota to the SDK docs","message":"Add Octavia quota to the SDK docs\n\nThe patch that added Octavia quota to the SDK forgot to update\nthe documentation.\nThis patch corrects that by adding the required section to the\npage.\n\nChange-Id: Icfb92c267a10c7526e91d678e00fc522058a92bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/582c88d8df7de1e5710694f13e8cb2d620aa8f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/582c88d8df7de1e5710694f13e8cb2d620aa8f54"}]},"branch":"refs/heads/master"},"36fe79f93c562775621b2160ee337e4bf795cbd7":{"kind":"REWORK","_number":2,"created":"2020-03-17 23:04:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/51/713551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/51/713551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/51/713551/2"}}},"commit":{"parents":[{"commit":"d041357f99272ed9c85211c20d0d1adbd0f3e58f","subject":"Merge \"Add description field to portforwarding NAT rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d041357f99272ed9c85211c20d0d1adbd0f3e58f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-17 22:54:33.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-17 23:04:33.000000000","tz":-420},"subject":"Add Octavia quota to the SDK docs","message":"Add Octavia quota to the SDK docs\n\nThe patch that added Octavia quota to the SDK forgot to update\nthe documentation.\nThis patch corrects that by adding the required section to the\npage.\n\nChange-Id: Icfb92c267a10c7526e91d678e00fc522058a92bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/36fe79f93c562775621b2160ee337e4bf795cbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/36fe79f93c562775621b2160ee337e4bf795cbd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
