)]}'
{"id":"openstack%2Fopenstacksdk~735858","triplet_id":"openstack%2Fopenstacksdk~master~Ifca2676ef89a81d373e3bebf58c3c25880ed1905","project":"openstack/openstacksdk","branch":"master","topic":"retire-neutron-fwaas","hashtags":[],"change_id":"Ifca2676ef89a81d373e3bebf58c3c25880ed1905","subject":"Remove neutron-fwaas from the jobs\u0027 required project","status":"MERGED","created":"2020-06-16 10:32:27.000000000","updated":"2020-06-25 18:47:13.000000000","submitted":"2020-06-25 18:40:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735858-1593110454455-47516c0e","meta_rev_id":"2a5a92a3a09e38a7e2acc0507646ce73834a518e","_number":735858,"virtual_id_number":735858,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 18:40:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-06-24 14:15:17.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-06-24 18:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-25 06:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-06-24 20:41:13.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 18:09:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-06-24 20:41:13.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-25 06:23:37.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-06-25 18:40:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"091866bd24037234b2d80c175f0d548232ca4bf4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 10:32:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcdea6ab7fbd0ffb5baf6d38d0377a98ce431819","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-16 12:45:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96dfe0b1b2ce211d6353abf6874a4b96b17c3ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:18:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c51403092f499fa22fa69b4fd47efa : SUCCESS in 10m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ddf62fe4487c4d51a9ad1aabe92d6d04 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2df6d3312844ffa4d5950381a3fe89 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc6e18a874c442eab13b9b5f326ab9ba : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a7acfad67bd40d99621e7f97e86ab28 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/432e195d3a6b4bb6b6eb30026826a1a8 : SUCCESS in 8m 25s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/660c087635fc479c8362999fd823c152 : SUCCESS in 53m 46s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/84bfd94539a84406ab5b23dc7f121af7 : SUCCESS in 9m 16s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1560361c2b3540a68e2cfdf5d0c0bfec : SUCCESS in 8m 34s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/d35728fc7cba4b709cd6b918058e1c4b : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4dc4315c9e843a8a05ca68c82023538 : SUCCESS in 10m 10s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/e52e0b9581d244c186a9277cf14defe3 : TIMED_OUT in 1h 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83c3df9b2b124521bbb59f5c3a0aa512 : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2bdd549743d74fa99952639bf1a1d1ed : FAILURE in 32m 24s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/b7483365560f4230a6a0c9083287a24a : SUCCESS in 56m 43s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/16625ec8661e446ba6f6b8222612854a : SUCCESS in 56m 12s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f973d65b87e34e37aad24c43b21ec065 : FAILURE in 34m 06s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/032b30b72db74969942428d9a0f78a3f : SUCCESS in 43m 34s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ab1ad7c949294908a8a4d1891ae1c1ff : SUCCESS in 57m 10s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c1325be28d904b51bcfdf1c1e819edc4 : SUCCESS in 1h 02m 04s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37c4807cc88646f8a9b98d8366266069 : SUCCESS in 21m 21s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/83df84fe570c445581cf5b1224b2c656 : SUCCESS in 41m 00s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0bcbfc39495440e39f86eb30ed6d318a : FAILURE in 36m 51s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/937dccf97b2e4b85a88ecb5711f17dcc : FAILURE in 43m 41s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/cdfb4197a406433986467ac946b885e4 : FAILURE in 40m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fcf081d4c3d445840e14ecd1731395c9f2fca04","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-18 21:58:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab5cab564767b8ae488dcc252d9a0d2513c08e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:29:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14c4e48518104d5fa539d0e32fbfd29c : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/025fe78aea4149d381b1f77dddccff98 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b2b1826afa4081a9e4df4dabdfddcd : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcd872200eaa463183a7a1cb3f2c0830 : FAILURE in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28a8a542b8c549f89351ea323359f0d7 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/817536228097494c8b73330e709c050f : SUCCESS in 8m 15s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9f334d520bfc42f7af4abc7705ec0d3c : SUCCESS in 53m 47s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c35720de1d7e424b9122f718de7870b7 : SUCCESS in 7m 07s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/42958d741ac341f8ab3869dcc7c35066 : SUCCESS in 5m 41s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/d5eb227d26b14dfcba089bf61e209753 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a72ec2c695462faa9efd46eaa80e07 : SUCCESS in 11m 45s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/19010e4c114f4a33becb584ad0845f32 : SUCCESS in 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5967c8778cbb4b13af1e85244611937f : SUCCESS in 49m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/75af9476e45644148fe31fa47d783fa2 : SUCCESS in 1h 13m 58s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/b810eb602746435794ed25d76316fbdf : SUCCESS in 51m 57s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5ce613b95a6c42848b36df6264600eee : SUCCESS in 49m 27s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a76894052c35469886682491469e28f2 : FAILURE in 32m 39s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/fcff723130f34e9d9ccdcecff0b77c51 : SUCCESS in 42m 57s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/719d00636eea47f9aec5bbb1c90f964e : SUCCESS in 52m 22s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a40849c7496849f291ec3acc00608d0f : SUCCESS in 1h 19m 11s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6952040370e54791af06b840d2b3e44c : SUCCESS in 18m 47s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/332648162b5c4d18bedd77ee4a5483cb : SUCCESS in 1h 06m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6eeccbb4568a4a8486e9f06bd06ab7fc : SUCCESS in 41m 48s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/23236766745e4481961228ab5cf68b8f : FAILURE in 33m 45s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/75ee7847ffa041ed9205a2db793e6eae : FAILURE in 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27315b936c196eed37b33d7f3f934ce17f073bb5","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-22 14:24:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4feaab23c32048611703d29fb6476e45328756be","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-22 14:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b44ad7c9ed184a686de5629359583d0b9fc4cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:44:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7a071c7bd5843e7b7e32af77121ec56 : SUCCESS in 7m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb6f16c4ca21475f83600b0998907434 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63037511b3ef4df9970287eb6e991bd8 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e2b634320514f9bb7845bb8755fe745 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/027b60d1df6744be9895c71c54bb0973 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd70842275734d078af9edb547abeb32 : SUCCESS in 5m 03s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7ffa5a60518e40e7b1c74630de89f8ed : SUCCESS in 57m 26s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6967de69c66941af88cf556166518ce8 : SUCCESS in 6m 33s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6da876358b374bcdad1be4fab3132ed3 : SUCCESS in 6m 53s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/50ddf00c91424c9787b408a7ee93bd3d : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447aa58e8fc24e3dad720650b0c878ae : SUCCESS in 12m 35s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/8555252f9c174cdf938499fe4a0f458f : SUCCESS in 36m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15ec35feba66461cb50d9adcd09b0534 : SUCCESS in 1h 10m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/246dbb13a7a84c9cb908a387b8c3c3a3 : FAILURE in 1h 35m 06s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/9ae90065199347ffbad2e1e8266aacd0 : SUCCESS in 1h 03m 20s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/baf28e7482c3460aa73871f380bcf6d3 : SUCCESS in 1h 08m 24s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c85447b9729d4680be91a6c2495e087a : POST_FAILURE in 29m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b6f3a37967344504b5828bd10713c123 : SUCCESS in 38m 31s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/59640bcfa71f41e187f520a190ba3241 : SUCCESS in 1h 05m 13s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1130663352454d6db7d0b8e70dfecfd7 : SUCCESS in 1h 03m 19s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91101a52001e4ffdb6786f33e5d03a9b : SUCCESS in 18m 35s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2415713635274084a1e45d5185439f73 : SUCCESS in 53m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c4e305b9015d4e358e3e6107e831f6e0 : SUCCESS in 44m 42s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/ec1984406b2a4b75b88c67205768cd98 : FAILURE in 36m 05s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/754c8b40b1c74d73b0ef1319cc8113f9 : FAILURE in 41m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3abc80ceb847cf422f6073e35014dec9f15a500","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-22 20:19:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2c8ac7b30024b8d4487e7cd2e82a159bc28c168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:36:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/856e99fbf05041e8b3ca505b7b14761e : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8fc576a692cd4a90b5b8d5f215e8dfc5 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fc668ce62641f588cf3377560c6023 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a01fe218c7e421b907c6411a0362b97 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38287dd2ff1d443ba6f54052176aeabf : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed8ac3a11ac4eeb88cccca06498a659 : SUCCESS in 4m 13s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5732443248194c8ca43bed29e26c2b3f : SUCCESS in 1h 03m 42s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b5a594d3b48b474e96435cb8067643f9 : SUCCESS in 4m 51s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e3fb33c2cc24efe9122617319977325 : SUCCESS in 4m 04s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/fee1ee714938459da1b4fc17556a7631 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55f179e50d84359ab62cd0edcb3cfac : SUCCESS in 7m 19s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/98248c50b5a04fbb91a5b534e25e6656 : SUCCESS in 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e10b26a9d43c4e61ab0d4fa1cbdb4518 : SUCCESS in 47m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5f509d0d962434ea87660f257c2b886 : FAILURE in 1h 15m 13s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/cbe2f171aff649f5ae4037ec3b5c44bf : SUCCESS in 51m 23s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7cae05b5acdd4b9fb596354f03168d63 : SUCCESS in 54m 39s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/13c95a6bd8654c718bb37e02e9fb1884 : POST_FAILURE in 30m 42s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/97ac9f0d37434b5c95e523952d486b2a : SUCCESS in 38m 41s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fb22db094d734c1e93ee7058521fb722 : SUCCESS in 50m 04s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d84781e2b82e49b5a9528e8e6af16cd3 : SUCCESS in 1h 00m 23s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f616eed0541d4a219f8f741047ec9cdb : SUCCESS in 15m 12s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/68b3c115262247eb879036fbe057913d : SUCCESS in 54m 39s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b7b04bfc0dc24eb7b1174ac483e470ee : SUCCESS in 43m 29s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/8670d54fb57e4667954901aab5896e5b : FAILURE in 35m 23s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/6b19432c74f94cf4aed295f93e04c282 : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d158996bd2bf5f22c9830b2fc0e17fd0ea4cb7dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-22 21:48:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"79d596b8e6ce20a7b2f5bf1e7a25d3c7ef1edecd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-22 22:17:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dc7c34182217762d00d13988c9dbf32526bcec5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 23:21:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/586089eff1ee4f04b898507997d49ff8 : SUCCESS in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba351d290cb54d29a182fa57b0d82358 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16753569fba4460b80fbadfeedcff3e0 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47224e48c5554e0fb21a4a61123d1fa9 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96c740d39f1040eaadfe0b5955c9fb02 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb1b255b9234ceba1f2d91e89249356 : SUCCESS in 4m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e23e206f809d4574a463f190752adfec : SUCCESS in 52m 51s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e3ba101f3e654df1aa232746d86d3a48 : SUCCESS in 5m 41s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a91f1d52c811493da1928e77c8442221 : SUCCESS in 4m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/305ef8b5e3274de1bb32eb8c0c4fcde0 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61dc58a882a74d5c95e170b40cfccb41 : SUCCESS in 9m 52s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/2ded63dac4d94647b65188cb7bc78e79 : SUCCESS in 35m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0069bbbc754d4fee9753fe472fc4c200 : SUCCESS in 1h 24m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/21db4ea6d8d6450ba6e00e9e19397d42 : FAILURE in 1h 13m 37s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/3d3c02fa29724dba880e181c674d779d : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/81fd5150cc3a492c90cff11ed13d1f0d : SUCCESS in 50m 14s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/2854a6cdd15d40fd82e2ab738735d761 : POST_FAILURE in 31m 02s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3936af2084dd489590152c2886b14d4c : SUCCESS in 37m 48s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/31ad6ab1d26e47bfb56d1dc26261d51c : SUCCESS in 51m 42s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/803a303bbf68445a8ac2a9c9644b4e4e : SUCCESS in 1h 01m 44s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0f2daa5188f45b8825272a7722721ee : SUCCESS in 15m 33s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/03db5def28cb466abd3a13237b8581b0 : SUCCESS in 56m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/fb080edfb8cc4966b4a9d9b1e9b9251a : SUCCESS in 44m 19s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/ebc15960d71a4984b13604b036c35bd1 : FAILURE in 33m 17s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/460a35baa44b4ee8b597423324e6585f : FAILURE in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0656305e3c6bf7c54f78a63d1a24c2150bb9f9bb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-23 06:46:13.000000000","message":"Patch Set 2:\n\nrecheck octavia unstable?","accounts_in_message":[],"_revision_number":2},{"id":"622aac2438162ba1972572bc75af50572564f40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:26:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b675ea2ebf094170bc64f043da740dea : SUCCESS in 6m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f8cfe3b790e472d92381e6bf180e0e7 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec81ae7a6dd4f28a36e955ebd5b075c : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e3b275fcf704b82b9a2d93467e4d3bb : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a323d406515b49b8b3d7f235d94a6eb2 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff516930f314db3a33fd5d135ee6933 : SUCCESS in 6m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/23c3db21d919451fa888252eae946ec6 : SUCCESS in 1h 08m 01s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0bb16289e0de4d67ae2e624328d101a6 : SUCCESS in 5m 32s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c0eaf15ce52d4e39a9cec34e38599762 : SUCCESS in 4m 56s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/fa3fe0bacf314a7cb484bd00d1a7b794 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85fa2a61390146a99234e1d621e2f02a : SUCCESS in 8m 16s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/1b3ec4d7b65d4327bf8c615f995c9872 : SUCCESS in 32m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc183e39b6114f1681bfc018847c70e1 : SUCCESS in 1h 17m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/110222aba61c450d8fac7c56ebc8d330 : FAILURE in 1h 29m 33s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/7bad1ce2e307448989fd17840d4f2bb7 : SUCCESS in 51m 35s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ab90f0266ba54641b77e94b3d08b0899 : SUCCESS in 51m 40s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1146172529194b4b8d7a16ab5d11eff8 : POST_FAILURE in 31m 11s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/35ffac223bae48e887245df3587be7dc : SUCCESS in 44m 42s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dac27ca7caa94d749959935fe79e41e9 : SUCCESS in 1h 00m 40s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3b9b59e7641d4c8a9be172de2a33ad8e : SUCCESS in 53m 49s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51b589a18d7f46049396dc5df89dbc61 : SUCCESS in 18m 06s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/09b355221fe848d1b531138711b3e076 : SUCCESS in 54m 09s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/39eaff7812e84bc08463707d2bd76f1d : SUCCESS in 42m 58s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/5930585b1237428b940f0f6355353237 : FAILURE in 36m 45s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/d9f1d2582c3f4f80b54a318c52d0fc38 : FAILURE in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ca783c746a9c975a6463fd27bfc25b10958b3c1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-23 09:12:02.000000000","message":"Patch Set 2:\n\nNot unstable, but broken, see https://review.opendev.org/737460","accounts_in_message":[],"_revision_number":2},{"id":"5bac0409617666f0b79760c89e4c82da2c828495","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-23 09:14:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6e16704f6944c064c5d4fcf6f18b84d83a86aa2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 11:19:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac0994cc2ef44094a1b9f5af34c789ef : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/757ead43ce8948c5be869ec98f67c480 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f4ec4d319d4065a657a19a19f3a94b : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12177841bf404f198709eebce6e2a0a7 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0eb8c2e2bed3425ca5d831e5ea507f37 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/460c191d98b34f21b3a3a9fcbf5faca8 : SUCCESS in 5m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/91c8ffe1a2a544118e10627fc3575f54 : SUCCESS in 50m 39s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/cf3bc0cdff9b45ae9a53ab1447d0ce9c : SUCCESS in 5m 46s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/367a1b7ab78348a3807121e6bc7c1e5f : SUCCESS in 5m 03s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/26e0cdcb4c9f46afb7baf392915aa399 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaee3a1794fe4004a4615321290ee9ff : SUCCESS in 11m 26s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/02e118aa223d4bb79d03a3d73ed5a782 : SUCCESS in 35m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d4e8c984c504e2aa79f47d19539e130 : SUCCESS in 58m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/775239062c994381b8d18f2292b5a375 : FAILURE in 1h 09m 23s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/1618a4ae3a524db78dc0c59b3a7f89ab : SUCCESS in 56m 56s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e02f7d7dcb3c41cd81c389df1f2a4e68 : SUCCESS in 59m 55s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b75f872faa18438592efb2dd637f204a : POST_FAILURE in 30m 43s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a1d5780e52084e9582146e40dd2c466b : SUCCESS in 43m 40s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6f2342b6ad1946fd80bad09e409cf4d1 : SUCCESS in 47m 16s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/64c49543a15a463dab6d965348789c35 : SUCCESS in 1h 12m 39s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90c03b7964ed457fa711432c53ebb617 : SUCCESS in 16m 40s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1ae0b260348548b4857f5d2d1995c111 : SUCCESS in 1h 02m 11s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f3977d8858e849eeab58deb01169c653 : SUCCESS in 47m 17s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/68a40f0a77394b3d9120e5ad186da041 : FAILURE in 29m 22s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/e4bdb077c6aa405ea611a8b2dd733f71 : FAILURE in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d79b6c36d3057321403d5862050b43bf62876b1","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-24 06:53:57.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7675d070ebc20a5f6f48a721edee159e464dd2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 08:44:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56e9552304b74825bcecfcd3a6d6801d : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff1f91b2893f471e8edab51a9c47cc83 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88bd84391a74433a1407086373a3679 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/501cca23a4974adf8b37816d47ee9840 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21afd345a4574da78eaa791d0bbaff8d : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f73b50b780b84562951915ecf8382a9f : SUCCESS in 4m 38s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e288df94000a46779e6d77e285d26159 : SUCCESS in 55m 47s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/580a3043daf54da9a9f092bd76d42da3 : SUCCESS in 4m 57s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/762f27ecb9254046950bd635848aa659 : SUCCESS in 4m 09s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/2021f5829dc94b7cafd68ef5813fbf6b : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0202b7544ff54d6fbfc77620f1b8fd52 : SUCCESS in 10m 14s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/1bbeb59940da43778db96d9a3cd68a9f : SUCCESS in 36m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c17820c93a7d41c6842c1abb9c9ffde6 : SUCCESS in 58m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/31c9cf89f0784c2e93fb660b3086623f : FAILURE in 1h 18m 16s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/f0e96640bf7d4f4e96b3d8e4a9ff3920 : SUCCESS in 48m 05s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/23c5f84a6fc0420cb6851994538f9e60 : SUCCESS in 49m 06s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/00cca1cf4b244ecdad1f15646027ab15 : POST_FAILURE in 40m 37s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/71dded4cb4524f2ea70a72dc3051b850 : SUCCESS in 37m 20s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b26dc4287c3d4e9f8d4f059c61a7892e : SUCCESS in 56m 13s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0b12af5e929943cb8d8534bdb2de2c8f : SUCCESS in 1h 01m 10s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf940399e9934833900dd28b5ce4aac2 : SUCCESS in 16m 46s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3ae4ecfdca9643a388e740c66f32ef8a : SUCCESS in 1h 00m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2966a6a6b9344712a2361a3ab4fa9f91 : SUCCESS in 45m 31s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/f7819fb9d4be42c79a1fbac7dd1b4ede : FAILURE in 40m 04s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/b2541e1c0c2b4627b47f6c26e1764495 : FAILURE in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25b75f659c5477159e152d1a0ac477a123c47820","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 10:18:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"53267c330d07dc116eabba6419653b6da60303f2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 10:18:34.000000000","message":"Patch Set 4:\n\noctavia fix was incomplete","accounts_in_message":[],"_revision_number":4},{"id":"a8c150d04ee83fec3749dab0d664305d5bd94f9b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-24 14:15:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"947497fd6c33bda54f57c7c45649ea72991ad1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 17:56:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42506c39f23847d5b6f09b2c911cf085 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7459a4b8cf947a99e709f79ad758706 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df58a479abb043f6ad38c679b93089e2 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef330f6a2c34203b2dfcf2dae33f49e : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77baa7aba40148a89547244bb6fee223 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05985f3101b4457392c8c9fa415134c8 : SUCCESS in 5m 34s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/98261426656a412f95f01071397cc054 : SUCCESS in 1h 05m 06s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a5229570c3ec4fdd87adc9330e09efa3 : SUCCESS in 6m 02s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2690e1d974f466e95d8aae9dae4e137 : SUCCESS in 5m 27s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/1e132224739649fe9ef1d9681b9528ac : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2193c592a994f69b1b2ef9b36475c3c : SUCCESS in 8m 30s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/d8974f0b150645028a3068877a67308f : SUCCESS in 38m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/107fd0800ae949dda4a84a03ea22f783 : SUCCESS in 1h 10m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/877dec98e0d24be581311402d016589b : SUCCESS in 1h 17m 29s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/575735e0f44b45de9fda731b262fd45c : SUCCESS in 52m 05s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5dc5d39df0694e2da8cd1142a238fc06 : SUCCESS in 47m 27s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0e7b90eac9bf4254a252b8c5f8bc4da5 : POST_FAILURE in 35m 48s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/353aea4e9e904667a8c3779a42e39fbd : SUCCESS in 40m 43s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8cc01cea246c4fa58832dbfcd21fe057 : SUCCESS in 37m 15s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0074083bbf7748079f08e9ceac628ef0 : SUCCESS in 1h 04m 46s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f2d9e6b22c44fbabbe21b572b371ea1 : SUCCESS in 18m 08s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a0e47baf70e84516ac7943ef62bf29db : SUCCESS in 35m 34s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/671d99f8ff444948ae4112310964d660 : SUCCESS in 43m 21s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/1618810c44524162b3c197774a55b291 : FAILURE in 34m 14s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/24dd10d9d920403a813432bd383755c7 : FAILURE in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49aa4eb345569503268785ba3f515a09bd935ef0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 18:09:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3131628a5f917daa3286bc63028ab49a2fbc11ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 18:09:58.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"6e4666d85c9f3bafbefe02d991989386725c1a34","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-24 20:41:10.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"530ac41fb1683ab54f22978c32747775e2fb8f22","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-24 20:41:13.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"04487b230a765826f2e263a23696b3b757e2f165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:41:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"44a9acf46d8ad5f9896313ff8a95d0f5318913c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:31:45.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/073c41adbae24a60b39ff12f5c1cf072 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ada29259b7a447284c9aed82830a007 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978282ca038749edbe3a149af15df91e : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce575e6a52e34de4afc77c2908f36d43 : FAILURE in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12dbbc71785743cf99d4e7f86bc890e7 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3a5d4ee7719f4c5fbfb93ea528262fa0 : SUCCESS in 49m 51s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c2475c6dff974645ae1b4fd3c8cc87c1 : SUCCESS in 8m 52s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3f23914ae5d9456f93aa0bb522243e75 : SUCCESS in 5m 40s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e9c3cd5a51484bdcaba015dd2d87481d : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293a66794f3c45e4b42a99172d8f077a : SUCCESS in 11m 41s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/ebcedeab80ad42c18e4c65979265fe78 : SUCCESS in 38m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49120fdc0b934b0d8a389454e7727b08 : SUCCESS in 51m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/58168f25bae446f586f1d22787ba3af8 : SUCCESS in 1h 16m 14s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/2d9ba67d19bb41528e6fe34331cb774e : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":4},{"id":"f6742145df3b50f43e30c0d3a5f20071e94b9f65","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-25 06:23:37.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c44cd69657cc474a191f85938c9726f593c6c1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:01:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d66d61bb31c04a7a9a10383877097e0e : SUCCESS in 9m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b169fa42a10d44ec9f103bc9b280c2aa : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b60a0c1c2d48b5a45b410956cd1f5b : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45bea9bb495945a3abb4925dc1ae4fa3 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9a9f6d0a21746979f0b49ff4aeec11d : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1e722e9dc34eb992734b5ee38dff91 : SUCCESS in 6m 53s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e4346ad6d42c4f329b1d588fd02cce20 : SUCCESS in 55m 01s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b394237cd47647efacf386ed64c91e4f : SUCCESS in 7m 08s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bcd6f4383f76494eb64276022e29a588 : SUCCESS in 8m 05s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e12db2509c3f40d6bc2a582b2f0f42d9 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ea29562c07421cad6b633fc29007c9 : SUCCESS in 12m 49s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/37af340279934a7f96f27c6ad302ef50 : SUCCESS in 50m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b090419de344b48ba58ccdd12990628 : SUCCESS in 40m 26s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1129d6cb004342d5b4f3de8efd13fc77 : SUCCESS in 1h 18m 41s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/b02a3c5e61744cb4a4d40ce6797101db : SUCCESS in 57m 37s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/67ec50819faf431f956ad547724eeeba : SUCCESS in 52m 46s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/929494cce2a74779a6fba617354a83e3 : POST_FAILURE in 29m 37s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9e89679a8a1f49598cd70a824bdc287b : SUCCESS in 37m 11s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/70f4a9cae2524074b0abbefcdf5fce76 : SUCCESS in 57m 52s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/13c1f0e4a81442d7982430aa55ec3b1d : SUCCESS in 1h 15m 19s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e36e17b04b1f46a4aca5ac397abb8649 : SUCCESS in 18m 04s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/e3a539482abc4884b41ba87d847371d1 : SUCCESS in 59m 35s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/bd972883aacd45e788ab806e76592925 : SUCCESS in 48m 11s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/f2570ad222af4f1f8244c391ee296e0d : FAILURE in 42m 46s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/27411c56b73249cc850d367de6e94c81 : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a64bdd189a37b0fa667b559fa0c9d5eda7f6d283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:01:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"574d42c4889853e58783412ec20935f2bca5c87b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:41:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4932c9c081045ef9c14e4e55273a846 : FAILURE in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a8fb434c2ab740b380d05cb6893562dc : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad67b8bda40470b94b9d8ba2112cc48 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1af53e2b04294066bba334756865b5b3 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ad98afa99374ec0b64aba17fc2af355 : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/61bcce54fef14abf8cb7b4b434fc36fd : SUCCESS in 51m 47s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7fd5bb2a7d6747d993771c3b8a935536 : SUCCESS in 5m 09s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71ee1c4c8141471fbd32b6270713f1a8 : SUCCESS in 4m 36s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/b9a6d7f9875a4983a9c3b8bd391544d6 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51460f075eda4c12955320d634a6e256 : SUCCESS in 9m 21s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/a4143a79d8f4430ea74d9202992974f8 : SUCCESS in 42m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd3b0cb551d94d489bf679b7032a0d3d : SUCCESS in 59m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6d5701732bf244348aca4b892d6eea11 : SUCCESS in 1h 17m 09s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/4ccdd19ec57c4106b2006eb5c7a50c94 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":4},{"id":"1f0dde26f102b4579a4486230adf44973207940e","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-25 10:52:03.000000000","message":"Patch Set 4:\n\nrecheck we will get it some time","accounts_in_message":[],"_revision_number":4},{"id":"b9bb4001e0b486171daa6abb72547c41ca438003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 12:47:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4eaa3065c34f408b754ab3dc1468d9 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/860bfcd62c8746be97bf00148d59134a : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96dd6d1f1d348ce83d0cfc99cd125ab : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39715a3280e4492e9ed1b80fec2155b7 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2725c968f9847eabd0fa7fecc5a7b46 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51d38a6f31f494fbdbc19a829853bb6 : SUCCESS in 3m 51s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a8e2dfe89eca4c39be52cdeecab9f40d : SUCCESS in 53m 25s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3384c997ca9d488c9dc6ebc58226ca5d : SUCCESS in 7m 38s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/90bae53c51d4485fbd65fadfef115dbf : SUCCESS in 4m 46s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0596defe8a4e4baf9e8e06e72af7c51c : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f2a1a9973d444ea847ebf9ace208b7 : SUCCESS in 10m 31s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/f9419343671f4a74aad658e4b06a5168 : SUCCESS in 35m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b81345dd08394aa890aafbf1921e6fc5 : SUCCESS in 52m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2a73a5245f7e4f3eb027aaf230ea4485 : SUCCESS in 1h 44m 33s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/8a7a9e15b31246c18a7c5d65b65fbb69 : SUCCESS in 50m 16s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/360926fa82414edc9c4da44cdc9b7daa : SUCCESS in 54m 52s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8b5cd2e522fd414c8c2694f373c9295f : POST_FAILURE in 34m 26s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7fd7383e47964907ba16269fae84c733 : SUCCESS in 36m 50s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/987624ec42d74c8f82ac8bcb8237ec7f : SUCCESS in 45m 47s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/bbdff1a95df54b738d49b011e3db3390 : SUCCESS in 1h 06m 33s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe6a1c1b13ab473ba87b7f0660f8d964 : SUCCESS in 18m 10s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/eb9da58e3f7b4255bd5b25fb48c760b2 : SUCCESS in 1h 00m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5448e8e4d55340eebfbfc69760432a52 : SUCCESS in 52m 05s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/3d8e70e46692401da2b5d1addeac858e : FAILURE in 40m 41s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/b51dfff382f04507a4cea82fddcb457b : FAILURE in 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ba5ab720617e9e2386ffba51827c73ae5670eb2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 12:48:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8afbe855cf8f4e63134da4cfce0ac9fc0af7d3d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:40:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01dbe1e39be74617803b28e0f280eb01 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/568eff7886494f82ad8315fdef145068 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bbad07296a24386a3b9f5cf475eb24f : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfdf95957564c4cafb5ef109bcc9dc0 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61dc10b64bf44e1284cf17be1cc77363 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e06db4ea02934ef5be798518ecee7baf : SUCCESS in 56m 14s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/56a66c75dc694c6e9fa0e33a7453514f : SUCCESS in 5m 52s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8ef8b0d768484236a132f8fa9d5e8df4 : SUCCESS in 4m 21s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/bfbd80e77d834108ad3773a13384445c : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0865f033d44ee4aa000cca8e49f0f8 : SUCCESS in 7m 41s\n- nodepool-functional-openstack-src https://zuul.opendev.org/t/openstack/build/6196764b62f046dba830e10a70d5d1c3 : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4835c57eaee24bf2a3a5635316cc099d : SUCCESS in 40m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/afc5df896e7c48c0b4e354d294764113 : SUCCESS in 1h 14m 32s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/c5ca50f5bf04412d987ac987e2274029 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":4},{"id":"90e127d096aa4bd076586cdb3173f1b98b33db0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:40:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8a92b8ec00e70f6a4173bcac41ea44999e367a05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:47:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b655c3667e144415b2f59b7de867df1c : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b23928b68e1f6e478851fb61f97596e572825337","revisions":{"b8e9694fa02ae482d7afe9da4878c3fd8f117c5c":{"kind":"REWORK","_number":1,"created":"2020-06-16 10:32:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/58/735858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/58/735858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/1"}}},"commit":{"parents":[{"commit":"1a032b8c602212c9ff650fc78a436d9be8af7a4a","subject":"Merge \"NIT: Fix application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1a032b8c602212c9ff650fc78a436d9be8af7a4a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 10:31:51.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 10:32:24.000000000","tz":120},"subject":"Remove neutron-fwaas from the jobs\u0027 required project","message":"Remove neutron-fwaas from the jobs\u0027 required project\n\nNeutron-fwaas is going to be deprecated in master branch with [1].\n\n[1] https://review.opendev.org/#/c/735828/\n\nChange-Id: Ifca2676ef89a81d373e3bebf58c3c25880ed1905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b8e9694fa02ae482d7afe9da4878c3fd8f117c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b8e9694fa02ae482d7afe9da4878c3fd8f117c5c"}]},"branch":"refs/heads/master"},"4862ca554fe137306ff0fa7ae770d7e02ffdedab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-22 21:48:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/58/735858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/58/735858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/2"}}},"commit":{"parents":[{"commit":"8b5d4fe7985d3f9319b23121e290ef96dc801863","subject":"Add query parameter \u0027id\u0027 for security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b5d4fe7985d3f9319b23121e290ef96dc801863"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 10:31:51.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-22 21:48:48.000000000","tz":0},"subject":"Remove neutron-fwaas from the jobs\u0027 required project","message":"Remove neutron-fwaas from the jobs\u0027 required project\n\nNeutron-fwaas is going to be deprecated in master branch with [1].\n\n[1] https://review.opendev.org/#/c/735828/\n\nChange-Id: Ifca2676ef89a81d373e3bebf58c3c25880ed1905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4862ca554fe137306ff0fa7ae770d7e02ffdedab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4862ca554fe137306ff0fa7ae770d7e02ffdedab"}]},"branch":"refs/heads/master"},"7ff5bc15c433d2cef4cf04b81c675ee5fd204bcf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-23 09:14:00.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/58/735858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/58/735858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/3"}}},"commit":{"parents":[{"commit":"8b5d4fe7985d3f9319b23121e290ef96dc801863","subject":"Add query parameter \u0027id\u0027 for security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b5d4fe7985d3f9319b23121e290ef96dc801863"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 10:31:51.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-23 09:13:57.000000000","tz":0},"subject":"Remove neutron-fwaas from the jobs\u0027 required project","message":"Remove neutron-fwaas from the jobs\u0027 required project\n\nNeutron-fwaas is going to be deprecated in master branch with [1].\n\n[1] https://review.opendev.org/#/c/735828/\n\nDepends-On: https://review.opendev.org/#/c/737460/\n\nChange-Id: Ifca2676ef89a81d373e3bebf58c3c25880ed1905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ff5bc15c433d2cef4cf04b81c675ee5fd204bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ff5bc15c433d2cef4cf04b81c675ee5fd204bcf"}]},"branch":"refs/heads/master"},"b23928b68e1f6e478851fb61f97596e572825337":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-24 10:18:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/58/735858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/58/735858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/58/735858/4"}}},"commit":{"parents":[{"commit":"8b5d4fe7985d3f9319b23121e290ef96dc801863","subject":"Add query parameter \u0027id\u0027 for security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b5d4fe7985d3f9319b23121e290ef96dc801863"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 10:31:51.000000000","tz":120},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-06-24 10:18:08.000000000","tz":0},"subject":"Remove neutron-fwaas from the jobs\u0027 required project","message":"Remove neutron-fwaas from the jobs\u0027 required project\n\nNeutron-fwaas is going to be deprecated in master branch with [1].\n\n[1] https://review.opendev.org/#/c/735828/\n\nDepends-On: https://review.opendev.org/737731\nChange-Id: Ifca2676ef89a81d373e3bebf58c3c25880ed1905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b23928b68e1f6e478851fb61f97596e572825337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b23928b68e1f6e478851fb61f97596e572825337"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
