)]}'
{"id":"openstack%2Fopenstacksdk~750831","triplet_id":"openstack%2Fopenstacksdk~master~If3ec085184fbba81f455164b803afb6f66e46fd0","project":"openstack/openstacksdk","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"If3ec085184fbba81f455164b803afb6f66e46fd0","subject":"Fix l-c testing for ubuntu focal","status":"MERGED","created":"2020-09-09 22:28:44.000000000","updated":"2020-09-23 11:50:01.000000000","submitted":"2020-09-10 17:15:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750831-1599758140632-e889b14a","meta_rev_id":"7dcdcf7dbc44555181786f54687912768a9ea45b","_number":750831,"virtual_id_number":750831,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 17:15:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2020-09-10 13:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2020-09-10 13:27:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 13:27:26.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-09-10 17:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f6d7518f8d28fe5b2324b526c1d0d409be58a09","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-09 22:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b609d3d5af1695e4716e639e33908df96110f1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:01:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/36f027ebce834ec4b7c071d7237f3e31 : FAILURE in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8a91f9b5fd44180b1816f0b98e246ef : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61f831321ac4d32a8b94e1d1eda50a2 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b4307d05db34b9d89e9e75290369616 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48e79b0beb674405abff557b194b6f18 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ef62a846da4b1dace816e18c588a6e : SUCCESS in 5m 31s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9212dd53c1d6433d97598b9a7bd4b1fc : SUCCESS in 57m 25s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/185e2ee73eed4ec5b4aec74da293b31b : SUCCESS in 5m 58s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f28a5e3c94af43108de662a624bb4130 : SUCCESS in 5m 40s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0b7ffeef9a724fd099726be2e4fc0225 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7e416de4984667a7f9ff4851fe6834 : SUCCESS in 20m 57s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/97ab4fea82ac4c85971c0930333ef59f : SUCCESS in 1h 03m 44s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/bdb1dcaaa1ae4d93be348ac7edc5c941 : SUCCESS in 16m 52s\n- nodepool-functional-container-openstack-siblings https://zuul.opendev.org/t/openstack/build/d8668c162319451d9489275829afcd85 : SUCCESS in 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2add7a70e064f7480ca762ccd16a75c : SUCCESS in 54m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5b827e8263964b47987a191ff0f43e45 : SUCCESS in 1h 51m 52s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/e39397b78a664880bf1ba2b8655e68df : SUCCESS in 51m 54s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/aeea60db8290430ab31b7cc171eeb831 : SUCCESS in 58m 55s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/12e1e0fec47447fba2c6cf753442608d : FAILURE in 52m 58s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/32b317d3ecf94271893cc008e9181a79 : POST_FAILURE in 58m 22s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/45daa4b89d6d43b9a3cbafa75d36836a : SUCCESS in 59m 05s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/031ccb1ca7524eec86f9e96c88cd905a : SUCCESS in 1h 08m 56s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/339174924ea24b35b1f201f9cf3889e2 : SUCCESS in 23m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b4e9b878c8b64eb982f098a8382edf1d : SUCCESS in 43m 04s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/885091870fcf4f908ebc133d3fdf40b3 : FAILURE in 38m 22s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/9321249f29e74a78a1c7e8ce034dd584 : FAILURE in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a249f4b7275c03644d54c94b7be40e02a77f5500","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-10 02:33:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a37dfc935250da206b4cf02b2be033c2e1a63a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:54:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3086047c53024c88badd5aaa31178069 : FAILURE in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd03d705b9dd4452b6427c31ba92f3e7 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af87780874814661b6454fc8aa2829af : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12cbcdbd0a3e49bd8163d7d1bf6ff018 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab6237fab7a24154b94accd7fb853ddd : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf732ec3c6324020873573f7cda0809f : SUCCESS in 4m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/01ff65812dff4fa287a9e19cfb6def8f : SUCCESS in 1h 03m 44s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d3d5b40ab8cd4f59b46b41fb76ad5359 : SUCCESS in 5m 35s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/302a8fac033c4034ace3695f798b5e1d : SUCCESS in 5m 35s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/b18369cd5efa4767be5b06b664eb4253 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb4d8533b164376ba5378d3f9b9221f : SUCCESS in 14m 58s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/459895e3b7744729a0416c71c3e676cb : SUCCESS in 57m 36s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f91bebdae8ae4126a09e5be385ec93cb : SUCCESS in 17m 42s\n- nodepool-functional-container-openstack-siblings https://zuul.opendev.org/t/openstack/build/e16a90cd9b72480cb8b982aec3179e65 : SUCCESS in 32m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9c04219e6654a6396e0489db8584adb : SUCCESS in 59m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23e1ee1f66ea468fa858a7e9a1ce318c : SUCCESS in 1h 13m 01s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/2e85c94bb8384124b9fa4bfee5197591 : SUCCESS in 37m 49s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/73a4bc01124844a48c2b4b1e0d52ad57 : SUCCESS in 59m 05s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6df1adacaf5847218fd21df6ace6b94c : FAILURE in 47m 55s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0704c09229df423689972005ca7ec6bf : POST_FAILURE in 50m 40s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/01e4e59507d14035848d3a20c5756f56 : SUCCESS in 59m 25s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4bb867082b4244a19242715838a8d250 : SUCCESS in 43m 43s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33422856c94a4fb4b4f69a67c9fc9572 : SUCCESS in 21m 54s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1fd1fb3e9e9a4486b82cb11ca91d9241 : SUCCESS in 50m 49s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/dfbbeefc44d24ea99572b029d7943503 : SUCCESS in 49m 07s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/699a85369c8a44e79d8b4a147eac31ba : FAILURE in 33m 31s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/5c042bf9b3d44f5a85696278ef86015a : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80abf7057499ffb0e93a540f96c36d9def946eb2","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-10 11:03:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3cd8a761e931dfbd3fedc4b97f893c79eab8a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 12:31:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58c34d9a843140beb4ca8a6f283aec5e : SUCCESS in 2m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b79dc1d32244140885cb084ae5f1b0e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5faf5cffba849af8c6afb66bdde5cbc : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3d411b4b0f45c9b2673e2a344f83b6 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a50dbc4685c4c9eb7e18d8d5d920c79 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d17b3fa0122b44cd9d8bc824da7a2da6 : SUCCESS in 4m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0f57963db03949afbe3652676c420fac : SUCCESS in 1h 02m 25s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9e6b262c1bef4088b20f54a29d98f1b8 : SUCCESS in 4m 22s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5fc5a0f6b3e24a79a7a6bdf33fe959a6 : SUCCESS in 4m 24s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/ed78b72b770f4dfabc253e416b4f3d73 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef06108b2fc54266b134ebf75fb452f7 : SUCCESS in 14m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/896c057008ef4cd29c124590c4ea4832 : SUCCESS in 1h 07m 58s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/93bdcc8b84af4783ba8afa3dae755aa9 : SUCCESS in 18m 29s\n- nodepool-functional-container-openstack-siblings https://zuul.opendev.org/t/openstack/build/d45d26bc57b64116b3712b098622ede2 : SUCCESS in 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c15db5894cd490ca0e1debb804ddb26 : SUCCESS in 52m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/28e3a101c8d244b1863802ad95549567 : SUCCESS in 1h 22m 22s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/ce2d4209a7d248a4b042121fa101c961 : SUCCESS in 54m 59s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/65c0948fb3a045048a653f65814ac06f : SUCCESS in 1h 01m 10s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/fe2bd00af2c840d8a75ea8d6ee80caa2 : FAILURE in 47m 01s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/38851fbc72da4d9a84219831d29cd7de : POST_FAILURE in 53m 41s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d96c6c9ce55a465f8a9bac017d5a7cf1 : SUCCESS in 58m 34s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ec940a82ca394416bf103c77faac5a20 : SUCCESS in 53m 38s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/024c2a9e8c25476e8a6559f2f088c41a : SUCCESS in 21m 46s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/312f148356c04ebca80c10ea46030f26 : SUCCESS in 43m 46s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7570c41db88c471486cd6d08bdb1f83c : SUCCESS in 43m 23s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/a281c1695712419ebe7ac966e5eab1b5 : FAILURE in 36m 55s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/5bda29fbc0624373894685303aaccf5c : FAILURE in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"97a4e02747d7c72b3a81cf51e90f04a103005e74","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-10 13:27:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\napproving in a single vote to unblock gate","accounts_in_message":[],"_revision_number":3},{"id":"5829a8f966e9400e90e26f61a47fe026a900d5ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 13:29:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ace63d17ae8bc04fd86a047dc5bc8e17b89246f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:15:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7bde08fa1cd4670999c0acc126a6fff : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/747252f32bc348c2afc5bf28c0e5d8ff : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719be1cacab64be8ac44ec3b573b1d43 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c8a453ef1a3412db58ec67aea98a91f : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9c17d08c1cd4c7f93d3997ce5f037ef : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f949924695b44c39a1a178cca0144a0d : SUCCESS in 1h 03m 35s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d7b62cb85fb74b9998852ca299c9c133 : SUCCESS in 4m 42s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6238920a84cd4925bb44dc84eeaa2e83 : SUCCESS in 4m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/ae740006ab044bdd9feb15a32239183e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc3638e1b394a08910b54f5b9e94acf : SUCCESS in 16m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ce9052f9930343c39a518f67490a29b0 : SUCCESS in 1h 18m 44s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b75fa7dd768642e6b87daafdd7872c13 : SUCCESS in 20m 29s\n- nodepool-functional-container-openstack-siblings https://zuul.opendev.org/t/openstack/build/83f3ce764d664c1ca4655b7d8e406322 : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53ea2b5ed9e4da1b7ef32d22dcd3ee4 : SUCCESS in 1h 00m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/63bbe1701aac4ef694c610c0754260ed : SUCCESS in 1h 16m 43s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/50aea15a52384c70a83796a014bec6b1 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":3},{"id":"61918873464c503a7610e24a09da0b4c438424db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:15:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4063703ab8b06607a3ef204517eee26850f048bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:22:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d65c0fddc0647b19e9d0c383f3e624b : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":3},{"id":"eca9faa1589e0f98f0dcedb642cf9adc99fc8d53","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-23 11:50:01.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 4927a26eae3af1cf03f8577b03400dd3d4582315","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49029b8792947733ae7d4940ef1ab3a09c1cf06d","revisions":{"24d028ea72d87f669cf8b1af535f9fb40df6c9ca":{"kind":"REWORK","_number":1,"created":"2020-09-09 22:28:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/750831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/31/750831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/1"}}},"commit":{"parents":[{"commit":"0ebfffe8356262128362b193f94bca91b0692a71","subject":"Merge \"Add TLS protocol support for Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0ebfffe8356262128362b193f94bca91b0692a71"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 22:28:18.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 22:28:43.000000000","tz":-300},"subject":"Fix l-c testing for ubuntu focal","message":"Fix l-c testing for ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\n- Bump the lower constraints for required deps which added python3.8 support\nin their later version.\n\nStory: #2007865\nTask: #40206\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: If3ec085184fbba81f455164b803afb6f66e46fd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/24d028ea72d87f669cf8b1af535f9fb40df6c9ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/24d028ea72d87f669cf8b1af535f9fb40df6c9ca"}]},"branch":"refs/heads/master"},"8fc7e61e200c8b26921a2135b44b35a355e3c28c":{"kind":"REWORK","_number":2,"created":"2020-09-10 02:33:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/750831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/31/750831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/2"}}},"commit":{"parents":[{"commit":"0ebfffe8356262128362b193f94bca91b0692a71","subject":"Merge \"Add TLS protocol support for Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0ebfffe8356262128362b193f94bca91b0692a71"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 22:28:18.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-10 02:32:52.000000000","tz":-300},"subject":"Fix l-c testing for ubuntu focal","message":"Fix l-c testing for ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\n- Bump the lower constraints for required deps which added python3.8 support\nin their later version.\n\nStory: #2007865\nTask: #40206\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: If3ec085184fbba81f455164b803afb6f66e46fd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8fc7e61e200c8b26921a2135b44b35a355e3c28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8fc7e61e200c8b26921a2135b44b35a355e3c28c"}]},"branch":"refs/heads/master"},"49029b8792947733ae7d4940ef1ab3a09c1cf06d":{"kind":"REWORK","_number":3,"created":"2020-09-10 11:03:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/31/750831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/31/750831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/31/750831/3"}}},"commit":{"parents":[{"commit":"0ebfffe8356262128362b193f94bca91b0692a71","subject":"Merge \"Add TLS protocol support for Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0ebfffe8356262128362b193f94bca91b0692a71"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 22:28:18.000000000","tz":-300},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-10 11:02:57.000000000","tz":120},"subject":"Fix l-c testing for ubuntu focal","message":"Fix l-c testing for ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\n- Bump the lower constraints for required deps which added python3.8 support\nin their later version.\n\nStory: #2007865\nTask: #40206\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: If3ec085184fbba81f455164b803afb6f66e46fd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/49029b8792947733ae7d4940ef1ab3a09c1cf06d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/49029b8792947733ae7d4940ef1ab3a09c1cf06d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
