)]}'
{"id":"openstack%2Fopenstacksdk~777354","triplet_id":"openstack%2Fopenstacksdk~master~I14f0f1ae5421fd0c53980c79d35f5ea2012ea320","project":"openstack/openstacksdk","branch":"master","topic":"fix-trivial-error-in-error-message","hashtags":[],"change_id":"I14f0f1ae5421fd0c53980c79d35f5ea2012ea320","subject":"Fix a trivial error in one of the error messages","status":"MERGED","created":"2021-02-24 10:43:56.000000000","updated":"2021-02-24 22:25:10.000000000","submitted":"2021-02-24 22:21:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777354-fix-trivial-error-in-error-message","meta_rev_id":"8982df351fe58f308678baf2498e9a7823926989","_number":777354,"virtual_id_number":777354,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 22:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-02-24 11:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-24 11:11:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-24 11:11:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 11:11:09.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-02-24 11:32:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-24 13:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0559b42d31fd936b6b16f7457777952ed98bf69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-24 10:43:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e83c5d7628066ad1355ad6c812d64044d29af5f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-24 11:11:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efb4e31777fba7123860f524d31792a0c4868e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-24 11:32:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd32474aba9611dda140a6a21a97aad6711c04d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03d385e782b84322a490711c841dc8d4 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1885b7d218448a9cb01ab608100274 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac4e8167c63450093de4bb5c48cf4cb : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b52cffbfe174f36ba6b116e4ea8e181 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5b80ca0c2bb490aae3f49282134eab8 : SUCCESS in 4m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0924a6c723f84c93a7957e755d28b01d : SUCCESS in 55m 05s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0bd5097f91144f37ac3037d1e2fafdd0 : SUCCESS in 4m 46s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5e9e0f8f59f8402aa73c2d555454d915 : SUCCESS in 4m 24s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/f0778f23fff443fa96b43e1a65255483 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5d6ef495724741ac04a89c18293ecd : SUCCESS in 9m 38s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/27a883302684433e87812bd23dcbf471 : SUCCESS in 1h 09m 31s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/d6ead6bfed6540fe8d51b088cecefc1d : SUCCESS in 20m 01s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/a940c8e9f6254f3fae6c95ee744402cc : SUCCESS in 40m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a9168edfb3149949f929787541bdbdd : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/843391b866da46149ca634850f95ffd2 : SUCCESS in 1h 52m 17s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/44b1492207244a5bb19e2bd543e01eed : SUCCESS in 1h 02m 32s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a3329eef272a4275a75e351ba004cb72 : SUCCESS in 52m 52s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8497dfd98d4a42f29c8d6eddea4a9722 : SUCCESS in 36m 46s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5be24b6fa4e2484d90d1e46e600b78a3 : SUCCESS in 39m 55s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d7d9d08789c54906b38b801a3ddd9f07 : SUCCESS in 50m 43s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/966b5a7b42f2474bae8bc0958ad997e3 : SUCCESS in 55m 14s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e40394d092485d81189ae2bf2c6faf : SUCCESS in 25m 32s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d9ae2a61b4544d8898b942abb0f4aa84 : SUCCESS in 50m 28s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/23cc70756de84a69b613f7b0c4bbff78 : SUCCESS in 38m 38s (non-voting)\n- openstacksdk-ansible-stable-2.8-functional-devstack https://zuul.opendev.org/t/openstack/build/f2f6a0ddd50e448cac476f603539aeda : FAILURE in 39m 32s (non-voting)\n- openstacksdk-ansible-stable-2.9-functional-devstack https://zuul.opendev.org/t/openstack/build/adf09b59689d4d7c8c0131545db8180b : FAILURE in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cbd57244014c1c7a49e3445ff48c7fd846ea985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:49:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9df1eb63bff4811d8f6bd8434a3714abb4e35e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:21:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49f970dd47084f5ea6d64e1ee60a4b0d : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f3c6d7cbcc40af836d09cbfef3fa72 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d56566832c3d4abb9eeaaae27cfa462c : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dba2e045a6a42aea7de3348e3c11765 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/690003ade7d446c78c7e23794dd27a9a : SUCCESS in 41m 24s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/211db1f0fbed463f9f256b37ac8e59bd : SUCCESS in 5m 43s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d2cb8fe16314e188228143d6b691332 : SUCCESS in 5m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a4e3a30f575e42e9b46f92542afb8896 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e2b16090974810a70b65666c13d8ec : SUCCESS in 7m 42s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9298b69a454a4d95a075d2feadcae627 : SUCCESS in 59m 24s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/dbc64fbb9bca4322949312ef177bb3df : SUCCESS in 16m 57s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/2789c1a3fc66497e989284890696ea40 : SUCCESS in 32m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da82571e49274e82b52d927e62b0dfc8 : SUCCESS in 51m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/612f3de720554903971e35635c9dbab7 : SUCCESS in 1h 15m 32s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/3e16f8cfc3fd4fba9fc718d2ba54d974 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"38e2ef4cf7e1df955e1a85db8076de67d62acafd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:21:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8982df351fe58f308678baf2498e9a7823926989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315042d7cf2f4e72b638f20099fc65ec : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c81c1f61a8d504d83478d64f959bd6171655a6a8","revisions":{"c81c1f61a8d504d83478d64f959bd6171655a6a8":{"kind":"REWORK","_number":1,"created":"2021-02-24 10:43:56.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/777354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/777354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/777354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/777354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/777354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/777354/1"}}},"commit":{"parents":[{"commit":"eecefcf6e90d6ce92c032467f9e969a8f08bfac7","subject":"Merge \"Add shared file systems support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/eecefcf6e90d6ce92c032467f9e969a8f08bfac7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-24 10:42:13.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-24 10:43:54.000000000","tz":60},"subject":"Fix a trivial error in one of the error messages","message":"Fix a trivial error in one of the error messages\n\nSome text got inserted in the error message making it sound odd.\n\nSafe to backport.\n\nTrivialFix\n\nChange-Id: I14f0f1ae5421fd0c53980c79d35f5ea2012ea320\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c81c1f61a8d504d83478d64f959bd6171655a6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c81c1f61a8d504d83478d64f959bd6171655a6a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
