)]}'
{"id":"openstack%2Fopenstacksdk~782704","triplet_id":"openstack%2Fopenstacksdk~master~If41a6b2d43d00e5bd78f6b2bd5bf33fa4a5d38d5","project":"openstack/openstacksdk","branch":"master","topic":"trivial","hashtags":[],"change_id":"If41a6b2d43d00e5bd78f6b2bd5bf33fa4a5d38d5","subject":"docs: Add intro doc to user guide","status":"MERGED","created":"2021-03-24 12:30:31.000000000","updated":"2021-04-04 15:53:57.000000000","submitted":"2021-04-04 15:52:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782704-trivial","meta_rev_id":"48bbf863331ce4960c72136f2ed50953eea7b823","_number":782704,"virtual_id_number":782704,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-04 15:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2021-04-01 15:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2021-04-01 15:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 13:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 15:40:12.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"adf6cf4734f801853b4ec3c6a64d4ce58fc0a558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-24 12:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e3ca10407ce952a42d5450ce7b0e845be24be89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 13:42:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d6abf6aeaeb41a998dd72a42b937152 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d58ca30dfc147afb6b5b136b35260dd : SUCCESS in 8m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9fdcec921435478e84e1a6c6a5210f78 : SUCCESS in 1h 08m 04s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0e2e697d871a4634a23cd4e9e866bd3f : SUCCESS in 20m 15s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/0dd1238de8a9419284842dcc34f051da : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dd71bdc43ba111c518167acc151996ce81f6c291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-24 13:46:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2d86f077afda2a7e059325c60b7e73189dd5d8c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 15:29:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1330b36e8cd7413680477532efce618d : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49aa07d5c6c6443aa06f5a57240c4c86 : SUCCESS in 10m 03s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d8acfc07b4594ea29eece80386616fde : SUCCESS in 1h 25m 46s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/14aa1a0da63c4358b25f67ab30d1c02c : SUCCESS in 20m 50s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/212c7bee80f64e58a80b482d6abf2e75 : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":2},{"id":"599bc5bba15d897ba54445452015d1384ca626c8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-04-01 15:40:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd874cd36d0f03e51aea0844ce70af3d14b50e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 14:41:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1a7d8f6af1172693fdc3633b6032f250839c0d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 15:52:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5285c4f8907f4342a543c26883a740ed : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cd124a730248f6a40990ad6cf85912 : SUCCESS in 8m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/66f21a2d61eb448887276751e2bada8f : SUCCESS in 1h 05m 58s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/8cae158832cd43109d8fdde5e2f149ad : SUCCESS in 18m 03s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/1996072d97dd4b18ab3d263683796297 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0d283794b9526b1d7eb90a458708fa0bf5742b4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 15:52:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48bbf863331ce4960c72136f2ed50953eea7b823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 15:53:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ccc4c3ecc0464c8ed52577c2a2d06f : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f280f7cdd67dece901f01bfe1a4cc888850f6e44","revisions":{"84e2485cf0085f9e35b05551240d3809cfc0dc91":{"kind":"REWORK","_number":1,"created":"2021-03-24 12:30:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/782704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/04/782704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/1"}}},"commit":{"parents":[{"commit":"7582a8140efed5b55f19357f1ae8ecc2a8d7d654","subject":"Improve README to provide example of Resource usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7582a8140efed5b55f19357f1ae8ecc2a8d7d654"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-24 12:15:33.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-24 12:28:51.000000000","tz":0},"subject":"docs: Add intro doc to user guide","message":"docs: Add intro doc to user guide\n\nThis is essentially a repeat of content from the recently updated\nREADME. We simply duplicate it rather than using complicated includes\nbecause $effort.\n\nChange-Id: If41a6b2d43d00e5bd78f6b2bd5bf33fa4a5d38d5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/84e2485cf0085f9e35b05551240d3809cfc0dc91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/84e2485cf0085f9e35b05551240d3809cfc0dc91"}]},"branch":"refs/heads/master"},"f280f7cdd67dece901f01bfe1a4cc888850f6e44":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-24 13:46:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/782704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/04/782704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/04/782704/2"}}},"commit":{"parents":[{"commit":"85a2f101598c9c68c8f3f824dbc38b88ab6cdde0","subject":"Improve README to provide example of Resource usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/85a2f101598c9c68c8f3f824dbc38b88ab6cdde0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-24 12:15:33.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-24 13:46:10.000000000","tz":0},"subject":"docs: Add intro doc to user guide","message":"docs: Add intro doc to user guide\n\nThis is essentially a repeat of content from the recently updated\nREADME. We simply duplicate it rather than using complicated includes\nbecause $effort.\n\nChange-Id: If41a6b2d43d00e5bd78f6b2bd5bf33fa4a5d38d5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f280f7cdd67dece901f01bfe1a4cc888850f6e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f280f7cdd67dece901f01bfe1a4cc888850f6e44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
