)]}'
{"id":"openstack%2Fopenstacksdk~789283","triplet_id":"openstack%2Fopenstacksdk~master~I4a49116db8e67329fb1f099b5a3806b07d47e5d4","project":"openstack/openstacksdk","branch":"master","topic":"ironic-ci","hashtags":[],"change_id":"I4a49116db8e67329fb1f099b5a3806b07d47e5d4","subject":"Increase RAM for the Ironic CI jobs","status":"MERGED","created":"2021-05-03 13:33:36.000000000","updated":"2021-05-03 17:31:24.000000000","submitted":"2021-05-03 17:28:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789283-ironic-ci","meta_rev_id":"2702de768c36b63aa267b28979eb21f3c5821a3f","_number":789283,"virtual_id_number":789283,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 17:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2021-05-03 14:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2021-05-03 14:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 14:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 14:52:50.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"758db40fa8bce30c68b708f0ebb01f78a5783bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-03 13:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d904074224b5c65e7f34be2fbfa70f941ce077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 14:47:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c98aba0acd4ecc9835b9183cbb44b8 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4cd1ea492824eb893b35bf99223301e : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa216a8206c940dcac36f2b58497438b : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06fd15633b0e4d9cb69e33ae716a5d28 : SUCCESS in 6m 31s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/37064e66a64941958b16105c457ededb : SUCCESS in 47m 03s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/242f414e4e6040b5a071748bfa5ecda1 : SUCCESS in 4m 53s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/56ec2396ba2f4ca2bfdc229d943d19ec : SUCCESS in 6m 31s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/19528f9a83664c4cb771d74a5671545a : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150de4e82065407cb366d90a13ed7002 : SUCCESS in 7m 17s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bdcd9ec5f02842aea4cdcf2aacbbef6d : SUCCESS in 55m 27s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/375729ef6a3f4834a858b9757e3c45de : SUCCESS in 19m 15s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b71e46948de947619fb0f5f216bc9adf : SUCCESS in 31m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/463122ceeb7a476a863508848a164a88 : SUCCESS in 48m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/380aea20aae64f36bc3aa66928fa8f1c : SUCCESS in 1h 11m 38s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/9491ed9e88394bc187df8d9f74bc23f4 : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d6cee3c90ced4c348565e643b1f7aa90 : SUCCESS in 47m 35s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6fd0a55e844648f48a31114b65d82c82 : SUCCESS in 29m 45s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1e094a909fb9490b9dd39f724b090c17 : SUCCESS in 35m 18s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6cf3e7748dbe4eda9ea34208110669fe : FAILURE in 49m 33s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/e650623dbd63455b8168b7480a847100 : SUCCESS in 51m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5efb15cf0de3419e9a0e8864750e9a60 : SUCCESS in 23m 27s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3690c2c7ab2e4fd8821d3a8d06c1e181 : SUCCESS in 54m 22s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8dcdbfaae2a74e1f8315f8bc0db507da : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89cbc2275439e9d1625fc3f34ac1ca9498d2aa44","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-05-03 14:52:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f1b5ae3199e95e92dba8bb87f354418362900d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 14:53:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbd3f561e08bdfcb40b559e4012df9eb17973d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:28:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8be96dd3e284e6fa0a3564acee2509e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fd614fdafd145efb9dbcdb63bdc9419 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca2e97eabc343798a7dfc598cf1be4a : SUCCESS in 4m 12s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b65af0e0a5a049458a593a63de4aa0e9 : SUCCESS in 1h 09m 02s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/98daa590c5bd44119b528122131a85c3 : SUCCESS in 4m 44s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b2bf0351b624210ae63970cd3d0b810 : SUCCESS in 3m 57s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/8bb0dda0fa034c61919066414a0c8665 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400eede429c04d0b8e9c84ce897f8601 : SUCCESS in 11m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/dfe214a60aef4e4f94fac23b588ef35c : SUCCESS in 1h 14m 57s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/85560c78e3e0447b8019508b5885208d : SUCCESS in 25m 15s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/e4419d592afe44c8b177090f3852b4e3 : SUCCESS in 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce4a8c31b99f46368876d543bc573175 : SUCCESS in 45m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0ec642d2d1d640d3bc08b5c5beeb463c : SUCCESS in 1h 35m 00s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/16d4c9d1498d466a8661854b64f60eed : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"cbb0b897b30bd80999617a87c46403fb1903f68e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:28:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2702de768c36b63aa267b28979eb21f3c5821a3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:31:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d687c68e5341f69f12fac6d37a67fd : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a237bcc1d4118e5a474020e73f4f237882d0dd3","revisions":{"6a237bcc1d4118e5a474020e73f4f237882d0dd3":{"kind":"REWORK","_number":1,"created":"2021-05-03 13:33:36.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/789283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/83/789283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/83/789283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/83/789283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/83/789283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/83/789283/1"}}},"commit":{"parents":[{"commit":"dc47f23697c5e28735fe440d87555156844325f1","subject":"Merge \"Add min_count and max_count attributes to Server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/dc47f23697c5e28735fe440d87555156844325f1"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-03 13:33:04.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-03 13:33:04.000000000","tz":120},"subject":"Increase RAM for the Ironic CI jobs","message":"Increase RAM for the Ironic CI jobs\n\nAnd decrease the number of testing VMs since we don\u0027t need so many.\n\nChange-Id: I4a49116db8e67329fb1f099b5a3806b07d47e5d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a237bcc1d4118e5a474020e73f4f237882d0dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a237bcc1d4118e5a474020e73f4f237882d0dd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
