)]}'
{"id":"openstack%2Fopenstacksdk~822254","triplet_id":"openstack%2Fopenstacksdk~master~If6dfc295e062da660d408af50a06cac69bf03da9","project":"openstack/openstacksdk","branch":"master","topic":"fix_misuse_of_assertTrue","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2021-12-20 14:41:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6dfc295e062da660d408af50a06cac69bf03da9","subject":"Fix misuse of assertTrue","status":"MERGED","created":"2021-12-19 06:01:15.000000000","updated":"2021-12-20 14:43:43.000000000","submitted":"2021-12-20 14:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822254-fix_misuse_of_assertTrue","meta_rev_id":"1220713fcf025d868ff1955a1869590d7c4cd212","_number":822254,"virtual_id_number":822254,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 14:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-12-20 10:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-12-20 10:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-19 07:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 10:37:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f25a71e4bbbe4f71980a74eac6de101553c8a4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-12-19 06:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa40b60cbf657049bb6deb5334ea5b0a9f0c493d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 07:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf0e3eba401437da6e9188ca8ae64a1 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10d9df86201c409c813aca0eec4948c3 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6adabd62ead4c43a34f7ddb21748aa7 : SUCCESS in 4m 22s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/84e147baf11741b3be0dc2ba12234133 : SUCCESS in 45m 59s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b1b4200a9d2a4999940bd7d6788162f2 : SUCCESS in 3m 59s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e5d21a0e71894e43a9273741b2943642 : SUCCESS in 4m 14s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c6193f48334048009bc7c011903be146 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65314910b2845a1ba00bbbe8bc77958 : SUCCESS in 6m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/af209f2298004a0ab21361135bf2b32b : SUCCESS in 23m 26s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ed772a04d1674f4f819e2bb1cdda1d26 : SUCCESS in 16m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71481744f1a8487585aff4d04950d80a : SUCCESS in 46m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1051bd9ac83c4a51a57293c7e728addc : SUCCESS in 1h 15m 18s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/7d865d01a75f4020ac6f51ab98ac60e4 : SUCCESS in 39m 33s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0241d8f7cabc4706809bebb635af2ca9 : SUCCESS in 52m 31s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/40cfb290dfa74d14bbf352c056dcd523 : SUCCESS in 32m 52s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0fba11f0add1405ebcd60fd1f3ebf0c4 : SUCCESS in 38m 25s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9a5eee551613435ca99b4424bc619f6f : SUCCESS in 57m 00s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9f328a7a0716441a9c89d61059e63493 : SUCCESS in 56m 00s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c804996ca4f45ad93b29d9467de5420 : SUCCESS in 28m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/751c2bf8f40f48338e8a28e4376e4fa4 : SUCCESS in 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c38a1ad21c1f64bb51a8e07e891c8ecd6fa048e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-20 10:37:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cef5e871ea95e7f30d5ebeed4d15bf9959a1dd15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:38:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fee0f89958a1fa504d44965616152b859594c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:41:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18460bcd9ccf4060a42eef72cfbe8614 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d7c7a2c2454c50ada122ae6b66c4df : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ff1e5ce0d0e446ab9b912f0956b7766 : SUCCESS in 6m 01s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9def9702bca34422a0c14a05c8b2acc6 : SUCCESS in 50m 59s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ab4c7c59248d46688567ccbe2560172a : SUCCESS in 6m 06s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b63c68783df84887b370f4520aec2e5d : SUCCESS in 7m 02s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/9e983d02a147434fa82b2c8a49f04352 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e635e21409e84071b69b040ad61941ab : SUCCESS in 9m 11s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/046efcec26d44d5a93d0963232f5f438 : SUCCESS in 25m 59s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/e4e0e8dd1c76445ebd0bf60d4bb7a978 : SUCCESS in 19m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecc4db430f804af593e9dadd0a2ff66d : SUCCESS in 47m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/48c682cb37f44035a3c524c3807ca4af : SUCCESS in 1h 20m 40s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/7e503b51f5cd44aba54b6eacdf74ef69 : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":1},{"id":"947725a85714276aedd0b69a0c9ccfbc58e95d02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:41:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1220713fcf025d868ff1955a1869590d7c4cd212","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:43:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ea7026a42546918949703b1f845869 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7098876455a3abbe53609f58d2db380460adc6d","revisions":{"d7098876455a3abbe53609f58d2db380460adc6d":{"kind":"REWORK","_number":1,"created":"2021-12-19 06:01:15.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/54/822254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/822254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/822254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/822254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/822254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/822254/1"}}},"commit":{"parents":[{"commit":"26c9bc22e372c46c8737e26823d0e99611c6b1b8","subject":"Merge \"Adds \"test_update_zone\" test case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/26c9bc22e372c46c8737e26823d0e99611c6b1b8"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-12-19 05:56:55.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-12-19 06:00:42.000000000","tz":540},"subject":"Fix misuse of assertTrue","message":"Fix misuse of assertTrue\n\nFix misuse of assertTrue by replacing with assertEqual.\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: If6dfc295e062da660d408af50a06cac69bf03da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d7098876455a3abbe53609f58d2db380460adc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d7098876455a3abbe53609f58d2db380460adc6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
