)]}'
{"id":"openstack%2Fopenstacksdk~823961","triplet_id":"openstack%2Fopenstacksdk~master~Iceb4377082d678b825774a69d211a158c64211b2","project":"openstack/openstacksdk","branch":"master","topic":"merge-branch","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2022-01-24 12:47:37.000000000","reason":"Artem Goncharov replied on the change"}},"hashtags":[],"change_id":"Iceb4377082d678b825774a69d211a158c64211b2","subject":"Merge feature/r1 branch into master","status":"MERGED","created":"2022-01-10 11:00:11.000000000","updated":"2022-01-24 14:13:57.000000000","submitted":"2022-01-24 14:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823961-merge-branch","meta_rev_id":"bea1329195e210cc30bb0181e6ea58e92210e4e2","_number":823961,"virtual_id_number":823961,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 14:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2022-01-21 17:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2022-01-24 12:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":1,"date":"2022-01-24 12:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 12:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-17 11:23:53.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"CC"},{"updated":"2022-01-21 17:51:53.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2022-01-24 12:35:53.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"REVIEWER"}],"messages":[{"id":"aa1e62c16c5e911c44b19293bf57d2b1cabd4b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-10 11:00:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05b4721ec7af17080f69a4217ad50f98b626bcfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:30:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdeb6499f7ed4d38819a05d249e1173e : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0367691c699546449e53e5380698a0e7 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bd835e103754f3b9cdc61f73df4a875 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8c596605ce8247b08232ae0318a1a3d0 : SUCCESS in 54m 10s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6db18dc40ecf4e0bb0262503f0301d28 : SUCCESS in 4m 46s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/35db058c20f94587b53ef891ba23e660 : SUCCESS in 5m 03s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a7f658162ee64f7cba50edb724e0c83c : FAILURE in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769c6e8031c7450199a1dffcecc33e1d : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7003f8b6d48646208c0a12bd48eb24b3 : SUCCESS in 4m 19s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/361b32cb4dbc4fd3982460668b4afac8 : SUCCESS in 28m 24s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/cbfa663be682461fbd21e7b4708006c8 : SUCCESS in 19m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5145b096577d4071b14ab9267a778545 : SUCCESS in 49m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2b1c9e1cee6d4fefa2cf49270da223cd : SUCCESS in 1h 21m 51s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/430e9d8d6902430b818e82baedfd30b5 : SUCCESS in 46m 38s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e1f8b0bc42a14d009417643de0864baa : SUCCESS in 48m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7448da97cd8246909127b5f448e8e413 : POST_FAILURE in 26m 13s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/83aa2c946b854c4ba8f3fd043a35903c : SUCCESS in 33m 11s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/4cf7f3f5197947ed90359fa64e5c2c30 : SUCCESS in 39m 38s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/70e18c89235947358b443c7290936298 : FAILURE in 58m 37s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b3da0521d67d4a0a883dd08edf155c15 : SUCCESS in 53m 57s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef77214cd9d4638bf5478476f84bba7 : SUCCESS in 29m 07s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8b6313db520b4aa8b926fee3a1e6abfd : FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1c4a1611fd0cbbd68857c8da8ca301dfa0de39c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-11 17:46:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"588dc5c73dc466f55533c3e38107621a4505dab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 19:09:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cb60cb5fae43b8b9b7da2f571c498b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1824e40ccf40688b5d8f3996e3e79b : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/890b7212f9db4c1c951ccc862b2c59ae : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4263cedf3f824741ba969ecd2d702489 : SUCCESS in 46m 08s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c42488e83cd0433b81addd3a7878cbae : SUCCESS in 4m 42s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/739eb77cfb194721bf0f2eaa3b365941 : SUCCESS in 5m 59s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/2f65eea3794b41aa9350a69d8ddb9379 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7443778b0a04c55a221caf8e6a89264 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f33875678f544b24a923b8588b9b3c97 : SUCCESS in 4m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/24cec2c902bd4113a50574cbd043100d : SUCCESS in 22m 43s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/815ef56a8006442b9a71064fc44164c4 : SUCCESS in 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5374d73036c45bf9b8f217c13b61038 : SUCCESS in 47m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dc0aa8094df34245826c3f19617564c1 : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/8f6fb7735ac34e86939807da9a1269ec : SUCCESS in 49m 37s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/67bb80872ffb4bc597ea58ed14bf9cbc : SUCCESS in 51m 06s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b679eaa48adb40dd87f700d9efaf5b73 : POST_FAILURE in 21m 41s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/fdf3b0e0cde6412bb68345bff58481bb : SUCCESS in 32m 24s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c1d712532386484494b5db81ab82ca37 : SUCCESS in 37m 39s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/90431972bcca4014b287c8bdecb1a738 : FAILURE in 53m 38s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7a54e5ebc88c45a1a6c39e6835f45987 : SUCCESS in 56m 25s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe015c1f3b5440ce97193a83b652ebdf : SUCCESS in 29m 13s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7c7ec193a013469aa2c5be9be69e2fc8 : FAILURE in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76dd245ff11b753ff18e3ee16e35c93e4bcd57be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-14 17:56:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e24def71883b36e84b7667d7af57563b322e9068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 17:56:48.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"462edc7dc55f73c7ad64a3d93b0a7cd22870c7a8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-16 07:21:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02a71d4ad968b34a4feef235985d4f046daa30cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-16 08:51:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a46ec3233440c49f609d3d3e75b1e0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/466329af178745fe85dac000958298ce : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4ca33611e14e108b0d4e8b7edfc873 : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fa3f7733c4f243709853846b129363a5 : SUCCESS in 39m 50s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2cdb6b33adcc4cdcb88a2a9d2f35d253 : SUCCESS in 5m 07s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0791accee84a48fe8de157035f826015 : SUCCESS in 4m 33s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e6cc08bd0e604878ba634334b7373c1b : FAILURE in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae47d1597594ba79e33cd6ef48cfcb8 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ca95ca0ab7486f922b27b6c41d21bf : SUCCESS in 3m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/390417bef7954365a06b4fe9cd3ae60c : SUCCESS in 28m 34s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/1b987ba565924d7f97e024690d6d868d : SUCCESS in 20m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93119fded3e34ffea599453b16430b03 : SUCCESS in 1h 07m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/34be9be9cb4a4832a70f3c55fcb151dc : SUCCESS in 1h 22m 17s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/3b0ee698e6324f5f98fb5b1be226ab22 : SUCCESS in 55m 13s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/855639508e3d4a3bb4b1a5168f4e3b81 : SUCCESS in 48m 25s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/faa07acd47e647eab63373b3d321691d : POST_FAILURE in 25m 01s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c750285654aa49418808b57fcef464ca : SUCCESS in 32m 22s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/495d259f5fd84c029ebbdaf1058322e2 : FAILURE in 42m 00s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6707e2beccab4400b20872bf8dfbe699 : SUCCESS in 59m 39s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c70986c4fd90410f8556b190ce0d391e : SUCCESS in 55m 19s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61ad736bffbf4351a17d38d71e348d7d : SUCCESS in 29m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/9d2720bd399e4490a5a6b19ca5d58268 : FAILURE in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c730a062de48a9390f4f01219ac206ff251c0c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-17 10:03:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c7a55bcd15db601b02758fd5ddfaff8590c7f71c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 10:34:27.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ab3699e1d92b3821052dd46b66f4c08e3d330562","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2022-01-17 11:23:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d72c50fa43d3dcb52922758c5ac02b730b2e96fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 12:52:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635ffef2efaa461aa963d12da8a85fba : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a575cfbd6184611829d31f725a3507a : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e137ae6f3a794a6899a623dda7159385 : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a9d08bc8400d4d788a0ad8bdb840a711 : SUCCESS in 53m 55s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/11fbc5e1b6b44cc494cb50b62fe0e745 : SUCCESS in 5m 16s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8c819c58f08847f59ce2eb7c96fcbad8 : SUCCESS in 4m 51s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/60a723d05798427cb0d4a3baa6f73601 : FAILURE in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47065a0b1c134de49af9b6408a66d55d : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/322581bcb1ef49c2b109e36187fddfc2 : SUCCESS in 4m 12s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d2f5a33d12304073b0484b43ebe9c9c9 : SUCCESS in 25m 17s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/fd8a047884bd4248b3dcc027ede9904b : SUCCESS in 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7458b2ade447497cb625eb8148787961 : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0244692462c46db858e6bdec31f0873 : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/09aced49b2704c8cbcd6866185d90149 : SUCCESS in 49m 11s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d9fde023122740568e359c51aedbeef6 : SUCCESS in 50m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f63a648e34c347b8bb7456691375ee5d : POST_FAILURE in 24m 37s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a15f4638e49b40ff90593266650bfb97 : SUCCESS in 36m 00s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/232af162fb984922bda494e448152460 : SUCCESS in 40m 50s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/613707ebc1ca449b8f4f28168c2c4b9c : SUCCESS in 56m 46s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7a18eb458c224ec7b951bcc3eeffd0f5 : SUCCESS in 1h 00m 04s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622764c204494bb5b7b4681f99991b14 : SUCCESS in 28m 58s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f0e3ac54bbb34589be48542a0359389e : FAILURE in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbe47253a4250aa1caea81f9041188cdbea1550f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-17 15:34:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4aff4458ecec5a8f783b0b872926c1ee28fc4075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 16:48:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64d03c1d75f4c7a88346fd9a2ea0184 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f45cbb184a142379fe8e6119471d747 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e1058d7c53a454faed6f7f3f26936bf : SUCCESS in 4m 19s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8365f41c700d46ec9e3abf9f1e90858d : SUCCESS in 50m 02s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d4ed6eb2028f4618a0aef4fa96aa1ddd : SUCCESS in 4m 07s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6c7df2bff7504bfaacaa630a51885ae0 : SUCCESS in 5m 46s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/b8a23914f1ac487bbce789abc73e0910 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc41869ae67e4b21a72c5bf2282448b9 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5843f94587845ed81eb3b658c5dd8e7 : SUCCESS in 5m 09s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1ddc492fc6544440a5db6236b33fdd4a : SUCCESS in 27m 09s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/de8258543c9d4f4abc22609319f502a5 : SUCCESS in 18m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f540472f64b46fab96046767e07983d : SUCCESS in 45m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/81960634bdfa45afae0a0bd238a143f3 : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/be8d5d253db44d03987607a4da07403c : SUCCESS in 35m 44s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b00ca36bf2214c81a628bb9d2abeb702 : SUCCESS in 45m 52s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5fa8ce2304be4de28cad7821c74a2c9d : POST_FAILURE in 21m 53s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ac140ad5939640daad69132a73e995e9 : SUCCESS in 24m 13s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d02f8a5c60614b21b85e2acd90b09a8c : SUCCESS in 34m 14s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/afa9135a1abc45f6bacd965d68f61765 : FAILURE in 58m 02s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c19092552f8d471fb737693d6de2bf8e : SUCCESS in 46m 03s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba515912958c44b8b1b797c03e5f5e45 : SUCCESS in 31m 42s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3a503e7074a64f728bf2f79a86a6a5b4 : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed84a32c17671a42ba926a57b1d8a256efff21a4","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2022-01-21 17:51:53.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e298329fdd79404f38e4db9da1a4856b1d8ca12","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2022-01-24 12:35:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4a552e5d2e7ba1e7121323bf642d2703154d4f19","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-01-24 12:47:37.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c4216c5ce6a00ac848c41850d31f0f70b4bc6ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:48:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"328460176ff7a0f55225eb8270e65236f0c75ac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:09:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7faeeeb8c7e54d5bbb84d384237f22df : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37e57e12a18c4b93b6ecbaba0f97cc44 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c886f19fbf4742e29788134f001fa170 : SUCCESS in 4m 19s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/31bd95f3b79e418da3956024117a3213 : SUCCESS in 34m 35s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/372b7d7693d34f118fd71994b6d1c420 : SUCCESS in 4m 20s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e985792e4f014342a5d4a046511ed789 : SUCCESS in 4m 04s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/df810738ff34471a8ca55fdf4e9bb703 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3da20b6c274cea8784459e581d918d : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179956dd90904e06968752791215d519 : SUCCESS in 4m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cdeb9d860cbc42bda7afdbd24fbeb263 : SUCCESS in 20m 19s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/d16742d2d52840e0bc717f4832df1163 : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8188c4e25314a1cb14d026345c1adda : SUCCESS in 42m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d68ee4e1441483297c9fc8902caa536 : SUCCESS in 1h 20m 14s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/2305405b29a8441497c7c466f4073e8d : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0eace5b474c732f305f61fe14d4f75275723cbbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:09:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bea1329195e210cc30bb0181e6ea58e92210e4e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:13:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63687d013f9a422b924a9e8121598ec7 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83e7ea32667641fd9419083c31b42271 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6c0209bf5a653c092bad07212d4213a70168ad7f","revisions":{"2b7b88d6a219bd100431ee2c9bba0c153f6ad81c":{"kind":"REWORK","_number":1,"created":"2022-01-10 11:00:11.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/61/823961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/61/823961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/1"}}},"commit":{"parents":[{"commit":"58384268487fa854f21c470b101641ab382c9897","subject":"Merge \"tests: Handle overridden \u0027verify_delete\u0027 (kw)args\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58384268487fa854f21c470b101641ab382c9897"}]},{"commit":"a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7","subject":"Merge \"Get rid of normalization in compute CL\" into feature/r1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-01-10 10:59:09.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-01-10 11:00:02.000000000","tz":60},"subject":"Merge remote-tracking branch \u0027origin/feature/r1\u0027 into merge-branch","message":"Merge remote-tracking branch \u0027origin/feature/r1\u0027 into merge-branch\n\nChange-Id: Iceb4377082d678b825774a69d211a158c64211b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2b7b88d6a219bd100431ee2c9bba0c153f6ad81c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2b7b88d6a219bd100431ee2c9bba0c153f6ad81c"}]},"branch":"refs/heads/master"},"43da826f02a54b94a6f6a8e9c923e899978bd5ce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-14 17:56:04.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/61/823961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/61/823961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/2"}}},"commit":{"parents":[{"commit":"58384268487fa854f21c470b101641ab382c9897","subject":"Merge \"tests: Handle overridden \u0027verify_delete\u0027 (kw)args\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58384268487fa854f21c470b101641ab382c9897"}]},{"commit":"a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7","subject":"Merge \"Get rid of normalization in compute CL\" into feature/r1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-01-10 10:59:09.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-01-14 17:56:04.000000000","tz":0},"subject":"Merge feature/r1 branch into master","message":"Merge feature/r1 branch into master\n\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/824746\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/824733\n\nChange-Id: Iceb4377082d678b825774a69d211a158c64211b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/43da826f02a54b94a6f6a8e9c923e899978bd5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/43da826f02a54b94a6f6a8e9c923e899978bd5ce"}]},"branch":"refs/heads/master","description":"Edit commit message"},"380656263bcc07e15afc2cca64d5408723ca48a6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-17 10:03:19.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/61/823961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/61/823961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/3"}}},"commit":{"parents":[{"commit":"58384268487fa854f21c470b101641ab382c9897","subject":"Merge \"tests: Handle overridden \u0027verify_delete\u0027 (kw)args\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58384268487fa854f21c470b101641ab382c9897"}]},{"commit":"a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7","subject":"Merge \"Get rid of normalization in compute CL\" into feature/r1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-01-10 10:59:09.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-01-17 10:03:19.000000000","tz":0},"subject":"Merge feature/r1 branch into master","message":"Merge feature/r1 branch into master\n\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/824746\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/824733\nDepends-On: https://review.opendev.org/c/openstack/osc-lib/+/824909\n\nChange-Id: Iceb4377082d678b825774a69d211a158c64211b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/380656263bcc07e15afc2cca64d5408723ca48a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/380656263bcc07e15afc2cca64d5408723ca48a6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6c0209bf5a653c092bad07212d4213a70168ad7f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-01-17 15:34:56.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/61/823961/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/61/823961/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/61/823961/4"}}},"commit":{"parents":[{"commit":"58384268487fa854f21c470b101641ab382c9897","subject":"Merge \"tests: Handle overridden \u0027verify_delete\u0027 (kw)args\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58384268487fa854f21c470b101641ab382c9897"}]},{"commit":"a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7","subject":"Merge \"Get rid of normalization in compute CL\" into feature/r1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a3edc2eff3bb4e2262e4978c2ecef5826b8af9f7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-01-10 10:59:09.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-01-17 15:34:56.000000000","tz":0},"subject":"Merge feature/r1 branch into master","message":"Merge feature/r1 branch into master\n\nDepends-On: https://review.opendev.org/c/openstack/osc-lib/+/824909\n\nChange-Id: Iceb4377082d678b825774a69d211a158c64211b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6c0209bf5a653c092bad07212d4213a70168ad7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6c0209bf5a653c092bad07212d4213a70168ad7f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
