)]}'
{"id":"openstack%2Fopenstacksdk~835466","triplet_id":"openstack%2Fopenstacksdk~stable%2Fyoga~Ia46013b416bf9d7103659190a5c18706a458c26c","project":"openstack/openstacksdk","branch":"stable/yoga","topic":"pedro/fix-group-assignment-on-domain","attention_set":{},"removed_from_attention_set":{"16363":{"account":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"last_update":"2023-05-16 09:02:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia46013b416bf9d7103659190a5c18706a458c26c","subject":"Fix adding roles to groups on different domains","status":"ABANDONED","created":"2022-03-28 14:38:48.000000000","updated":"2023-05-16 09:02:50.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f83b96dd1a2961f7db69376c4d0aeefa3dbbd1e","_number":835466,"virtual_id_number":835466,"owner":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-28 17:14:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 17:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 11:39:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"ec98e53415fcc2bd94d3c525c9f5cdcd57706b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-28 14:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb05b7a03e7d6eb680d34afc5e145806a9f3739e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-28 14:39:31.000000000","message":"Topic pedro/fix-group-assignment-on-domain-yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"272330144256c10c70891e06e84d7a4a27105095","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-28 14:39:50.000000000","message":"Topic set to pedro/fix-group-assignment-on-domain","accounts_in_message":[],"_revision_number":1},{"id":"1739c0ad51778af4ae1cbdfe43b27f8b11a2acad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-28 15:48:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd776c13fc3aea77efa64842ae24ab6a569daa46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 17:14:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fd2735f13e4285986981fc1d22d034 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56e435c8e81d41c0800047d4e273f61f : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf78326791c943f3ae303676b2b908b1 : SUCCESS in 7m 11s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1a9700011d644f7996f1eba2c87f409e : SUCCESS in 59m 12s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8989bdec9e2d4c1eaebd6566af096550 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44fa48ab309b4c429314d60ca03aea98 : SUCCESS in 10m 03s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b0b1cbb75d6b4b3bb1ae23e26fd6751d : SUCCESS in 24m 34s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b4aa21ea513f442f994407d320631f3a : SUCCESS in 17m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9f03ac1aa284a9f90363e5a4c071109 : SUCCESS in 48m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ae2acd2d5c4644638dd6564b49eae1a0 : SUCCESS in 1h 20m 22s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/1bb289b305024a5dbab72d79a5a8e7df : SUCCESS in 59m 47s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/6857c2b7f48240ac845f79a0d8df4778 : SUCCESS in 49m 48s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a35eb04cccbb4322b8c417dad8dfce67 : SUCCESS in 30m 53s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/668666907e3c4444b3fd6f4c99e49d0e : SUCCESS in 43m 19s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/81f5a6261a0a4d40a423e7db04cb9462 : SUCCESS in 1h 13m 56s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a361d2aa14f9409199a81972a571b296 : SUCCESS in 57m 05s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e5dbd0fd3634cf684c10b9cc9b4ae43 : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e68893b5b134b46c47899210f7c1d9b81c3c936a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-30 09:51:04.000000000","message":"Topic pedro/fix-group-assignment-on-domain-yoga removed","accounts_in_message":[],"_revision_number":2},{"id":"79eca344999f3b1eece2d2e602ac7e53af222298","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-03-30 09:51:23.000000000","message":"Topic set to pedro/fix-group-assignment-on-domain","accounts_in_message":[],"_revision_number":2},{"id":"6037aedec0fd9a340c7c8584ea8fb183cec5934f","author":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"date":"2022-04-01 14:34:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8c182c6b322a6f4bd7cc84970077756b7b4188c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-22 11:39:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f83b96dd1a2961f7db69376c4d0aeefa3dbbd1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-16 09:02:50.000000000","message":"Abandoned\n\nOpen questions for a few months. If we can get some resolution on them, we can reopen and merge this. Until then, let\u0027s leave this","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80b0c833bfc7db26176799aca9aa42ed9513ed85","revisions":{"577d1464d33397d0274b71304e8eed3e7954d546":{"kind":"REWORK","_number":1,"created":"2022-03-28 14:38:48.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/66/835466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/66/835466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/1"}}},"commit":{"parents":[{"commit":"fd300de63f5a7a9c7760d69a2d7ce248e058bd61","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fd300de63f5a7a9c7760d69a2d7ce248e058bd61"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","date":"2022-01-31 09:35:40.000000000","tz":0},"committer":{"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","date":"2022-03-28 14:38:42.000000000","tz":60},"subject":"Fix adding roles to groups on different domains","message":"Fix adding roles to groups on different domains\n\nSimilar to commit 3fce613297865a6f9ab6657b4c84ce1e269f58d4 but for\ngroups.\n\nChange-Id: Ia46013b416bf9d7103659190a5c18706a458c26c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/577d1464d33397d0274b71304e8eed3e7954d546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/577d1464d33397d0274b71304e8eed3e7954d546"}]},"branch":"refs/heads/stable/yoga"},"80b0c833bfc7db26176799aca9aa42ed9513ed85":{"kind":"REWORK","_number":2,"created":"2022-03-28 15:48:25.000000000","uploader":{"_account_id":16363,"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","username":"pedroalvarez"},"ref":"refs/changes/66/835466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/66/835466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/66/835466/2"}}},"commit":{"parents":[{"commit":"fd300de63f5a7a9c7760d69a2d7ce248e058bd61","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fd300de63f5a7a9c7760d69a2d7ce248e058bd61"}]}],"author":{"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","date":"2022-01-31 09:35:40.000000000","tz":0},"committer":{"name":"Pedro Alvarez","email":"pedro.alvarez@softiron.com","date":"2022-03-28 15:48:11.000000000","tz":60},"subject":"Fix adding roles to groups on different domains","message":"Fix adding roles to groups on different domains\n\nSimilar to commit 3fce613297865a6f9ab6657b4c84ce1e269f58d4 but for\ngroups.\n\nChange-Id: Ia46013b416bf9d7103659190a5c18706a458c26c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/80b0c833bfc7db26176799aca9aa42ed9513ed85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/80b0c833bfc7db26176799aca9aa42ed9513ed85"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
