)]}'
{"id":"openstack%2Fopenstacksdk~853703","triplet_id":"openstack%2Fopenstacksdk~master~I05101d5154ee5087d50c5cbf347054fdb1f582fe","project":"openstack/openstacksdk","branch":"master","topic":"add_support_for_server_api_fault_field","attention_set":{},"removed_from_attention_set":{"34257":{"account":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"last_update":"2022-09-23 07:12:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05101d5154ee5087d50c5cbf347054fdb1f582fe","subject":"Add support for fault object per Server API","status":"MERGED","created":"2022-08-18 20:36:21.000000000","updated":"2022-09-23 07:13:45.000000000","submitted":"2022-09-23 07:12:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853703-add_support_for_server_api_fault_field","meta_rev_id":"35af28096bce5c02238bea978aa5fc4ce9cf7239","_number":853703,"virtual_id_number":853703,"owner":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 07:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-09-20 10:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-23 05:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2022-09-23 06:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-23 05:49:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 22:17:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 10:05:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-09-23 05:49:32.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-09-23 06:10:35.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"4f56b98abdf1ea2c88786c27b0347d36f1d5d2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-18 20:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1956ed9e04d02223b5bbe17ce266e9bb805109d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-18 20:56:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3579012e947edbe1a9dbd1f5640ca905c61bd971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-18 20:57:41.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f87fad54cd7df180b0e2b974719365dc1622888b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 22:17:54.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65ed30cf5c9d4ef78d303023b9ca7fd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cd81111b7349669937318f58e884d1 : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fee4d6632d56427686e2d2f9cc817d7e : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d23d338827a547848e574d2f9887035f : FAILURE in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f95894356dd342aca1c6b36d35af4c27 : FAILURE in 4m 26s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d67ab2d043dd44c783f16c1683cf4f63 : SUCCESS in 48m 45s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3cef06338ac44507849b27cc3e0a0745 : FAILURE in 4m 42s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/02eafb24f4924b81816abb11ae1ce45b : SUCCESS in 4m 34s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2bf381978967460cacaaf19ebfaa0407 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348ee2e1f82647338f9a071d4afd559a : SUCCESS in 7m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/503b83e8043e4678aed45b66718b0b59 : SUCCESS in 1h 06m 00s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/39c52a4e622548aeae67b32e19412834 : SUCCESS in 11m 52s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/a5fde604a1774218baa8a2a64bb3397c : SUCCESS in 39m 55s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a161f3ceb2e24bc79c10b4f6da421f77 : SUCCESS in 46m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/afb13615cecb4314919ff37f055cfdeb : SUCCESS in 1h 15m 01s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/9876857152df456fb2af76132b014079 : SUCCESS in 41m 51s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d9d06e3ea48e49bea87e5f343bdd1b6e : SUCCESS in 44m 32s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c071c25898d74068bc645a9af11299be : SUCCESS in 24m 20s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/bb3ef507d3ec49c8ac82191dd3612b51 : SUCCESS in 25m 43s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c85a5112d5084eec9bb73129fb9962a3 : SUCCESS in 49m 33s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/2d332c4c401b4b199c0ca99b6af248a1 : FAILURE in 34m 19s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/88f921a560ab4307b3269dfa5fbfb9f7 : SUCCESS in 49m 42s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b2d5dea83433477aa06f64c1a203e95f : SUCCESS in 55m 42s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ff5be51ac2466e923e4dfd7f1285f4 : FAILURE in 12m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/98d64a70a4b64f9486de68ba7e397699 : SUCCESS in 46m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47c050900b748746e9d9d7c7a4539e477b2836b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-19 18:29:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff7ff021b19f6fefc462d150f723dc2400620f6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-19 18:31:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"815ba398053a1c4c97ee3dcc5f4777f15493ecf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:55:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87484a87edcd42c09abb8a2e767b27b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b6fb30fc674501bf6a04f42297e7ee : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f5e15f0777472892ed784cae4aa92f : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8482f8d8604943eca66a19287ca48ffb : FAILURE in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52e67cbfdca1437e9e205e6fd6f7515a : FAILURE in 4m 35s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/433985c5b239459ea00c31a338bff53e : SUCCESS in 50m 50s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/15b24d9ff24a48d9a918687f53aae159 : FAILURE in 4m 55s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7a664bb7ded7435e8ee35cc8be201422 : SUCCESS in 4m 19s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d470bc03c5674f22b6523efdc2c70ada : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502c0743c2bc45e1835fcdc6cd37bd9f : SUCCESS in 7m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cbdad5af1d50483aa056c763a48e4499 : SUCCESS in 53m 09s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f15fd7fd9aa3476f8953524d822bb115 : SUCCESS in 13m 00s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/df442de3678a41408dc00d1ef56b0614 : SUCCESS in 35m 49s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e20262b8160462cbc6b8a300ee7196b : SUCCESS in 42m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89047e38742e429a80b7e1a000a561eb : SUCCESS in 1h 22m 19s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/e6193618b90541848b93486532075f14 : SUCCESS in 39m 34s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/532099b5b7434617bd219ad6ad7ca748 : SUCCESS in 47m 40s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b673f9cc460d431e814f27fd62623e0f : SUCCESS in 30m 35s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c599f12305bd46a287c46b899417cfdd : SUCCESS in 25m 02s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7661c01edeb243759f40db3b46acc3fa : SUCCESS in 39m 41s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/d420b7322d764268b7f5406b1c14589e : FAILURE in 28m 51s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a4e8a91d4dbe4b24bfddad197d4f8bae : SUCCESS in 55m 22s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/60c5722965d347778bf3d1ee57197561 : SUCCESS in 56m 33s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595f4fc3df524399a14f50f4c8f943bf : FAILURE in 13m 05s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/fb1a1ccc64c04c149892f37a3d47ba56 : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31e239a1c9ffa8c54553b41331403d27189328e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-20 01:24:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"21a8d34f8d36e37deb085e6dbc55516d12e39707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 02:47:23.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7e4179f186948e08b732a41996a4cc1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284fef4fa77940d2b6cc3835f9b68d25 : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bf9987c7f94e43bb94bb5badfd5fe1 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49e7a3c6718f49c5b944839ed5a0520b : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7098bed30eb4fb0bd075af6f92dbca8 : SUCCESS in 4m 44s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/46550dc2b19e40fb9982f2246852b884 : SUCCESS in 42m 20s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/04ba045ee58e4cc983e02acd1fec7fa9 : SUCCESS in 4m 16s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c863be75d1f0497085eb7fe5218a8b46 : SUCCESS in 4m 33s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dd02f42ce49747ebbeb63c5ad8bb0542 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cdf096384e4668b3fe40175e74decb : SUCCESS in 8m 16s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7b15dec0dff24c2592b58cc776ad52a7 : SUCCESS in 1h 00m 54s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/e89381cff7d64730b5fe6513e3e3bb7a : SUCCESS in 14m 39s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/0036ad3b380544c586833f1049881538 : SUCCESS in 40m 45s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/854f2ad83bd94ae680b80aa41084cc3d : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c050060ef5bd45f9ba15dadc4b7e5db5 : SUCCESS in 1h 20m 58s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/f377ea91d0424ea3bda329b7aacf21fd : SUCCESS in 50m 10s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e9feacab56c648f8a1d207c73bd21206 : SUCCESS in 40m 28s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c02707cbffd146e396a26005179700aa : SUCCESS in 22m 50s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0dd3dea8b84643c0af21ced6d9f6a5b1 : SUCCESS in 32m 06s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8cdc4b985962489b83878ed23b20e617 : SUCCESS in 47m 11s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/aa068f3942684a2eb3888f6a467ac31b : FAILURE in 27m 03s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8db011d2eb70443585af3464ce313ebf : SUCCESS in 50m 09s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c69b54107ad949c88d4657932fe99670 : SUCCESS in 54m 16s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c367c1d1590e4c51816b2381624d73f9 : FAILURE in 10m 16s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4ab58533716349cb84547db769d45009 : SUCCESS in 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e14f8e5e46d7d9c04b8a356053b0a5825d7ad01f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-20 12:50:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"78456bee2320d41a3a0286b0cf5471207e40dc0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 14:36:41.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1996a9d002a342f4a6ac9be4d7b56bdc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5ef1ec80304260a63d3667f7b71991 : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/832c3e5541f0438bbe5ff6d1d56bb4db : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e21185175184c578ae01ae4a93e2655 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cc38c71fb7d4c98aef4e8ad80a8086b : SUCCESS in 9m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/93f465106230409c8a63fc58a4e74b0c : SUCCESS in 39m 15s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/43f02c65b1b1453ea8c3a0fa18016de6 : SUCCESS in 8m 27s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b5b2a6429c0949aa85bfaf62b52c1446 : SUCCESS in 4m 07s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/07432e0f3bce469f87bd502708507f90 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e1bc001d81414b868e6413cd8c2ad9 : SUCCESS in 6m 50s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/33fa8b5b0e3a47c4947a8ae1e0d1b56f : SUCCESS in 1h 05m 45s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/9a0c72a25ffe4851b584e22df982aaa9 : SUCCESS in 16m 35s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/9f525e7af7f0474b922a310d746c41a0 : SUCCESS in 38m 34s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d8e2876f19e4c61ba9b881e98a487ef : SUCCESS in 43m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/39d3273cd4a24862b7ca1f311feb9375 : SUCCESS in 1h 30m 12s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/828ed11d793d4d20a3458626dd695b08 : SUCCESS in 40m 50s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9cefff21d58b4ff3a64660f4e8f8d3f3 : SUCCESS in 58m 42s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e26d22ca9fdf4301b0201d3c760c6310 : SUCCESS in 23m 28s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/89703f1504ea453f80c7bd6804ca964d : SUCCESS in 39m 45s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f3c9f330f1b6400d9e68ed8c75ab5282 : SUCCESS in 33m 34s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/cc6e94e5951b4f459eae8906aa39b36e : FAILURE in 42m 59s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7970ad13a14646c38cb472c50631fdad : SUCCESS in 46m 12s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/933d741c9569427abbf6cb59cef9e489 : SUCCESS in 1h 07m 23s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a47962bd30944e9eaa50725d710e6696 : FAILURE in 11m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a98a8f975880491bb0f84faa41853422 : SUCCESS in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ad0c71471db4bdec737ab7cd46c29c92b91bd12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-20 14:53:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"28835c501c8d71a9bb20cadd0ad292d16bd3a7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 16:29:57.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86fe4a31ffc24c928fea2577db3028eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e16b9ae28445019298eb7743004a0f : FAILURE in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70bc208d3246406ba83979dd8964304c : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4d05a4c8c9341fa978898fc8bc3d45b : SUCCESS in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/406982c296bb48a68a83437993d19012 : SUCCESS in 8m 31s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8f8ebaf0c98a4e0a915786bbb15845a0 : SUCCESS in 1h 00m 19s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ce3cf90a17a149f49b8cbc3185eb2c43 : SUCCESS in 7m 27s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3754a9a13d0545c4a95fb29c35b1dfc3 : SUCCESS in 6m 44s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d8af4ab2f5c04b84b90a5dc68d33858f : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f21a984f4104914831d8afeacad0baf : SUCCESS in 11m 51s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5925e8351d0548028201f88eb1bb80a1 : SUCCESS in 1h 04m 53s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/bea04132f96f4c5485ebaa4f41ea02ac : SUCCESS in 17m 43s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/f38c547218814631906b6c89c56a2a3f : SUCCESS in 37m 41s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e70713a81e48639c3762852cf83cdf : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b7f9a6d0f6b040b8b755a44f285e6e48 : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/aafc9ecf36ea49b5866ebaf3d611f0b2 : SUCCESS in 59m 10s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/77c3f392cba6442daffa1c326ec44e03 : SUCCESS in 1h 02m 08s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/043d3637bc194ea2954449247d7a4e71 : SUCCESS in 38m 43s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/442061069a0b49a6b54ab7a97f5e9667 : SUCCESS in 41m 12s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9a17d911f48b4d9db53b8a049640947a : SUCCESS in 48m 35s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/2b77a00eb91e479c9a848c5b06cfc30b : FAILURE in 42m 28s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/141ebb208f534b99a67799cd12a57d1b : SUCCESS in 1h 03m 54s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1fc85df09dea415292178ef685c08f0f : SUCCESS in 1h 11m 05s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c719cded36b480a901b8cf1c45ca79d : FAILURE in 15m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6ae4eea5f22b4b42821b6752c958ff3a : SUCCESS in 1h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"60db1cb41f7935bea9b8f995144095e9cfe662b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-08-20 19:02:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"47dfe2762763d58e5ae02b4ed016119955d85296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 20:38:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01743476c57346a1bb4206940fe7f30d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12ba001fa7145cb8b6028b94b4c539c : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f9901e8b9744921beec096451a36461 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1285a2b0f776445aa50934fa2e3774e6 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eb72909d48e4c028ad5dff730cfdcd9 : SUCCESS in 4m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dd324d3801ca45db9045ac7cf75c5290 : SUCCESS in 57m 11s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ac0b6d4185474df7aac16f8997a2833e : SUCCESS in 4m 45s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d776c132cda04cff9b195708f5ad76ba : SUCCESS in 4m 07s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/aa3fa6b37bf7460897050f74be032357 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594e83b8cfec463895c66dfe919409af : SUCCESS in 8m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/043554d1d5684c599570a5f4dd81de70 : SUCCESS in 1h 04m 58s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/8d676da0e7ac4743a05391fb46622ea6 : SUCCESS in 15m 03s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/f9e397f9baf7451abcbb9cd17ead5595 : SUCCESS in 37m 17s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0848cbbb4374d218f1a5d0a093353e8 : SUCCESS in 56m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4f565b6bb1524a81aed9d2b102a865ce : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/1931bb074f7c46e0aec9605e1340b896 : SUCCESS in 57m 30s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/194a98e014a749b79641763dabc2917a : SUCCESS in 56m 33s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/be37a2aaf51a4e159350d7c0a00f0036 : SUCCESS in 31m 26s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a817c97438b04d8b9f15b3cd110d3e44 : SUCCESS in 39m 15s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1ed4602e951f4b529e580b376c1ff838 : SUCCESS in 43m 22s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/26f941c36dae400b8fa938f40134b192 : FAILURE in 39m 13s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/863f26c8afd24bbcbd11d82b6905f3ff : SUCCESS in 1h 00m 38s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d3d9c89f640847938b6cc60a51fc8973 : SUCCESS in 1h 07m 46s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b543471b9ba744e6ae0cb89975322a07 : FAILURE in 13m 02s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/9cccd12e0faf4ca28ed6b7c1a6411fc2 : SUCCESS in 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ced3fc0c1057578296cedf4a26152a1ad1b53c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"date":"2022-09-02 19:52:18.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"7620d7cb0aebc11ce13fa0380cecf8eece3d8f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 21:20:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2037ee5acd034fad9e05a112fa487c12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65171b01dc6f44db8201134b3cca464b : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c72e78d30264a5897c1956c6d3f23a1 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a140d96d6042f299a88146bea07b32 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cbfdeabf0d44b19985605d494700cc3 : SUCCESS in 3m 21s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/09b1adc516374f5098439d6f08d01041 : SUCCESS in 1h 20m 21s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fcfef6e0b2fe4f43ae22ca8600653319 : SUCCESS in 3m 41s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7eed451c1c2641b5b5b6e89b2d9616b0 : SUCCESS in 3m 32s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ce45a40e29a04894af632d242176bbf7 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb0672c7dc0419f9494675674807eaa : SUCCESS in 9m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/20e9374d0b50429bb855e0fbc71ecf5e : SUCCESS in 1h 00m 55s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/1dd9bedf056a4c9c90a6e50417c34d5b : SUCCESS in 14m 30s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/16ab7778a375495bb074929447e2f50e : SUCCESS in 34m 25s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdb82ad7d3db48a3bf5e4365915c3566 : SUCCESS in 1h 11m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0e86b59e6bff4458aa1d981f9a6b0d02 : SUCCESS in 1h 22m 52s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/6921a4180bc642fcbf165d918bf4785e : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a150253a609f4269b0e0ce1c4a699d2d : SUCCESS in 53m 44s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/8a0262fdcd324a16b0c5084c859199bb : SUCCESS in 31m 00s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c5033269903245a8b32aaa71a2503876 : SUCCESS in 35m 28s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5641a605767d41e283e1c0666a45ddab : SUCCESS in 42m 35s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/2feab3bdb1ed4061a41a6e602b73dbf4 : FAILURE in 38m 06s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7e3d9eeee4a84c5ab0453144fabed337 : SUCCESS in 57m 28s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6f2b7515880844ea91d256d5edb47dee : SUCCESS in 55m 42s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f12b218485084bc1b29b82c7e359b6d3 : FAILURE in 14m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/bfe458181c2b48808df0deaa4829960e : SUCCESS in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e2a48bc55b7148b0e3a455fce59e4caca3f5d9d6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-20 10:05:14.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e7a193fd101707502756e13148ad335f1ab828f9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-09-23 05:49:32.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2a886b9d67fee3f388f023d61a2ba84a54e1a1cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 05:50:17.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c8eecbdb568cd34d347c18002637d3c01e3d68e0","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-09-23 06:10:35.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"80113562cbda52cf780c00dc62ca620765d7a937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 07:12:36.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b023e08b861b4724a6f51d97023181a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7dfc57dac042a9a9b19601c9df5882 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a19fad3804b042bea32f58abadb41803 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b889455bb6fa49628ebe6e3d5a94f5b7 : SUCCESS in 4m 37s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5db9d78afde54ee59a28abc761ab364d : SUCCESS in 52m 35s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/45b5c72cdf77443aa8736fb488ba466c : SUCCESS in 5m 03s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/95c3bf009c3f4d4f80b634242e9beec7 : SUCCESS in 3m 55s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/029306e3e1344929bdad0ea58d9bce42 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141369ca0e894129af7ebae310e48f23 : SUCCESS in 7m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b40f5b74decc4b1a849256d968519327 : SUCCESS in 56m 47s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/4c4f3687a42b43c1b4122ec362811b6c : SUCCESS in 15m 14s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b1b960e02a5344f995786490f05329c5 : SUCCESS in 35m 39s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06a6683ba6c3448586660b06a790263d : SUCCESS in 44m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f1728e16ffed460cbd18a9499033316c : SUCCESS in 1h 21m 26s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/3b6bf7c38b704be487b9eba3ca2b060f : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":10},{"id":"da7003a839d6e2f127233fc20a1f0e4e0642760d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 07:12:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"35af28096bce5c02238bea978aa5fc4ce9cf7239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 07:13:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f55da945195347b48af8f4c672232d05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264623ac24854f5589bb2bfd3c9ea656 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"0536405a445ca906f832d48ca12d5073b5a09cdc","revisions":{"8377d3859732cc26e88a1d974e0aa826c48402f0":{"kind":"REWORK","_number":1,"created":"2022-08-18 20:36:21.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/1"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:36:20.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8377d3859732cc26e88a1d974e0aa826c48402f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8377d3859732cc26e88a1d974e0aa826c48402f0"}]},"branch":"refs/heads/master"},"1866a6ac6e84cb3335b50d3a1e189cd43fe7906b":{"kind":"REWORK","_number":2,"created":"2022-08-18 20:56:56.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/2"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:55:31.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1866a6ac6e84cb3335b50d3a1e189cd43fe7906b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1866a6ac6e84cb3335b50d3a1e189cd43fe7906b"}]},"branch":"refs/heads/master"},"e97aaecd982fdd9769a05b89b65f7459b1307e88":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-08-18 20:57:41.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/3"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai wu","email":"haiwu.us@gmail.com","date":"2022-08-18 20:57:41.000000000","tz":0},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nhttps://docs.openstack.org/api-ref/compute/?expanded\u003dshow-server-details-detail\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e97aaecd982fdd9769a05b89b65f7459b1307e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e97aaecd982fdd9769a05b89b65f7459b1307e88"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a7fb0bf758373bb55088dfac8c9f34c1e3fd7c4b":{"kind":"REWORK","_number":4,"created":"2022-08-19 18:29:04.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/4"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-19 18:28:40.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a7fb0bf758373bb55088dfac8c9f34c1e3fd7c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a7fb0bf758373bb55088dfac8c9f34c1e3fd7c4b"}]},"branch":"refs/heads/master"},"ed78eabdf7fd02f07d9337fba43f87574254e3a8":{"kind":"REWORK","_number":5,"created":"2022-08-19 18:31:19.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/5"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-19 18:31:03.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ed78eabdf7fd02f07d9337fba43f87574254e3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ed78eabdf7fd02f07d9337fba43f87574254e3a8"}]},"branch":"refs/heads/master"},"fa5415ccaa6779ab1f2fc790fdb5793934a7635c":{"kind":"REWORK","_number":6,"created":"2022-08-20 01:24:12.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/6"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-20 01:23:52.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fa5415ccaa6779ab1f2fc790fdb5793934a7635c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fa5415ccaa6779ab1f2fc790fdb5793934a7635c"}]},"branch":"refs/heads/master"},"5467c6c71e4a3a5a53e67b5b89b0f1c93ae0ae31":{"kind":"REWORK","_number":7,"created":"2022-08-20 12:50:33.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/7"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-20 12:50:00.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5467c6c71e4a3a5a53e67b5b89b0f1c93ae0ae31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5467c6c71e4a3a5a53e67b5b89b0f1c93ae0ae31"}]},"branch":"refs/heads/master"},"cb6678a6c99cf0aa3676c97825ed93ff8ab1bc93":{"kind":"REWORK","_number":8,"created":"2022-08-20 14:53:05.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/8"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-20 14:51:56.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/cb6678a6c99cf0aa3676c97825ed93ff8ab1bc93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/cb6678a6c99cf0aa3676c97825ed93ff8ab1bc93"}]},"branch":"refs/heads/master"},"81b04881b45ba79eada3b214461cf4406f435593":{"kind":"REWORK","_number":9,"created":"2022-08-20 19:02:29.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/9"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-20 18:02:06.000000000","tz":-300},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/81b04881b45ba79eada3b214461cf4406f435593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/81b04881b45ba79eada3b214461cf4406f435593"}]},"branch":"refs/heads/master"},"0536405a445ca906f832d48ca12d5073b5a09cdc":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-09-02 19:52:18.000000000","uploader":{"_account_id":34257,"name":"hai wu","email":"haiwu.us@gmail.com","username":"hai"},"ref":"refs/changes/03/853703/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/853703/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/853703/10"}}},"commit":{"parents":[{"commit":"433815dde50cc253e33f56b5e7babd64137174e0","subject":"Merge \"resource: Merge unnecessary separation of logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/433815dde50cc253e33f56b5e7babd64137174e0"}]}],"author":{"name":"hai","email":"haiwu.us@gmail.com","date":"2022-08-18 20:32:29.000000000","tz":-300},"committer":{"name":"hai wu","email":"haiwu.us@gmail.com","date":"2022-09-02 19:52:18.000000000","tz":0},"subject":"Add support for fault object per Server API","message":"Add support for fault object per Server API\n\nIn Openstack CLI, we could do this: openstack server show VM_UUID -f json | jq \u0027.fault.message\u0027.\nBut the same functionality is missing in openstacksdk.\nThis particular patch will address this issue.\n\nChange-Id: I05101d5154ee5087d50c5cbf347054fdb1f582fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0536405a445ca906f832d48ca12d5073b5a09cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0536405a445ca906f832d48ca12d5073b5a09cdc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[]}
