)]}'
{"id":"openstack%2Fopenstacksdk~860399","triplet_id":"openstack%2Fopenstacksdk~master~I95af9b81e6abd51af2a7dd91cae14b56926a869c","project":"openstack/openstacksdk","branch":"master","topic":"better_tests","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-03-15 07:43:24.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I95af9b81e6abd51af2a7dd91cae14b56926a869c","subject":"Implement acceptance test job","status":"MERGED","created":"2022-10-05 10:27:19.000000000","updated":"2023-03-28 08:47:16.000000000","submitted":"2023-03-28 08:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860399-better_tests","meta_rev_id":"44f7cbcf6ab968fdbf3a8040bf2d980c847dfbc4","_number":860399,"virtual_id_number":860399,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-28 08:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-28 07:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-28 07:16:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-15 13:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 11:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 09:41:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"151db464daa092eea2e77e6b54f3cd7284953148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 10:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b677affde73290e0f8b8e936c1be3711e6faece5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 10:27:46.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"204be3d12afcb76fadf1050e32d49ad23a542d45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 10:30:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d6dc576544b95c2fa6876b1b232889eed833943","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 10:45:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f4b862e7a74d7e19419ccfa21939d196c7fb635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:10:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1a83f19641843c0966be0274eac0833\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/ff70e97e5a9e4a7c9ada7d1692d2c9fc : POST_FAILURE in 23m 24s","accounts_in_message":[],"_revision_number":3},{"id":"913e05f17d9ab74fb26790fdf859fbbea936d02f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 11:10:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee5037f9f8f6b0b34610249dc9e95a3c7d62ce3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:40:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f5e3cfa949049dea62c8e6892f331bd\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/2d86371ad6ea49068bfb0cabafc2c441 : POST_FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":4},{"id":"863b8f3073a0e2b75697c3ee668d6f00db1be009","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 11:51:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"13e426f94949f871df9c69fe85c3276ed24c7313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:27:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6825b3be51c4f9fb5272538e237c69a\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/f439957eb48f4fe4805b072c3c4c5b4b : POST_FAILURE in 34m 06s","accounts_in_message":[],"_revision_number":5},{"id":"08baa45c0e53f80165a978c9b6dd95f6da6b576f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 12:51:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7fb7aaa35c266e07fe9498b0fa157dab7c5be8f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:33:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afc9bf6b4c70469a8f0fa3ae2190e207\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/26d1faef7e114ca7a4044f1915c4f34d : POST_FAILURE in 36m 50s","accounts_in_message":[],"_revision_number":6},{"id":"480e68f86d13b3752ddd137fe3f794f3b9e5142b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 13:40:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"144c6eb7254fda087031fe4b4a4a5a920da468bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:36:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3be66f882044b8aac5e606429fd6be8\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/090c136e2cc2498a85bfdfffb0eb1aab : POST_FAILURE in 50m 36s","accounts_in_message":[],"_revision_number":7},{"id":"7778141acd54e67078cfd6843efcb4f05990834c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 14:37:15.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"b4245d1d29b0f5b1d0fbd811c4e6ec828de1536b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 15:19:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a49814391dbe4cfbba2356a0b772412a\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/eb74c634172d48e0a5d22500d251934d : POST_FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":8},{"id":"22354777e987fad0a778a15c0299817a579865ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 15:55:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"018b2181e139ba33df72f37e4fa750f9b5ccc40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 16:24:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2cf2ba4fdbb4d4f80c111a83264b2e7\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/8a126c716ea145eeade80e8ba5dd2ef1 : POST_FAILURE in 27m 46s","accounts_in_message":[],"_revision_number":9},{"id":"a7c09a689199edbab543b410666a0ba28dc6e96c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 16:29:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0088bb0b02ae3fd579678c9e09a3886982e08579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:19:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6ec64963f544428a950c1ed66075078\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/f99e579373cb460b8e84f6de0aa3acbb : POST_FAILURE in 43m 54s","accounts_in_message":[],"_revision_number":10},{"id":"72efa4bdbcf189c61436e5c5613b2b34875f9bcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-05 18:39:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"73cb9841d4af3eae75b5b03b7e2641b8a89047f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 19:08:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1736aee677114822a4162401ec27bcc8\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/10f623cb707e4f94a4c74a968192d103 : POST_FAILURE in 28m 15s","accounts_in_message":[],"_revision_number":11},{"id":"a0e0555a3042b47fbf1829908561ae1bcde1bc2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 06:39:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2f1fee83155a17c6d48e5e7fdf7ff42975c5e10b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 07:22:26.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d95d71ce328d46d2afa3d8483d04fd8d\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/f6675d22e6a044ed98e46e6c235976ad : FAILURE in 37m 18s","accounts_in_message":[],"_revision_number":12},{"id":"0cc9ab5be3e1c2cb057e3064646368a413101d3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 11:07:50.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"99f36f655b8367814ab45ecbd5ba06cca840025a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:56:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba9bcd8b33a94150ac3e56fc3362cc52\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/d52d868425894d978edca2ac238bd4c5 : FAILURE in 41m 42s","accounts_in_message":[],"_revision_number":13},{"id":"280128141605122fbf388553f343da25f13485ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 12:08:53.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5bae800af8a2756ffdcb1d75987eeee265b67dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:49:48.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e4d87235ff0421ca447b79b972635e3\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/c97e551e542f4efcb4ec1b860b85703c : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":14},{"id":"67b39739bb42750aa58e7e64153851cfec4d2358","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 13:20:09.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8ff4464f0f063df1009acfc834f366d8714e90cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:04:21.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73e7d22eeda34cf8b44d6b344e61a464\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/d50b19020365496e90151e9f360c58b5 : FAILURE in 42m 25s","accounts_in_message":[],"_revision_number":15},{"id":"8d2a77c6561f0656c47da25cd711d53d1d84e3b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 14:25:10.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6d6095d08a3034a34c484b70e4889c83fa9a2f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:03:58.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3287dbf3ecd4653b9ad947b49c76313\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/48faa8499ddc4f8fad708b7d4d13775c : POST_FAILURE in 36m 32s","accounts_in_message":[],"_revision_number":16},{"id":"fb585e101ad10b38ee8bf0c704e0e351b7574b0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 15:18:51.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"882b1f2ba961608eb6359e25ccef128086c391f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:54:09.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9817c0cb4d643bb95c0b10e3aecce4b\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/a4ba7487079f4e7d832eda17ef480fbb : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":17},{"id":"657d4b50d0f423766b9be4c58d0f2a2e8b3dd11a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-06 16:05:06.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9889b407d4a776b36d0ee6983d4d31c7e970e046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:38:59.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/915a74e8beba4437a4d359f19123f7d1\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/579d46656503492fbce3045a21585140 : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":18},{"id":"943fe1c5788e94551e210b62d1d195c616ccd7ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-07 07:11:57.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"dc7e1c78cd3bcf07c0ffdb10304310ab3b973c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 07:37:36.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eedb052533c4492d939dce350b7406bb\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/396e65f0279e4430a124bd182352ac79 : POST_FAILURE in 24m 00s","accounts_in_message":[],"_revision_number":19},{"id":"97314ccc6d1d454a0fd622e766515abb47b48723","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-07 10:05:24.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"2be2f430631bee8e62e1eded58d34561efd53093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:26:42.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b6aff90d64f4e57b5a9a104c86eb8bc\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/4af09e1d2d0f4efda8eb30b522a59b02 : POST_FAILURE in 19m 29s","accounts_in_message":[],"_revision_number":20},{"id":"f7447e554415358bcd153be7b131121119f726c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-12-02 13:02:56.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"370cb60f97565778755bda5856d688ae5c6ef40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 13:30:08.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f80f0094266c46e69de0ad156d2d72c5\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/599bad6660cd4d0c8b6eb2c6d4c8ca94 : POST_FAILURE in 25m 01s","accounts_in_message":[],"_revision_number":21},{"id":"868bfd64e30d2320e4244d6dbfc730333b84f127","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-12-05 11:50:08.000000000","message":"Patch Set 21:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":21},{"id":"4d9d2ed9e4c3bc726109ebeac0c34c7a44db5288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 12:29:10.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5410936a268d4b1fbbabeb74f0389be9\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/c79f9475feca47648c3548a55cc8f71d : POST_FAILURE in 33m 17s","accounts_in_message":[],"_revision_number":21},{"id":"32214684e64d70f084cf08b8b6d93d4ab479d8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-01-04 08:45:40.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"c0567647ec531b2676fdb7f1be1b9ca6d73401f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 09:26:00.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d46905a9f53425c85e6b0c311e9e148\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/7cae6b7107aa4ccd8f7ba2e5f94b89d1 : POST_FAILURE in 19m 22s","accounts_in_message":[],"_revision_number":22},{"id":"d096758f07dd05df085256bae2889fe657d8c5e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-01 08:25:22.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"cdfef128e14b1b206ee176df3b757c1d70e89181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 08:58:14.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94f24a1ba29d499090fc71cf479aa543\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/47f78528bcdb4946be5e1d0aa21c962a : POST_FAILURE in 31m 06s","accounts_in_message":[],"_revision_number":23},{"id":"8d5325ea9c7c4c9c37c1df07db3071ce9e42d495","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-15 07:45:01.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"41c2bf795f082d0e7a2aa4185a6f0c0fa9b9c57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:12:08.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c3433b0da8b4599a676be044c79fba7\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/e567ab6da42849a5ad5045a1c4bb32bb : POST_FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":23},{"id":"0f80847afeb7a6fba805feac698e0134d5ed94f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-22 09:15:05.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"6d5a5c16b740e2db4b20ab131c2e941f3b05f5e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:43:06.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bcfc7651b1549289b419cda8dbbbd77\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/43b3e290d5874601a6417f70b40ac64f : POST_FAILURE in 27m 04s","accounts_in_message":[],"_revision_number":24},{"id":"ea705c1f44c09aaea5d57a4d3dcd71b4b306b304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-22 10:05:35.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"562250c2f52642b5157549b329acc7d3c6f74e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:54:33.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ba89849f7247d79c6dc1c37aea8c67\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/effaf566f33843cca2d4db834fcdc1ee : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":25},{"id":"d58f0225d92c5e9003817b61d11566b60b589977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-22 13:37:23.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"b7e4a0b4cee1fcd88e5c6472d360fa6c112d516b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 15:02:01.000000000","message":"Patch Set 26: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f031a0fd8a064dd7b49aa783794b5853\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff422ff92374b479bd7a7f034cf841b : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f11b45789a8b4171b7f6cc201a33a018 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d313b739aed4c3880f0efb9c3ee6704 : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f726eb89fceb4ecc8b240876cccf9233 : SUCCESS in 1h 00m 03s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/555aef1574fa4740b73c42403fcac150 : SUCCESS in 3m 45s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e5f1f6daf9834221805415066bfa0cca : SUCCESS in 4m 23s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b79b83d5ba324081a90cfd300a340985 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a338b651a64cdeb564aeb6427426c4 : SUCCESS in 6m 57s\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/cb2cde5b413c491a91fc9f8cbcbaaa92 : SUCCESS in 40m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fb88294da9284cf7aac1ef2cf73a492c : SUCCESS in 54m 33s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/8d16a595959f4c49b84738cf03976c38 : SUCCESS in 13m 36s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/5f2b9342acaf4dce98726c1ed40dc04f : SUCCESS in 33m 24s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0ebebb82742484caeec2ffe899c7044 : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2f452942c7a5485db7d7ad7931efd4f3 : SUCCESS in 1h 20m 11s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/c21c3e66610c4d5cbadf621dabf6b937 : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/32f764c3b8a6435f8743c82cdd55325a : SUCCESS in 50m 08s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d48fcb5afb9a471097e91348dbdeb88c : SUCCESS in 23m 22s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c895141b6401480f9e3b7bfeacd432fd : SUCCESS in 25m 24s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/68d3c331760e4a27b6509d909d11fee5 : SUCCESS in 31m 49s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/03a613f4e7a44fb0a709447c3175f012 : FAILURE in 28m 24s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/227e2f750ec34e5c8d4d44a36007d66d : SUCCESS in 1h 02m 28s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0e77137877e3417eb8061202e6208522 : SUCCESS in 52m 41s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/65056690ce1e4cf1a4a7eef8f307c0c8 : SUCCESS in 33m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/467101ad9b7842848a4af1ee4a3f1f4f : SUCCESS in 51m 51s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"5f0aaf485b096e859a2ab06b3b58eb7a0cd70265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 07:42:29.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"c4e12c3350fae7efd1e2293da9482b33cb020e7f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 07:43:24.000000000","message":"Patch Set 27: Allow-Post-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"221582c1cb74a7ed738defd7ca5b06c73047e87d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 07:54:02.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"18892a7d8ad519950a3bf7b5533f057577d4039f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 07:54:51.000000000","message":"Patch Set 28: Allow-Post-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"c79b08787816e4e7043a0a2d87ffe2f8a6e48e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 09:17:16.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc517b8a0eea4e4db21bdf4fe2e85947\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649b859d1b3049458e3e10d82bc65abd : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c018ecbd7fd428396f1b5902b9e4acb : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70484d45375d4187a35b0e8fe5f56f40 : SUCCESS in 4m 30s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fa938d50a5ed493ebccc6cffbc41e16a : SUCCESS in 55m 37s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9f5086d5b58f4e1aa84ab8d3bb274dad : SUCCESS in 4m 20s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/84af2199b85346148836ea06657859fe : SUCCESS in 4m 14s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/688ab66ec25142a197862e6100149416 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35cc5e36ae02402580ab48b246c0ec22 : SUCCESS in 10m 23s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8bb4c7bc4c2c47fbb6b6c435777a8e6f : SUCCESS in 52m 45s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/267e32d8dd7a425d81a7fa92c3983000 : SUCCESS in 16m 08s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b977d6ef747f4ac59a558d87640b8d48 : SUCCESS in 31m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e2807014b1047ca8e0652d811af77f7 : SUCCESS in 1h 02m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4acd78c31ca240728029a913b810fd79 : SUCCESS in 1h 20m 46s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/5f174ce2b698466db78ea27ff5ceda87 : SUCCESS in 54m 59s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/44431d06ad1944a3a1de4c4b60266afd : SUCCESS in 38m 09s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4c5cd3eb40f44351b9e4bc4d22e759ad : SUCCESS in 18m 00s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0a2e5aae97cd4e9599afbd3b6bb6b305 : SUCCESS in 38m 52s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/21b8715c354b4ea3b9689b5fb95d75a5 : SUCCESS in 42m 10s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/4e4ee93c3388411abe7cbfe6a9bf46e6 : FAILURE in 39m 31s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2e9225463179487aaa7f8405288ae224 : SUCCESS in 1h 00m 40s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6bcf29b12ea4405c8995503009493a7e : SUCCESS in 58m 41s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a8feb79c26c649beaad815e621cb01b8 : FAILURE in 9m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/47e04291fcaf442d8641fae29b65604e : SUCCESS in 1h 09m 29s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3d945199d0a3ee1980558b990c20ad58242d5f86","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-15 09:41:37.000000000","message":"Patch Set 28: Allow-Post-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"98f91a60cd00535b659bf6e2839f5f853de8e363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 11:00:45.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"5fd69bbf5e88ae33a2416f73ced4bdad212343fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 11:27:46.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"494ada29826aadd67755e6fa339992a42c23b8e4","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 11:28:19.000000000","message":"Patch Set 30: Allow-Post-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"1c25b92b3693c6a00c6b0004014d91c35c711b1f","tag":"autogenerated:zuul:post-review","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 12:20:21.000000000","message":"Patch Set 30:\n\nBuild failed (post-review pipeline). For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/660d76c768a4438bbb56a362e14bf706\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/14bb051389224fd39359ac10cf8f5140 : POST_FAILURE in 45m 29s","accounts_in_message":[],"_revision_number":30},{"id":"f55d6d6059946f7f91cb919155e8014a80debff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 13:14:53.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9effc71edcf648dfb405c855a9b46323\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f08c2f75064464bb6f942650f3413f : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d133338aff497196c9a591aa8b19c1 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8a3b09b8d184ed09f66c76032c45f86 : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e8fcaed3d006429eb6a47ee8a7a05eef : SUCCESS in 1h 04m 01s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7ed032d2f4984234a9ce7d347ab797e9 : SUCCESS in 5m 01s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ce26bc4ffcdb4bb3957046e9b219d40b : SUCCESS in 3m 58s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c17a959dd13e4a2c8764913ed42caa64 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dec6df5cc78426c83a35ea30a8c2b7f : SUCCESS in 7m 58s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e9b7deddf7684b35990cfe1faaa34da9 : SUCCESS in 1h 12m 19s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/7cbe92a4d1d34602bbcd3d18708f07cc : SUCCESS in 13m 59s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/be85c13382ae4d73811ef88483fa386a : SUCCESS in 30m 52s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0326ec16a1d8475ca3a1b3cb5409c7a7 : SUCCESS in 1h 02m 26s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/45e83896cd0f40a68f93fa4c19582ec8 : SUCCESS in 1h 30m 54s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/89df42a988204a3da3ccde3f0a83d0d5 : SUCCESS in 1h 04m 26s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2bc7fab1ee6148818210b029ac934731 : SUCCESS in 52m 47s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/06429e34f07f4a7fbbc47c059e9937fc : SUCCESS in 30m 31s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/29ecc5eb89bd4925887a8c86cef2a1b9 : SUCCESS in 29m 06s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8c2d83fcc5884e7289f6b1a5ba4eed98 : SUCCESS in 40m 16s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/de8ce6940349457587301e82bd9f5bba : FAILURE in 36m 24s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/98ae0a986db742a0956782355480a1a6 : SUCCESS in 1h 07m 26s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8cdcf9f3bfe84c81a62bef6cfc5fae60 : SUCCESS in 58m 59s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/943e5b3058f74efc9007b483159e1bd9 : FAILURE in 9m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/e9bffd7819874a5c8fd4138ed117ac26 : SUCCESS in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"82911393502b9d438996a12b9e28a4bc070832d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 13:49:33.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"2a55cd2572e4d7e0d27260c18eedc712c639df57","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-15 13:49:41.000000000","message":"Patch Set 31: Allow-Post-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"a4d5513b2f301627614479a6a853de1a9c5d43a3","tag":"autogenerated:zuul:post-review","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:32:26.000000000","message":"Patch Set 31:\n\nBuild succeeded (post-review pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cc16a15855a4317bc65f71926af2825\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/26d4cf7b80564f8fbc90eb6abbec86b7 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":31},{"id":"03aff904d0743d659d85108f494e50be296977cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 15:02:36.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2f4daf3c4b481b9c2a60da3c854fce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9aeed702734061a3c88254d27dfb66 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e724765676fa4eaa885d46973c71c257 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feae7c63589c4eff8753c9d40b1fc496 : SUCCESS in 3m 20s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/740da13d3c7649e082065de615de497e : SUCCESS in 57m 09s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/af690a2da39e48ebacf57dbc547c9e3d : SUCCESS in 4m 57s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0fd858d27970460c895960492809f36d : SUCCESS in 3m 47s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a28cb71085c945a68ebe9faf8345cf48 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469710eb48c54c3e91fdab25ee585ab6 : SUCCESS in 7m 30s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/499c14ee0d854a82a1f93f68fb819141 : SUCCESS in 1h 07m 25s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/c34547ce606c4bee9603adcbbf95f566 : SUCCESS in 13m 39s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b0c2e35d091b4c6389f7716d942f510a : SUCCESS in 40m 21s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d70a47d1dc634d978b3e435436eb51d2 : SUCCESS in 50m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/04454b9cb67442089f449d722d316873 : SUCCESS in 1h 09m 40s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/2284b9b3023a41aab49d1943bd753d0b : SUCCESS in 50m 03s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cadd0ff4e6104765ab65cc6b615df382 : SUCCESS in 53m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ca427e4fa34b4e9a858fa31593c7afc9 : SUCCESS in 34m 20s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/443556359ab7441488cee6289ae57d1f : SUCCESS in 35m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/29af13085ecb44f4a232b70dccacbbb1 : SUCCESS in 34m 23s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/85c018ffa3e045379a52f9338410544d : FAILURE in 33m 52s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e9a119bc3ca14a9e92e76d1fbaaa7833 : SUCCESS in 1h 02m 17s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d3305886c0744da8a0cca5ef3092e4cc : SUCCESS in 45m 35s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4ae63ed8383b45548f992ecd15c7826a : FAILURE in 9m 30s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/81d7ee85b533438d8feae178203a063b : SUCCESS in 51m 35s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c7a0c14161b8e063f16bca462056a9741edea1be","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-28 07:16:24.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"8397048998d2be86159ead6dbfaaabadec7b71ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 07:17:09.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"f1dd953a6d51ca7f4a5509a92964cff3ce9f190c","tag":"autogenerated:zuul:post-review","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:07:54.000000000","message":"Patch Set 31:\n\nBuild succeeded (post-review pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ab57a7cf024d69b12c06f01e1c0785\n\n- openstacksdk-acceptance-devstack https://zuul.opendev.org/t/openstack/build/d62601d18fa441f59182964e9772c521 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":31},{"id":"79c654b3c5b540b2da298169f71d7a5f587511e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:44:55.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0fc82cdcbbc43c7b97a588e7dabe4f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401daf1362014bda9b7e6ecc0c8d8d4b : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f965578743f14d5d955da778c9f0495c : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b88d101635014d2dad2eb4c306646c25 : SUCCESS in 4m 25s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dd621af364754f42b6c82143687501ab : SUCCESS in 44m 30s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0043780a2d66400ba0cd723990b5c038 : SUCCESS in 4m 01s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5b6068f5a0444dff9293e118d08161ef : SUCCESS in 4m 02s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5c307907cf4246668339893aceea3e2b : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15a40915fda4e81b6d4d2834f8e08c9 : SUCCESS in 8m 45s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5f6e6647a906404a82c06d46c683b497 : SUCCESS in 51m 52s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/52ad7ebd3d834004b3ca14a16936c88c : SUCCESS in 14m 14s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/a849bfdc0a2747a6966edaf3b3e17cfd : SUCCESS in 32m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8278fb5467e74999859dfce98a7f76e7 : SUCCESS in 1h 03m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c929982d5a5f481d8f916bf573700465 : SUCCESS in 1h 17m 17s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/ad5ea0e968004260a42aa6a8cb86c3cc : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":31},{"id":"5ba2da9831b12b1e4e68eff66617404a4c402b62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:44:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":31},{"id":"44f7cbcf6ab968fdbf3a8040bf2d980c847dfbc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:47:16.000000000","message":"Patch Set 31:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa099657d11e4589ac32d8b9caea38f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7cd7fd0a44c41f0a62820e0c2be1061 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"43ab59d8b38152e255ebd0501a4d763eb815ad03","revisions":{"5bf3a95c68c17507ae2c36da04f87fb3f12d73cd":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:27:19.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/1"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5bf3a95c68c17507ae2c36da04f87fb3f12d73cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5bf3a95c68c17507ae2c36da04f87fb3f12d73cd"}]},"branch":"refs/heads/master"},"24db36779737c57edf895f6925f4513d2711f3dd":{"kind":"REWORK","_number":2,"created":"2022-10-05 10:30:53.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/2"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:30:45.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/24db36779737c57edf895f6925f4513d2711f3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/24db36779737c57edf895f6925f4513d2711f3dd"}]},"branch":"refs/heads/master"},"a65632b5461f1041e8f859e920c5417287317fd2":{"kind":"REWORK","_number":3,"created":"2022-10-05 10:45:15.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/3"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:45:07.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a65632b5461f1041e8f859e920c5417287317fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a65632b5461f1041e8f859e920c5417287317fd2"}]},"branch":"refs/heads/master"},"27fa6b7e4ff7d2372c9f5ce70a42a9f23ad19c3e":{"kind":"REWORK","_number":4,"created":"2022-10-05 11:10:29.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/4"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 11:10:20.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/27fa6b7e4ff7d2372c9f5ce70a42a9f23ad19c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/27fa6b7e4ff7d2372c9f5ce70a42a9f23ad19c3e"}]},"branch":"refs/heads/master"},"64aec37edec42ea9ea34c08f09bc5253b1f10de2":{"kind":"REWORK","_number":5,"created":"2022-10-05 11:51:25.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/5"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 11:51:14.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64aec37edec42ea9ea34c08f09bc5253b1f10de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64aec37edec42ea9ea34c08f09bc5253b1f10de2"}]},"branch":"refs/heads/master"},"84ac7bb9886b62eedb6261095b0ac4c6799f475b":{"kind":"REWORK","_number":6,"created":"2022-10-05 12:51:18.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/6"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 12:42:26.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/84ac7bb9886b62eedb6261095b0ac4c6799f475b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/84ac7bb9886b62eedb6261095b0ac4c6799f475b"}]},"branch":"refs/heads/master"},"2fdc0da6ae49e9e743578c46b34a72f2d3f9042b":{"kind":"REWORK","_number":7,"created":"2022-10-05 13:40:19.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/7"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 13:40:10.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2fdc0da6ae49e9e743578c46b34a72f2d3f9042b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2fdc0da6ae49e9e743578c46b34a72f2d3f9042b"}]},"branch":"refs/heads/master"},"f11dad476dfdda3479ac7fcd09ab3667bf99cb1c":{"kind":"NO_CHANGE","_number":8,"created":"2022-10-05 14:37:15.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/8"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 14:37:04.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f11dad476dfdda3479ac7fcd09ab3667bf99cb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f11dad476dfdda3479ac7fcd09ab3667bf99cb1c"}]},"branch":"refs/heads/master"},"9cf4243de274ae97c3352de3b8c91a5c6e0566fd":{"kind":"REWORK","_number":9,"created":"2022-10-05 15:55:07.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/9"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 15:54:57.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9cf4243de274ae97c3352de3b8c91a5c6e0566fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9cf4243de274ae97c3352de3b8c91a5c6e0566fd"}]},"branch":"refs/heads/master"},"0faaad496dee02a8347986a8807cc789b8c9ad71":{"kind":"REWORK","_number":10,"created":"2022-10-05 16:29:36.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/10"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 16:29:24.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0faaad496dee02a8347986a8807cc789b8c9ad71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0faaad496dee02a8347986a8807cc789b8c9ad71"}]},"branch":"refs/heads/master"},"a412d3faffc82e849e9c937e63bd6f65e558277c":{"kind":"REWORK","_number":11,"created":"2022-10-05 18:39:27.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/11"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 18:39:18.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a412d3faffc82e849e9c937e63bd6f65e558277c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a412d3faffc82e849e9c937e63bd6f65e558277c"}]},"branch":"refs/heads/master"},"db755a39a6cb147524f5cbcae485df616400560d":{"kind":"REWORK","_number":12,"created":"2022-10-06 06:39:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/12"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 06:39:29.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/db755a39a6cb147524f5cbcae485df616400560d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/db755a39a6cb147524f5cbcae485df616400560d"}]},"branch":"refs/heads/master"},"b13ab68c0a2d4648c0ee170bf4f9ccf93d610351":{"kind":"REWORK","_number":13,"created":"2022-10-06 11:07:50.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/13"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 11:07:39.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b13ab68c0a2d4648c0ee170bf4f9ccf93d610351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b13ab68c0a2d4648c0ee170bf4f9ccf93d610351"}]},"branch":"refs/heads/master"},"020646faa354eee18d967573cea662dd917eaee7":{"kind":"REWORK","_number":14,"created":"2022-10-06 12:08:53.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/14"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 12:08:42.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/020646faa354eee18d967573cea662dd917eaee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/020646faa354eee18d967573cea662dd917eaee7"}]},"branch":"refs/heads/master"},"bc9de693c05f7a7e8008e8f3a59b88ecfe2def91":{"kind":"REWORK","_number":15,"created":"2022-10-06 13:20:09.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/15"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 13:19:59.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc9de693c05f7a7e8008e8f3a59b88ecfe2def91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc9de693c05f7a7e8008e8f3a59b88ecfe2def91"}]},"branch":"refs/heads/master"},"5e00c0ed220c8c80be7edcb947c6671d2cb927cd":{"kind":"REWORK","_number":16,"created":"2022-10-06 14:25:10.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/16"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 14:24:59.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5e00c0ed220c8c80be7edcb947c6671d2cb927cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5e00c0ed220c8c80be7edcb947c6671d2cb927cd"}]},"branch":"refs/heads/master"},"358d4d91c49ca349500aa55278b7293ea4ba2958":{"kind":"REWORK","_number":17,"created":"2022-10-06 15:18:51.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/17"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 15:18:42.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/358d4d91c49ca349500aa55278b7293ea4ba2958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/358d4d91c49ca349500aa55278b7293ea4ba2958"}]},"branch":"refs/heads/master"},"da4aa1f14e9f5463f1442161778e21730878010c":{"kind":"REWORK","_number":18,"created":"2022-10-06 16:05:06.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/18"}}},"commit":{"parents":[{"commit":"d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d94221cde7f2fdc0e3ce89c0e3ce56f58c94e7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-06 16:04:55.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/da4aa1f14e9f5463f1442161778e21730878010c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/da4aa1f14e9f5463f1442161778e21730878010c"}]},"branch":"refs/heads/master"},"b24dbf9b43f29c5cb2e17ad93e129e9d8e4e8999":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-10-07 07:11:57.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/19"}}},"commit":{"parents":[{"commit":"1e6366bf83ec6a5a811845be88f90cf4d534c400","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1e6366bf83ec6a5a811845be88f90cf4d534c400"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-07 07:11:21.000000000","tz":120},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b24dbf9b43f29c5cb2e17ad93e129e9d8e4e8999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b24dbf9b43f29c5cb2e17ad93e129e9d8e4e8999"}]},"branch":"refs/heads/master"},"19fe7e9c0f6caa9d097142af7c74a5cbd7c1ac15":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-10-07 10:05:24.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/20"}}},"commit":{"parents":[{"commit":"236f986f6727f67efa8537b21ca2270d34836d8a","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/236f986f6727f67efa8537b21ca2270d34836d8a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-07 10:05:24.000000000","tz":0},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/19fe7e9c0f6caa9d097142af7c74a5cbd7c1ac15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/19fe7e9c0f6caa9d097142af7c74a5cbd7c1ac15"}]},"branch":"refs/heads/master","description":"Rebase"},"f7e38958cd59072df37577dd86d00ec64f1afb1c":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-12-02 13:02:56.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/21"}}},"commit":{"parents":[{"commit":"ae22f1532d8454e84fdc6f541f2458c877c1c194","subject":"Whitelist cloud functional tests in acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ae22f1532d8454e84fdc6f541f2458c877c1c194"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-12-02 13:02:35.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f7e38958cd59072df37577dd86d00ec64f1afb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f7e38958cd59072df37577dd86d00ec64f1afb1c"}]},"branch":"refs/heads/master"},"4f6384e1eba79aaba194629c582f106b1469a6dc":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-01-04 08:45:40.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/22"}}},"commit":{"parents":[{"commit":"d57c1fcab3b6cbe806cbae735fefa4983b200ab2","subject":"Merge \"Add \u0027details\u0027 parameter to various \u0027find\u0027 proxy methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d57c1fcab3b6cbe806cbae735fefa4983b200ab2"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-01-04 08:45:40.000000000","tz":0},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4f6384e1eba79aaba194629c582f106b1469a6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4f6384e1eba79aaba194629c582f106b1469a6dc"}]},"branch":"refs/heads/master","description":"Rebase"},"c26e9986e0968bdfed3ccb54c580052122a2e305":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-02-01 08:25:22.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/23"}}},"commit":{"parents":[{"commit":"3876f3f6d9d45f22de7429e14744e66439d90fdf","subject":"Update README to indicate COE resource/proxy support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3876f3f6d9d45f22de7429e14744e66439d90fdf"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-02-01 08:25:22.000000000","tz":0},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c26e9986e0968bdfed3ccb54c580052122a2e305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c26e9986e0968bdfed3ccb54c580052122a2e305"}]},"branch":"refs/heads/master","description":"Rebase"},"58de3777c87489be7b8325e46b3bd14e10bd3b6b":{"kind":"REWORK","_number":24,"created":"2023-02-22 09:15:05.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/24"}}},"commit":{"parents":[{"commit":"3876f3f6d9d45f22de7429e14744e66439d90fdf","subject":"Update README to indicate COE resource/proxy support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3876f3f6d9d45f22de7429e14744e66439d90fdf"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-02-22 09:13:48.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58de3777c87489be7b8325e46b3bd14e10bd3b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58de3777c87489be7b8325e46b3bd14e10bd3b6b"}]},"branch":"refs/heads/master"},"b3e60bf771f403cbdacda301ff81a2d0a73482fd":{"kind":"REWORK","_number":25,"created":"2023-02-22 10:05:35.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/25"}}},"commit":{"parents":[{"commit":"3876f3f6d9d45f22de7429e14744e66439d90fdf","subject":"Update README to indicate COE resource/proxy support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3876f3f6d9d45f22de7429e14744e66439d90fdf"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-02-22 10:05:29.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b3e60bf771f403cbdacda301ff81a2d0a73482fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b3e60bf771f403cbdacda301ff81a2d0a73482fd"}]},"branch":"refs/heads/master"},"5243fc9dd220cf35debeef07fc9ba44e60331562":{"kind":"REWORK","_number":26,"created":"2023-02-22 13:37:23.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/26"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-02-22 13:37:02.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5243fc9dd220cf35debeef07fc9ba44e60331562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5243fc9dd220cf35debeef07fc9ba44e60331562"}]},"branch":"refs/heads/master"},"157cbc050157bf1043d9d1fdd5e4fef064989b8f":{"kind":"REWORK","_number":27,"created":"2023-03-15 07:42:29.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/27"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-03-15 07:42:16.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/157cbc050157bf1043d9d1fdd5e4fef064989b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/157cbc050157bf1043d9d1fdd5e4fef064989b8f"}]},"branch":"refs/heads/master"},"38e917d8319bbc19070fd93a7f49a5bd05080cec":{"kind":"REWORK","_number":28,"created":"2023-03-15 07:54:02.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/28"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-03-15 07:53:13.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/38e917d8319bbc19070fd93a7f49a5bd05080cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/38e917d8319bbc19070fd93a7f49a5bd05080cec"}]},"branch":"refs/heads/master"},"f2e5b7aaa1d418c92d265e1242e61dd762b086f5":{"kind":"REWORK","_number":29,"created":"2023-03-15 11:00:45.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/29"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-03-15 11:00:37.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f2e5b7aaa1d418c92d265e1242e61dd762b086f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f2e5b7aaa1d418c92d265e1242e61dd762b086f5"}]},"branch":"refs/heads/master"},"75072bc1ca951354e4a8791c2f249131e2494725":{"kind":"REWORK","_number":30,"created":"2023-03-15 11:27:46.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/30"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-03-15 11:27:38.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/75072bc1ca951354e4a8791c2f249131e2494725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/75072bc1ca951354e4a8791c2f249131e2494725"}]},"branch":"refs/heads/master"},"43ab59d8b38152e255ebd0501a4d763eb815ad03":{"kind":"REWORK","_number":31,"created":"2023-03-15 13:49:33.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/860399/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/860399/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/860399/31"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-05 10:23:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-03-15 13:49:27.000000000","tz":60},"subject":"Implement acceptance test job","message":"Implement acceptance test job\n\nImplement acceptance tests. Those jobs will run in the post-review\npipeline requiring access to secrets containing credentials of friendly\npublic clouds to test sdk with them.\nBase job is generating a token from the given credentials and writes\nclouds.yaml file with the token inside instead of password. As a post\nstep the token is physically revoked. This is done to prevent potential\nleakage of real credentials from the test jobs/logs.\n\nSince devstack is not a real cloud we do not use zuul secrets.\n\nChange-Id: I95af9b81e6abd51af2a7dd91cae14b56926a869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/43ab59d8b38152e255ebd0501a4d763eb815ad03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/43ab59d8b38152e255ebd0501a4d763eb815ad03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
