)]}'
{"id":"openstack%2Fopenstacksdk~875530","triplet_id":"openstack%2Fopenstacksdk~master~I77ffa307d2797368591693cb296606045e949144","project":"openstack/openstacksdk","branch":"master","topic":"bug/1999774","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-24 11:18:48.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-03-21 10:44:16.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-03-24 11:18:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77ffa307d2797368591693cb296606045e949144","subject":"Add BGPVPN to SDK","status":"MERGED","created":"2023-02-27 15:54:17.000000000","updated":"2023-03-24 11:18:48.000000000","submitted":"2023-03-24 11:18:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875530-bug/1999774","meta_rev_id":"ee68cff2a36e01732135bc2d3203c1c47dd8af71","_number":875530,"virtual_id_number":875530,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-24 11:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-03-21 10:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-24 09:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-24 09:44:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 17:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 10:20:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-20 11:49:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-03-24 09:44:52.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"6e89235af16e540cd8a436ff1b87f191c9ffb45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-27 15:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3b75daccbdcf4c099dc9582a4e61162e0bebe70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 17:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa6b1d8ce68f46ce9cbd0910ae6867de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cfd066475b4b4f8a0e30a73acd7132 : FAILURE in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e2ad947b4c144ed8925d65b4fc99371 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7faa2a800a84f779b0323de213d520c : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c7572899388f420d8ce899dfe8718999 : SUCCESS in 57m 55s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/57a675cd8ede4c598c4a542003917689 : SUCCESS in 4m 36s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/cbc0811a32ec4135a146379179771491 : SUCCESS in 3m 23s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fd28bca745be490ea7c460bb569d6957 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c697531fa2a431ba59f033fa55298f5 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62d953be563b48608ae54c1ef3467291 : SUCCESS in 4m 38s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/eb0e60b2b65b41e6bf2412b2fb36fe41 : SUCCESS in 45m 27s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0223316021cb43e6bff781ed2d4bcdde : SUCCESS in 10m 46s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/494898a4b7ad4f34a2dbfe19e21ab5e3 : SUCCESS in 28m 49s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2141094b81ad4185be9bd8a26995372f : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/00feaeb7087948f9aaac45867778f86d : SUCCESS in 1h 29m 45s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/03d8812087f74bae93480291cc0e73c3 : SUCCESS in 55m 22s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ad35c433d97d475fa10e410f07d42719 : SUCCESS in 45m 44s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/183f29aa3b004695a7872ccc448a4120 : SUCCESS in 18m 52s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/679bb9077d7b4b2b82712aa5bed60ba8 : SUCCESS in 38m 56s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c5a6e2ddd1f84cc8a26e886839e6d138 : SUCCESS in 41m 45s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/4b046771770948a799921c228058ec89 : FAILURE in 33m 47s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/89c3721fa2e2474e86282d0386e0cceb : SUCCESS in 48m 53s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0dc8dfe87ee04eb2a3e7526ccdb0bc06 : SUCCESS in 46m 17s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a40597cec83549e9b65311b8212c3bd7 : SUCCESS in 40m 17s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3e77c9567b9743ecbcd92ecfea4e052f : SUCCESS in 1h 18m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fd906b2653c604db9349ee1ccd8ba958c53e68e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-27 19:47:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9eb2bb6f2641217d4d691f899f92044d02a1d31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 21:43:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c3b3f4af2a47d6b5094a27e320d85a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5699e80f0cb4134a6e9142c78c2336b : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10a46359275f44bfac64b1c59ec65cb8 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e04ad4410e842dc89e8a8b4ff1d328a : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1466f38ae23e461d9a98754464122550 : SUCCESS in 1h 00m 14s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/67f3411b9f8d4b6c9852f23e9143da00 : SUCCESS in 3m 46s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9b080df41d814de18d50c880366bcc1b : SUCCESS in 3m 03s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dfe135013f1244a28e77e68576479a95 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00be538eb2746828457410af49e6c0f : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed32f58faa854271b633f808f3a2cef5 : SUCCESS in 3m 12s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/dce7ecd1a6264de9bb725a51f771a83c : SUCCESS in 1h 50m 49s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/6535d5d3a13d4e41a440ff9cc28d89f9 : SUCCESS in 18m 40s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/18672da3d80148d6a34361a79500328f : SUCCESS in 1h 17m 39s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79e32bf66e8a4ff6be81e69e8200d18c : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7567fe9bcd164bd18b58b039b435066f : SUCCESS in 1h 28m 35s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/1098bc953fdc47e9b48b2511b6ba88b7 : SUCCESS in 45m 37s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a6a26abdd5fa458c8483a170d343e077 : SUCCESS in 1h 15m 02s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bc355cec49e6462584472fbb83544fe0 : SUCCESS in 29m 18s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c4b78699197847beab034ae016b7c941 : SUCCESS in 21m 17s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/62a9d6d474fc4a9c9fd204a82b831099 : SUCCESS in 40m 40s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/df22b27ff07b4127b18972f36b36e2e0 : FAILURE in 26m 52s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5d0aff40f2cd412f965644326f05f5c8 : SUCCESS in 1h 02m 24s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a792231e1b924e15be3b8849d9d3ab84 : SUCCESS in 1h 02m 58s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0550e527b5e644179cb52bd12191b8c5 : SUCCESS in 41m 32s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2378fead994b42039aa415c40184a1d6 : SUCCESS in 59m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"222f120b6582b55fbabd7367696a5d26e31fc002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-28 13:42:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"570e3bcc3a115cf2b4bf81112ca437d32601e9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:00:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1376761707f4b3b8e7cdd4c51990d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667bc9e6d02d4445a35cb071b05a0ee9 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09efdad1f4d1453f94cd265b27fa1f53 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08f859a2a10f40568f6b76361db7b53b : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5da1094dc9ec48c1896a3ae7e2a20be0 : SUCCESS in 56m 48s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/90026d89c63e48e98819611e4e199a21 : SUCCESS in 3m 57s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8380a08a035f4bb4a7f0071844c9c0fa : SUCCESS in 3m 36s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/725e717d6f334a1daa05b66f057ff7c9 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d8b87feb38499f895801f2774242af : FAILURE in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f552da4f28bd426f927e762e7d4bb48d : SUCCESS in 3m 40s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/778eea05120d42668e37d38e4bd8bafe : SUCCESS in 1h 07m 02s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/dbc6fdc026cd43fb8d2294b2d8fc19d5 : SUCCESS in 13m 19s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/820e826d7aff4e268a28879829787ef2 : SUCCESS in 39m 51s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6f6d623e0734018a4e520e64ec201f4 : SUCCESS in 59m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8dd11293782d43228253e25935cbdc50 : SUCCESS in 1h 27m 18s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/77b0bdfffb9541d8b88bd829f6eb6e62 : SUCCESS in 48m 46s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/735d099f42824a2e885a37ce5b56d1dd : SUCCESS in 39m 52s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/a88d39444ebc4f6386f87645cdc1c585 : SUCCESS in 20m 38s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e0ff1fa8cab94531a6ea8f874e2293ea : SUCCESS in 24m 17s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3c140e1f6b7e44ccb6eb518a2df8b3b6 : SUCCESS in 33m 22s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/f6456b28b4ee4b5a8c985c6bbbccb245 : FAILURE in 24m 04s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d01b83be4610481a9559a759d517df51 : SUCCESS in 1h 04m 02s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/96c804e1ed784193b265351aa3992095 : SUCCESS in 1h 05m 03s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/181d8e3e831d42a5b8cfbdaf33cddcd4 : SUCCESS in 36m 01s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a13cfef2eeac4964ab92e3e459bd44e6 : FAILURE in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"852c60495a046ec36e07652e08214f28ef27e60e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-08 13:59:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"047ca68a1f35bf2ba9f3a485852c4ccf3c7666a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 15:37:23.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88a1e0ee10454af5a9bec9b1af84e5c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00606d2da0644b0cb0d875573a5ffaca : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a81a4e8d40848c484a540b8f35343b3 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2539cdb31494940b8a440fb519c3686 : SUCCESS in 6m 30s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8fcc83ca427645848d1a04a27b04f6c7 : SUCCESS in 1h 10m 45s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5665becd37ab41de8c1ed9006b6b19c0 : SUCCESS in 7m 06s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/126f8adc8b56433fb93d88e933ff3b56 : SUCCESS in 3m 29s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b06fde8ce9c84e32bf34c80fca90eccb : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72312f013f514403a48f7d15655ab27e : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e827b80df94d42bf83aa8ea3ddc5d5 : SUCCESS in 3m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d5c3899f9ba7492e8aa9c7b7b8a2e159 : SUCCESS in 46m 52s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/558663bc192a487eb05606aab99acae9 : SUCCESS in 13m 25s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/98d4b0fbd81d4cf4bc1504ba14ba9c1a : SUCCESS in 27m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dd3d421a8844ec6b0bf429b520611dc : SUCCESS in 1h 09m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e51df6afb755480ea0d8eb07ff2ac1fb : SUCCESS in 1h 22m 01s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/10db79fcbb1e447592c5d5d2fce6433b : SUCCESS in 54m 17s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d8e42ca0a0164839b087dc6d2435e218 : SUCCESS in 51m 44s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7ca3816242354ed391f4f761d17195a1 : SUCCESS in 29m 32s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/50c4997d4f5746b39c4d08d34cef3a36 : SUCCESS in 32m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ef537090240444d5bc3c723589fef029 : SUCCESS in 39m 22s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/44acc7a433f84462b2f500a26f59e47d : FAILURE in 37m 16s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9441a2c2a43243e5870abb7220f28de1 : SUCCESS in 1h 01m 17s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/23524aed43464f8fa1cc753bc55f838b : SUCCESS in 55m 15s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/5606c1169cd64a9c88e9cbe5eb52e3d8 : SUCCESS in 39m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/e2913e784d4f46659bbef31117aeca87 : SUCCESS in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99870d5ba5319a3b88525bed326a4e4ec938405c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-08 17:49:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c6b19ceb9e6b16c36b199b2f27b88cf79e90ad73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 19:22:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cef2f6e653594dba9e3c8e600077c24c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d614b35fc26f465391b9a358a3490804 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ced9985b78fa43e9bba9662a34938b89 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f5af6a8b12b416790a04a4a68f27654 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/620fec46004d43d1b24012f3b69b07d3 : SUCCESS in 1h 04m 57s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ebf2d3efcece4f8c9750c30328bb6bac : SUCCESS in 3m 49s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2b88d593e36c4806bb2cd1f7ea99187d : SUCCESS in 3m 01s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6d62cdb829794a46845b598900a6f1a2 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3c696163f04e8ca1d53c3b0c7c6115 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0dd9d4573c480aaaaa697a0d109601 : SUCCESS in 3m 15s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8a9ede2a16654a369b88c27d85d49a0b : SUCCESS in 1h 08m 04s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ce375a1a2a154ef6a50c2d636504db34 : SUCCESS in 13m 41s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/2653f448924d4198bc1b0451fc2c13e1 : SUCCESS in 40m 00s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc1515b49fd24ace981c92851b62b369 : SUCCESS in 1h 03m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0320d9b32c4c40c993663f9dd8394b1f : SUCCESS in 1h 20m 15s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/760b74d81e104c6fad61420835e9bfae : SUCCESS in 58m 44s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/8b0e9ff5e4af4395ad882eef31170f92 : SUCCESS in 49m 14s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/689ed73298624e84914950d6bdb756e7 : SUCCESS in 35m 05s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1bede0425f4a460aa6a1cffa7f7fd4a6 : SUCCESS in 22m 11s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/02516020e236482abd0947d23cac0129 : FAILURE in 31m 21s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/133b2669d92245619582d40845290f3f : FAILURE in 40m 15s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c3ee9090c5c847d59655fb641a28a1a7 : SUCCESS in 1h 15m 48s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a4b1ecd47d5b4ad18f458fe614d4901c : SUCCESS in 47m 45s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b31a7faa37774434ae09c01d146c745f : SUCCESS in 33m 25s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6fced3e34b614676aef3125379a56167 : SUCCESS in 1h 00m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"47d4e4f5ab165426fbfaf3f2805e9a14a982e9ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-10 07:37:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"203ffb485f25e4bb6f2c469c78ae0c88d203ac10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 09:10:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e3d2a802c5847ba8736dc832cffd150\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d984bf15be724a69a3d614a641b4638c : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19631fbf2974db5aefa35836b56d9f7 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a93d1693d88e493ab1d683208f4307ab : SUCCESS in 3m 10s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6c536d69d4a34716805e7c5021cb959a : SUCCESS in 41m 22s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fbadd6d9aea242798e3daf9fee3fbaa2 : SUCCESS in 3m 49s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/07d4d23eb1ca46b6a632e2344f6fc9f4 : SUCCESS in 4m 10s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dc758ab2100f4a03a20e18c13033b58b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b98aa7508b42a789b035301da368b4 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/525140efe1774d3caef171803b40d4f7 : SUCCESS in 3m 30s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6b96048caa3b4de596bb9adc224f7d6b : SUCCESS in 1h 00m 00s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/4129572df9694a98a81a592101c8c303 : SUCCESS in 13m 38s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ce482db4923d4e0a88b1fe307c8dc571 : SUCCESS in 32m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2750e83b92b249b88174151c8d10faa9 : SUCCESS in 39m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/37234ba6f0b84e47aa65cb16fc931297 : SUCCESS in 1h 20m 31s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/dbd74f996e2d483e93a414462f8bbc49 : SUCCESS in 44m 07s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c534381087c94e109bb3a95c2f89890e : SUCCESS in 42m 31s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e3e6d751b91b4b7a97a6cd07f63a6398 : SUCCESS in 23m 35s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9d852e2b48e544518fcf49d1541a783a : SUCCESS in 31m 47s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3c9623e4fdf541dca6a284100643bef3 : SUCCESS in 40m 45s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/57db727e769d4992aabd8fc0c0d11403 : FAILURE in 25m 48s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/14f686a9310342b695ccfa7c57fb765c : SUCCESS in 49m 42s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/32ef77ea90db421b9f0024534927bd18 : SUCCESS in 58m 04s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1b46a5ac9e8e4bc8a2c8cf41be74a533 : SUCCESS in 33m 11s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3cf0e0e75c2b4fdf8fdacc9bffe20123 : SUCCESS in 1h 06m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"212824c4a1314bcff9e6941b9ab56f308f9a5e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-10 10:15:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1fdb2d6d658e85caf4fe380ab730c73bc004f19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:42:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b7b8f12f12d410d8014765cefdf5323\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc95d85ac30431ab76feb4e672f1e4d : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d61bd01e338476ebc2bbac2f5e9d1dd : FAILURE in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09a4208210f24315adc7a360158967f8 : FAILURE in 3m 28s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/04fc4437a88d44ddbef61d4347b162c1 : SUCCESS in 1h 00m 51s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3c0a194f5eed4fbfa840c358f531f3f1 : FAILURE in 4m 03s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/58dd183e9e3b42fd925811284a8b61c9 : SUCCESS in 3m 09s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b06e47ae0f4043569bf81738858526df : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32bafced310a4a8a85a3fb8bc3ef6dfb : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dbe9e4ec6664f058732b43a8fa21262 : SUCCESS in 3m 09s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/98e0bb8408cf4a59a36ece5e7b12af13 : SUCCESS in 1h 00m 23s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ac473811743f41a5bd45ab63774caa47 : SUCCESS in 16m 56s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/11c2b16a4ff3470aba30c673e562e3d9 : SUCCESS in 37m 24s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6329809c677c48419dd63ea07c9c48b9 : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/40ba25a77c35490eb3d3c37209e0d31a : SUCCESS in 1h 23m 07s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/6ac161ff0c264b60bae6b1c4a5f8e6b0 : SUCCESS in 51m 57s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e1da788b0c044e05a70832f4817189ec : SUCCESS in 52m 07s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b698ebfc9f3e4914bae6e0f40dfea40b : SUCCESS in 28m 48s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1825c2e40ebd4daebcddbb75c343f544 : SUCCESS in 32m 25s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3e9e2fd53c524ba2a223c721c9db913c : SUCCESS in 38m 48s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/3f5198eefb1143e18a61c6b24a90ebda : FAILURE in 36m 47s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1bc18460e3564be0956eea79bac156d0 : SUCCESS in 59m 31s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/abaac1de3652421985d35b9507453a73 : SUCCESS in 59m 37s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9f1bdc877a3545ad9aa2597030637541 : SUCCESS in 38m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/68fa7763ff3840ce895498937e3becd2 : SUCCESS in 1h 07m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6122d56e8cc150fcb5de7c336b95a8168f26f338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-13 13:22:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6a080c51899631b98e55eba4775ba687536ac270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 15:08:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d345b1ee502d42b19bed551a3bab16d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d8b6f6e18a42f08c73f7d894dc5c95 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9736d6d530c94ad5b787502ba6f42019 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f243a8436c1b430aad02b4f9b09ef949 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5cfe6736b8b643fdbfb5ca9ecf8ed340 : SUCCESS in 1h 00m 53s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/90fc41b5e6204e298c37212b6dcc8d6f : SUCCESS in 3m 55s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e21b8120eb2e4624a9e75aa938c1698a : SUCCESS in 3m 11s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/258655ad9d494247bb99225ff3d79655 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9ae015baad4d8b9ad7789a1416d198 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8062accc91e44fabb247b67577d7f668 : SUCCESS in 3m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3d65e20e77ca4ee48e33c94e8d60d656 : SUCCESS in 1h 02m 30s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/97a994758b0f4562b30676ca12c5b1b5 : SUCCESS in 15m 00s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/fb24b894030c47469eca832f68ff4b2e : SUCCESS in 34m 23s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa128677a4d446718b54fac02ad8dec6 : SUCCESS in 56m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/259a950d2d3d48e5a785a97682514814 : SUCCESS in 1h 36m 52s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/3f01737935e0492e969a4d4823d3d1b7 : SUCCESS in 59m 11s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5ffbb229eb134eb8b40e575e4bc0d9c9 : SUCCESS in 54m 11s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/0f4722e35a2a4621b4a64965769d3a48 : SUCCESS in 35m 23s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/14edd4412adc49bd8e269d3528b4b738 : SUCCESS in 32m 54s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0631ef95520243988b23ee809c58631d : SUCCESS in 31m 28s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/612912ff42834f3a95de4393087dfbdb : FAILURE in 37m 15s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/24b5715b550d4ebb9476dfc5070162e3 : SUCCESS in 1h 07m 44s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ca200d63bca54b8e883119c596e7b07a : SUCCESS in 55m 14s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ac5c749b60d44692b32139cbeab4dd9d : SUCCESS in 39m 28s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6814707643464a9b9b3e042422c605af : SUCCESS in 1h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"23b7fc6e7f2fe07cfc06a4966905f40bd831fc00","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-20 11:49:50.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c1bfabffbf2174959520ac2f964ab2c2469b314a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-20 13:45:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"15665b8d45d57c870c02c988a269e7874da4c010","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-20 13:46:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"df7cdb413b6c2561bd11ba515cabeac205aa4ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 15:14:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2004d24c294d40aab489fa6440dd524c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc04b6a37c44025bd2868a36313e562 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c90d812be284980a2f60424974ee5d4 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ab2ffae2bd7415fa778c3eefaefaacb : SUCCESS in 3m 10s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9afd3b5fe4914a189fbaf50100d96c76 : SUCCESS in 41m 00s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/4e4da7aa47234d27b069a06e6874cef1 : SUCCESS in 3m 09s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d1806737ec4748a3a0b5afd829317ce4 : SUCCESS in 3m 03s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/bbb75d419f894dab9e1595a6b6f3e097 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815a48dc110943ee958353e2a7c44545 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e6655aed48450ca308ab50e8df8851 : SUCCESS in 3m 44s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1098efca7b4743298ea6858528f6dcd5 : SUCCESS in 1h 21m 27s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/449a031cd260471cb8ec80ca9bf59db4 : SUCCESS in 15m 48s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ce208b5678904b8f837503d6aa4a498e : SUCCESS in 52m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea138b3ac6454fd9973e1730591b5a0c : SUCCESS in 1h 06m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dd9c0f39350e4fceafa889fd6270cdc8 : SUCCESS in 1h 03m 43s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/9466b4f801c74b6c8a3f06bbc8454e11 : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9e42d1d5cfb645daad89f803de2668d6 : SUCCESS in 56m 51s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/8c18b83ff28646b38fb412ce2c000a99 : SUCCESS in 17m 57s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6a9747e9fb834c6f8c9c33ae219db549 : SUCCESS in 31m 49s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/eb88b95a6bd54918a1a3f917f43870b1 : SUCCESS in 42m 05s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/df853333563b4177a161daf157a48cab : FAILURE in 23m 26s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3dda362c4ef3458da0934a57bc80b581 : SUCCESS in 1h 19m 28s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/5a83ae5504eb44eab42c93b15e38c689 : SUCCESS in 40m 12s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/35d387ee790d45f89132d4620550f524 : SUCCESS in 41m 10s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5b8e26cbb37849adab6a12cfa866f016 : SUCCESS in 1h 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6a8fb1c184bd8802c0fb4afe0d078b19c4bd5a2b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-21 10:44:16.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"87cbfffff3f5f776750c39afceb29b33b789b9b9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-03-24 09:44:52.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"be7562859974321896214122935dbee0474639e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 09:45:36.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5cd36fac166a1adfa899d8ff980cbd850d0a2914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 11:18:48.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5daa802e0371448ca097365e7e9df015\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe817dd329e4538821796c3820a1240 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7abe7b5c491640ddad3c3cd057ba64d0 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6960fdcb2b314b2f9392b731f18093e6 : SUCCESS in 3m 28s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/984f84b9c3bb49939d89f7bff151fa8e : SUCCESS in 1h 01m 53s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f26d1a6b31b64f35b98a2aa0766d4151 : SUCCESS in 3m 38s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3567068c2ede40289b1254e646fef221 : SUCCESS in 3m 13s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b67b4705543a41349ea054971f539bd9 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5014a0eea8b4199b716889aab8f9882 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae5086c282d14703847ed6d4bdc9eb12 : SUCCESS in 3m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4b50b3d4fd014bbeabc5cd1ec9e6182b : SUCCESS in 58m 56s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b02c2d981d234db3aff4da732f83d8d0 : SUCCESS in 12m 10s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/4b33916377394e0791957b095b2d1680 : SUCCESS in 41m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea07d66422114bd58ebb1be7e7875038 : SUCCESS in 1h 08m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5466416757149bb84b670116e11328f : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/94011657231d4e2ca9b5ddb21631886c : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":9},{"id":"ee68cff2a36e01732135bc2d3203c1c47dd8af71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 11:18:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1b241d74c021d2efd80cd4f4bb4b72d51e9cdd01","revisions":{"75c275b252d16772166446a84ff2734b6d2a2b4e":{"kind":"REWORK","_number":1,"created":"2023-02-27 15:54:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/1"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 15:47:19.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nTODO:\n* unit tests\n* functional tests\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/75c275b252d16772166446a84ff2734b6d2a2b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/75c275b252d16772166446a84ff2734b6d2a2b4e"}]},"branch":"refs/heads/master"},"c3d3aeec47c9ffe2a57edf47b30b8cf976b47311":{"kind":"REWORK","_number":2,"created":"2023-02-27 19:47:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/2"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 19:46:54.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nTODO:\n* unit tests\n* functional tests\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c3d3aeec47c9ffe2a57edf47b30b8cf976b47311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c3d3aeec47c9ffe2a57edf47b30b8cf976b47311"}]},"branch":"refs/heads/master"},"4b905c7013cb500ed50c3c7c4bf6dec10e5c5729":{"kind":"REWORK","_number":3,"created":"2023-02-28 13:42:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/3"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-28 12:26:09.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nTODO:\n* unit tests\n* functional tests\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4b905c7013cb500ed50c3c7c4bf6dec10e5c5729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4b905c7013cb500ed50c3c7c4bf6dec10e5c5729"}]},"branch":"refs/heads/master"},"8b9a0a8868702e5eadf049afa25021e44f381c12":{"kind":"REWORK","_number":4,"created":"2023-03-08 13:59:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/4"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-08 13:58:51.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b9a0a8868702e5eadf049afa25021e44f381c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b9a0a8868702e5eadf049afa25021e44f381c12"}]},"branch":"refs/heads/master"},"9da95ba5f62ad38a07c1d11efbb24817f774ccdd":{"kind":"REWORK","_number":5,"created":"2023-03-08 17:49:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/5"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-08 17:49:37.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9da95ba5f62ad38a07c1d11efbb24817f774ccdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9da95ba5f62ad38a07c1d11efbb24817f774ccdd"}]},"branch":"refs/heads/master"},"f8e79407ef3094a6eb4978f2aec45e47634d2b57":{"kind":"REWORK","_number":6,"created":"2023-03-10 07:37:18.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/6"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-10 07:37:06.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f8e79407ef3094a6eb4978f2aec45e47634d2b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f8e79407ef3094a6eb4978f2aec45e47634d2b57"}]},"branch":"refs/heads/master"},"081f17a9a4e7230707533f53ab4d9c55237ef456":{"kind":"REWORK","_number":7,"created":"2023-03-10 10:15:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/7"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-10 10:14:18.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/081f17a9a4e7230707533f53ab4d9c55237ef456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/081f17a9a4e7230707533f53ab4d9c55237ef456"}]},"branch":"refs/heads/master"},"d78f5a6a20ed032f79e185814a9010e5d2bf0c8b":{"kind":"REWORK","_number":8,"created":"2023-03-13 13:22:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/8"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-13 13:22:19.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d78f5a6a20ed032f79e185814a9010e5d2bf0c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d78f5a6a20ed032f79e185814a9010e5d2bf0c8b"}]},"branch":"refs/heads/master"},"1b241d74c021d2efd80cd4f4bb4b72d51e9cdd01":{"kind":"REWORK","_number":9,"created":"2023-03-20 13:45:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/875530/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/30/875530/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/30/875530/9"}}},"commit":{"parents":[{"commit":"74f8869fd9ca436071d2227a8a668463e1f414e6","subject":"Remove \"feature/r1\" override from manila job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/74f8869fd9ca436071d2227a8a668463e1f414e6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-27 09:00:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-20 13:44:57.000000000","tz":60},"subject":"Add BGPVPN to SDK","message":"Add BGPVPN to SDK\n\nIncluded resources are:\n* BgpVpn\n* BgpVpnNetworkAssociation\n* BgpVpnPortAssociation\n* BgpVpnRouterAssociation\n\nChange-Id: I77ffa307d2797368591693cb296606045e949144\nRelated-Bug: #1999774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1b241d74c021d2efd80cd4f4bb4b72d51e9cdd01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1b241d74c021d2efd80cd4f4bb4b72d51e9cdd01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
