)]}'
{"id":"openstack%2Fopenstacksdk~900713","triplet_id":"openstack%2Fopenstacksdk~master~Id238b870de85a4663032710d1c15f6e200d2d543","project":"openstack/openstacksdk","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-10 10:15:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-01-10 13:07:44.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-10 13:07:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id238b870de85a4663032710d1c15f6e200d2d543","subject":"mypy: Address issues with openstack.network","status":"MERGED","created":"2023-11-11 18:11:30.000000000","updated":"2024-01-10 13:08:54.000000000","submitted":"2024-01-10 13:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"900713-mypy","meta_rev_id":"3c4e597fbb146145d67b93ac70c7f22307a0c84c","_number":900713,"virtual_id_number":900713,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-10 13:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-01-09 11:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-01-09 11:58:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-11 19:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 12:59:18.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-01-09 09:30:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"19c261e500feb710ffd9b3c9fa71fbeb36eccba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-11 18:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c31df21a54f692416161db4fa437304837ac376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 19:53:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bfd14f3225a4ed3ad1a7b683005df46\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c5e4ad5feb845c9b915ce4d20ce2e0c : SUCCESS in 2m 12s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/60f2273fcb5044e59c7f3072bfb7b222 : SUCCESS in 47m 25s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/914de00d43eb4237bd4cbc2d6748aaad : SUCCESS in 33m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae205e99c064ee9a5c24318dd5629e7 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcedf11f44954279a898fa88c7f523b9 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe4f831d2b4c422184141dbfe2f70b23 : FAILURE in 6m 22s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7ac5bd9bd70a4272b6f721f05c918f0f : SUCCESS in 1h 08m 32s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/eacef505af80432e84c72c8638fe3c47 : SUCCESS in 4m 45s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/42603714941244cd94db29a8a79aa088 : SUCCESS in 4m 41s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/47eec63bb0bf4727a85d2cdbf438fff0 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e464243c654c4b9d4a0cdfc6bd2ea6 : SUCCESS in 8m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d6a67410b31d420daff1e1bfe2b54938 : SUCCESS in 56m 42s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/de29d433e07a41e28d703e1e1cacea92 : SUCCESS in 16m 35s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/dd5fdf6668e545da803ce1db9cd54eca : SUCCESS in 31m 26s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dca0bc6e4be34fef8976749726655fbb : SUCCESS in 56m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7fbfd6f41c1e457d83ceab9ea87e6edf : SUCCESS in 1h 28m 13s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/63c084c349a04159b0cb3880e99b1bb3 : SUCCESS in 1h 31m 42s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/94413c1119c14d6ea4b2a4a63239fe6c : SUCCESS in 55m 57s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cd04b0407ae5476486710b4f2b0ad67e : SUCCESS in 48m 14s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/6b512da39bee44218d3661952a187c31 : SUCCESS in 28m 05s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c00a2b568e0b4e43a054e0ad2c14c635 : POST_FAILURE in 30m 11s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7689daeea2b24555bbec53485ab860c8 : SUCCESS in 43m 52s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b40ba859727a425d9d4ed9c935dc0258 : SUCCESS in 56m 52s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a6b84d6aae464ee596e91f64aa4f7d22 : SUCCESS in 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fae84f86915a57566289d7b5f167dfcf53198aba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-13 17:04:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27f4e8619c5431b4990105ea63289f7cddcea8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 18:43:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18178f92be3643e38ca04d2fa811d48c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6023ed4d5773458fb95d8154bbde05ec : SUCCESS in 2m 12s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6c2e0b994a9545229101fd7fdb46ecda : SUCCESS in 47m 39s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/df8b6a00d9334b218d48eeae854b6387 : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0dd425202d94633a06f5f072fc01686 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b05a4f8a791449c5ae0d68459c183504 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac059ee228d24e308b50ac6623f35860 : SUCCESS in 3m 59s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d45ed964a6414fbda92add68d122ed0f : SUCCESS in 1h 07m 27s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d30829e67db24619803e2cf093445b28 : SUCCESS in 3m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/eab125badbc44b379a6c0b79579e6e0a : SUCCESS in 4m 28s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/c5d17c078532488ab2f98207843e1e22 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2cfd9af96b43afac1029729fd0ff9c : SUCCESS in 8m 52s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/58b36546f93a421babf19908053f0673 : SUCCESS in 1h 09m 02s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0fdce086a9a24844ae33247eb6e76417 : SUCCESS in 13m 20s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b5243a547c344a3fade5970ab82d3d93 : SUCCESS in 41m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d4958551e80414abe20a2ec8a8adf8e : SUCCESS in 1h 04m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9ece7e0bf9eb4b8eb80f945c940982df : SUCCESS in 1h 29m 43s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a5d2eb8e3a7e4bd8b1969a35b02d3d79 : SUCCESS in 1h 28m 56s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/e328d5602aca447f9f2179b6f49701d7 : SUCCESS in 59m 18s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/02bb332e72da4cf3990406625d4651b5 : SUCCESS in 59m 31s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/3789e29654bc488ca1278f05dccd40bb : SUCCESS in 40m 08s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/96cdfb1691064c19a019c47000f3be68 : POST_FAILURE in 33m 44s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/31bec84949164895b3429d35d93941e9 : SUCCESS in 46m 32s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fd019779eecf46a6be939216d67f29a4 : SUCCESS in 1h 08m 53s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/dbe9b4d983ee47a1b27f7d587a47ad24 : SUCCESS in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f32c2b8e7297a22ac29f3e0bc13d6bbaa19d6ab6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-12-13 12:59:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f8249662ddec45ceaf605cf34fec59e8b6915dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 13:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db26ae842d070c3391afdd3b2e7e16679a226a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 14:34:44.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/900712, https://review.opendev.org/900711, https://review.opendev.org/900710, https://review.opendev.org/900709, https://review.opendev.org/900708 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"640b6b24d09a083971a9706c21952ff1ce0f87b7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-08 16:14:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71b370dd4c7edd8d3d3ecc9d9262d4067bfa806c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 18:05:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ba31073bfbe4b219adbb1e8e1b9f2c2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a90d9827deaf46a9a3a810ff00865f55 : SUCCESS in 1m 58s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/420bc58ba4d34cd1bbb348ba1a4302a2 : SUCCESS in 1h 04m 03s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/99ac66df67934baebeb123dbbbeb7cdc : FAILURE in 2m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232fe75034104727b52e384f65e4b559 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231c508e3cee479da4f1c5d11025692c : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bc9eb83acfa4c44b64c3dddc86c0847 : SUCCESS in 4m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0e5fa156d3ce4636be78844f21fa6c23 : SUCCESS in 1h 11m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c27a215aa56f4d0dadcff7a01440a8f0 : SUCCESS in 3m 56s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/51f73546c3984d59bcb45fe45e06c285 : SUCCESS in 4m 36s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1ffc9cb80a7c487eb5617be3599827a2 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7891a7c805e4633a5438c01fbd64616 : SUCCESS in 9m 27s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/51af9cbc15b34f0e9e207dedf6b6e81c : SUCCESS in 1h 06m 27s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/05cf5d44cc3b4ddfbda02a68dba55b8c : SUCCESS in 17m 12s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/759aeabe677b44399f96431de07c481a : SUCCESS in 39m 37s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31763b5e8f84491b8eccc59c9d6c6032 : SUCCESS in 53m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fd7a2ab92ebe4513914ba13bcadbf27f : SUCCESS in 1h 18m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f3b968f3ee584135aa6fc6121ce70ee9 : FAILURE in 1h 43m 18s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/398bd5b0362c492d805506adb2825014 : SUCCESS in 1h 01m 18s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cad0f76c1ca648a08fb31a4facf7293a : SUCCESS in 1h 11m 21s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1bd5ae93a31548898c84eeea0c053384 : SUCCESS in 40m 47s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6394b8afc7e740b3b8039fc1b3aa93bd : SUCCESS in 46m 20s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f26abd134648410d9d14180864c0a3dd : SUCCESS in 31m 56s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/930a71d61a6245f99dc882c3f9b16aa0 : SUCCESS in 1h 08m 06s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ddd13d0a7164476084f26777723e2450 : FAILURE in 1h 07m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2646cdec1156548b39e57c68011cad3453a57e1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-08 18:17:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"192c38fc2b7fc8004e98c7a8604e6c6b6397d774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 20:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa76fa4347149aebcc250fd4c4afef8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5787024840c4b918cc8fe7d73729f5f : SUCCESS in 2m 17s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7534d9483415461d9e322908f5e196c0 : SUCCESS in 53m 37s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ff01917774b247ef89d6b57b3da62a41 : FAILURE in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7f5f5a22654333af5851c309984bd6 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8251f7855a540bb99ea67b8c5288fb3 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5467b30a10aa47ffa78bb675f8fe0c0d : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f0e70d45bd964716b28823d9e6f07bba : SUCCESS in 56m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2be08a5cff7c41a9955d91386387b716 : SUCCESS in 6m 08s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/58adcdbab8234d86b82e28a27568d8dd : SUCCESS in 4m 44s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/4257b12d613c4cde9a4e53ef55f3c724 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6fef649cf84c56a4f74f998f950a93 : SUCCESS in 9m 59s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9f947140ea714d998f605dbf8d0b1f9a : SUCCESS in 59m 25s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/01765a84e96a41af8c209445e8dae75e : SUCCESS in 16m 35s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/dd703ced601a4d9497743d70cfdd3183 : SUCCESS in 34m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/245ffa2d65ae4cbab356bc2aa63b879f : SUCCESS in 58m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e70135e817064159bebea5c397c80fb3 : SUCCESS in 1h 14m 03s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4b3b2f367b9344b79bfdb81c8444a454 : SUCCESS in 1h 35m 22s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/ade294ac9369406891a6f7b6fce5fd3b : SUCCESS in 1h 00m 43s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/58f685025c834b16a8c790a11f8916fa : SUCCESS in 54m 34s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/3697461f4db94e639f28f4732084a2f3 : SUCCESS in 23m 58s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d473c51f19e740a3a08b289dd56d122c : SUCCESS in 34m 43s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/61afc14f63dc44ce8996c4fd9382d118 : SUCCESS in 34m 20s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7b310c5caf0848b7a4db04c65c396afd : SUCCESS in 1h 09m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d6dd3e8807674e13a5c5fc635cdf0f26 : FAILURE in 55m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46e67dcd20231ab9199ee2a96cd54e21dcc6e82d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-09 09:30:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b1d7266883694e93ec590434440f010028e609f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-09 09:34:22.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96f6b648d73d0ea904ccbfbaecdc3c074cea7b71","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-09 11:15:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28c578232d73d233cb82b602e03b6c49598f8fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-09 11:42:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8cc7b5ea44db2c49e8c5928b0d9e634419c7c9d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-09 11:58:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8afa87c7dbba2b96f42d596dcc917305348065b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 13:02:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c08f90a9abd45a1872f7e03fc7ccd85\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f9ae401c3254815ac4c66d7365f42ac : SUCCESS in 2m 11s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ef1817f5c6224b369dc09145f82c82b8 : SUCCESS in 46m 28s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9d9f2ae249f1414e9cac4c942ca2a0bc : SUCCESS in 38m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/602df227ed664135befc920b641ff1f7 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/905c1a744fb14947b5ddc772e0c40854 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e33ab4b12b58473a84fd5e83b8768b39 : SUCCESS in 4m 07s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2e735af5e08b4c92b78a9ad90fbea058 : SUCCESS in 40m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/09f73e07d4be49aab106e0bdfa61857c : SUCCESS in 5m 05s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b69c226792ec477091c74cf890209659 : SUCCESS in 4m 38s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1a2f4c11bfb64968ad5f3333f8d76415 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96476534d1ca475bb61edf4ad624171b : SUCCESS in 9m 44s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/aac62693c6b144ed91686b4ec28a1b5b : SUCCESS in 50m 06s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a0d4f146b3c040e6bd9ad43a346c63ac : SUCCESS in 14m 31s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ace952bbae344ec6922165f9ad0d1d4b : SUCCESS in 29m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c3214412a594196bb4a5ab75784b2e7 : SUCCESS in 29m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/71c5862b600247e399da59b18078c552 : SUCCESS in 37m 16s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/58fb1e205b004a919d935ef07265da91 : SUCCESS in 44m 07s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/92efd7a4090f49fe98a34c32807451a4 : SUCCESS in 48m 35s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/8e84905852e14f08be722cfd9e622d2c : SUCCESS in 40m 19s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/15ad87cfd9a042c39a6c6b1b2fe59f5c : SUCCESS in 34m 20s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/34b89a3cbcf34469b0be53012ff19fef : SUCCESS in 23m 52s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/99d0f52b970748ad9ee88623589ec9eb : SUCCESS in 42m 47s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8e63ab10f77c43b6be5f04c95c64e43c : SUCCESS in 1h 07m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5b05362a27e64cc8bd3a33aabc90dd85 : FAILURE in 1h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"766b511a0d3c07d2c11c6838cf9d38f3e4d5c64c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 13:34:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a95005c63934b6c40af3aa545cf45e15568c8b5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:40:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/900712, https://review.opendev.org/900711, https://review.opendev.org/900710 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"5bc2b4d67ee3e929f7a90bc74b89ea3232cc46c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-09 23:02:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"701de977fbae2e101b0fe7a107f7624cea2b3694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 00:21:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cbdfd01a3bd4427a5ca13da29ba6610\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/91e5eaebbd104c21b0949fe35fea5620 : SUCCESS in 1m 58s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/59e0c431835745a6a24348204f094ceb : SUCCESS in 51m 45s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3c0ec53805714d218ad553f7146d2b93 : SUCCESS in 44m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6125e4452aee467eb3c89bdc8a302e31 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/169fcf17e08b47959e0b343fa646e148 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a837cae2e2614571a571e967fedc1497 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e142c411fd114acfbc5f168e3ca84638 : SUCCESS in 22m 13s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5a62783d2ac64d319cb4e257c5c2e980 : SUCCESS in 4m 02s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f62c15a7d56d40e6aa52b5583675e8f6 : SUCCESS in 4m 38s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/de0ef46e8d3c4e01a6e8fbf1d4868283 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f09cca95e44d9985bb1f4114d12e09 : SUCCESS in 7m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4f9bf4687a7a46b69a89f0179b6757c7 : SUCCESS in 57m 37s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ddbbd1251be845688de1b87979447d01 : SUCCESS in 14m 33s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ffb0a4b685d5448db8fde8fb8f97a6fc : SUCCESS in 35m 24s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99b119bb05f440a093c1168df902e8af : SUCCESS in 39m 19s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c56a1a58f7b74c02a98e9d8043d2a5f1 : SUCCESS in 38m 42s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a91ba30354d54d388c07ee31687c7b42 : SUCCESS in 37m 23s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/0f49947da72d4ce3be87d7b468d17b79 : SUCCESS in 24m 49s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4c6a4f6ef5974440a247b47709cb5cb1 : SUCCESS in 39m 51s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/79695389b65f4c7bbcf15a361ca3024e : SUCCESS in 30m 51s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e97ef95fadcb478fbb2496b8e8b010d4 : SUCCESS in 32m 24s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/987550f2529f40188f83caa6b2ed5cf1 : SUCCESS in 31m 52s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/973d4ca6f5b7480c9b83e4b58908b810 : SUCCESS in 57m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/75bd960e69944a1b9f72488d4ffe8013 : FAILURE in 1h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9feb3886b9def79998cacec03ad526cf02bb1ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 00:22:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fba37d2c29791bd616abd5a52b2621acc8028eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 01:22:51.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/900712 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ffeb583909c24b953f4acf4c9631fa83df6b9603","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-10 10:15:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76e4dd5beaeaf5bd17fc8a37e97c7ae0564cf1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 11:42:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6792d9386a1c4fe3a1673eec722fa233\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba71623336ee4a7f97f3c8ace242a3dd : SUCCESS in 2m 14s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/238cd6c4f5d74198a45aa5c851e3c27d : SUCCESS in 1h 02m 26s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/57c0fae240b341d59f7802c27c6a903e : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a1e39efe7440c693abc75193ca49f9 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e42cfa340e77452a99f26a9487a74121 : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26e2b7e8a1ca4ff6aa8437655150b6ad : SUCCESS in 4m 21s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dd651aab9dfb4a5da3dcceb0dd59d263 : SUCCESS in 25m 06s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6bfa6bcfeaa14aaabd3737897722a82d : SUCCESS in 4m 12s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0549369d7a434d198d65a4c4ecbdbd12 : SUCCESS in 4m 13s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/25b564aed5f6433698c74b0dc2e3fe14 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dea7b3d6cae4d7e8f2c215e6942b30a : SUCCESS in 8m 15s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9cd19e1e8216416b85adf45edd7a2ff6 : SUCCESS in 55m 40s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/5ae4b92b959140928448a86b3f9529a4 : SUCCESS in 11m 31s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/87942017cfa24fad8dad5d59729e3400 : SUCCESS in 32m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f27411ddd4d04901aa86f59a260d10cf : SUCCESS in 30m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/51b8ca795ffe4e7baee8125f9ae67b4e : SUCCESS in 35m 18s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1aa1c4a53efd4dd9996a24b016033bfa : SUCCESS in 39m 24s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/eed0abfd7b5141b1b31fbd0446f20309 : SUCCESS in 35m 21s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0ba5437eddfa416592bb6a6ec22995d4 : SUCCESS in 36m 04s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/68ccf63706b34dbc9902f447b38348ba : SUCCESS in 23m 47s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9a36ffb5c3b7460097d8cdb139805a92 : SUCCESS in 37m 35s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/956d65c9b142483d88bdcef2d9abcf62 : SUCCESS in 38m 21s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8d09480b493d41f09a735752053d246c : SUCCESS in 1h 08m 02s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a8c0f7ed35a14fd8931a5487b4bc9194 : FAILURE in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a6fcabdd9b251e93cf13fc57e134638e7f27a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 12:07:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3021b0df19aed19b16751eb9898060d23d6ad2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 13:07:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c0fcc4b3ff49a39212525a5b02abdc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d82b77fb81b64b86a3a0188bac56128c : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8464b03a07fa468b9cfff8a49c1ba47a : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/647c14c1114c41d691edd5c72b2c65cd : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76b31987172e43aaab960ecfe85dedd2 : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f60d8e63e91b45d28228c0070e7212e9 : SUCCESS in 24m 21s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3e834f7aeba44d70b24a2bf417c8342c : SUCCESS in 3m 46s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ced3a6bab10a41498de036a572606ca2 : SUCCESS in 4m 45s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9634208be12b434992b075d842465b67 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c735432c4b434fc8854048cddf3b84b5 : SUCCESS in 9m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/343433848b6640d19a47668615a7cc10 : SUCCESS in 58m 42s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b94e3de41aac47f6957cd993d3b20b8a : SUCCESS in 16m 04s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/952ec1c690024f419eb3abcc9b7a9c8d : SUCCESS in 36m 47s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f816ada81158454c80683407cc7a4f34 : SUCCESS in 28m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8bfb2043a4544dd6b4d6463bb2d72cac : SUCCESS in 29m 10s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8a5e947a980b4b919aa0d68ead8e36bf : SUCCESS in 39m 01s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/b76b1a22f87d4a1baee5a0876bb58c3e : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"234df824f638cdf59720b446654288214d406825","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 13:07:44.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"3c4e597fbb146145d67b93ac70c7f22307a0c84c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 13:08:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b4dc4463094a068a19c559a110aec2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12a44590b99409aaae0320d735c7208 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f8d4102f5e73963d2decc3da12e5505a8e08f5b","revisions":{"ada2dbc9dfb4db9dece225b9ca3fe947aae38b2a":{"kind":"REWORK","_number":1,"created":"2023-11-11 18:11:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/900713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/13/900713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/1"}}},"commit":{"parents":[{"commit":"c57bea11d2b7af5e88a847d174ca352222df7bba","subject":"identity: Add test for \u0027role_assignments_filter\u0027 proxy method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c57bea11d2b7af5e88a847d174ca352222df7bba"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-26 16:32:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-11 17:33:56.000000000","tz":0},"subject":"mypy: Address issues with openstack.network","message":"mypy: Address issues with openstack.network\n\nJust an incorrect type annotation to correct here. We need to figure out\na way to properly type the \u0027Resource\u0027 class with all its magic.\n\nChange-Id: Id238b870de85a4663032710d1c15f6e200d2d543\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ada2dbc9dfb4db9dece225b9ca3fe947aae38b2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ada2dbc9dfb4db9dece225b9ca3fe947aae38b2a"}]},"branch":"refs/heads/master"},"4f8d4102f5e73963d2decc3da12e5505a8e08f5b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-09 11:42:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/900713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/13/900713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/13/900713/2"}}},"commit":{"parents":[{"commit":"f8a0ec9a637a452e4918539551992e3f77b32e3d","subject":"identity: Add test for \u0027role_assignments_filter\u0027 proxy method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f8a0ec9a637a452e4918539551992e3f77b32e3d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-26 16:32:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-01-09 10:54:07.000000000","tz":0},"subject":"mypy: Address issues with openstack.network","message":"mypy: Address issues with openstack.network\n\nJust an incorrect type annotation to correct here. We need to figure out\na way to properly type the \u0027Resource\u0027 class with all its magic.\n\nChange-Id: Id238b870de85a4663032710d1c15f6e200d2d543\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4f8d4102f5e73963d2decc3da12e5505a8e08f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4f8d4102f5e73963d2decc3da12e5505a8e08f5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
