)]}'
{"id":"openstack%2Fopenstacksdk~900719","triplet_id":"openstack%2Fopenstacksdk~master~I9a08341002cbf90ba9dccaacaa5f2d3e7d7560a2","project":"openstack/openstacksdk","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-30 09:09:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a08341002cbf90ba9dccaacaa5f2d3e7d7560a2","subject":"mypy: Add typing to openstack._log","status":"MERGED","created":"2023-11-11 18:11:30.000000000","updated":"2024-01-30 09:09:23.000000000","submitted":"2024-01-30 09:09:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900719-mypy","meta_rev_id":"15296814e052737d31e7311962ac7e9a5d588de1","_number":900719,"virtual_id_number":900719,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-30 09:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-01-30 08:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-01-30 08:03:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-11 19:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 08:03:10.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"5b7b41deb94b5038d04c9f6bcfb77bb917e48241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-11 18:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c59da85362ac714189f34f50857e16c74dc77817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 19:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55c0c795cfa34ec9a308d40216211e34\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f6e593b9a486457f9bc94ee878dd2a68 : SUCCESS in 57m 26s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ff94fc1b3fd746818b96f2109e87b113 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a5182cff6e48ce84262e82b191da67 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95778f87064446a4950a296158da1480 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b5fa9ab38bb458a80bef7beea5c25e1 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d9c11ffd8fce468696feaf1dbd07ead4 : SUCCESS in 1h 08m 26s\n- os-client-config-tox-py38-tips https://zuul.opendev.org/t/openstack/build/59761adc16dc4ad3b74587228bd984b6 : SUCCESS in 3m 28s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/4fba4f28c2774c8e8deafab5508fc953 : SUCCESS in 5m 30s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/4e143daf1c77454aad634fa2d74d02f2 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72735c0f654c4a2bb5d22134f4c00041 : SUCCESS in 8m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/78396d77ac944c6890fdd0fd1e1b975f : SUCCESS in 1h 03m 25s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/538b36bf80394c76b89eb13d1d5d8bf4 : SUCCESS in 15m 32s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ebe4cc32832a4bd194e770c0eb302846 : SUCCESS in 34m 45s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f6fdea558344b34914adbffdb129924 : SUCCESS in 1h 02m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9c4f0ec2eb9f4872b84c489df3c19546 : SUCCESS in 1h 05m 18s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1e8ff99a95c740d3b53d50ec5ad60918 : SUCCESS in 1h 23m 37s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/d185379044ef4fe78aefa8a9535be8ea : SUCCESS in 1h 01m 52s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b26b4a1eb3d94a02b79e3c17491bf455 : SUCCESS in 42m 50s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ce24d8aa6edc4121a28b5c1fd4fe7d4e : SUCCESS in 28m 19s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/bddc7b233df1402d8347d993be0c5999 : POST_FAILURE in 31m 49s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8f357baad17a4c1cbaac9477e1dfcc62 : SUCCESS in 45m 40s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/294e7bfd5be3427e871b760a13f5cf84 : SUCCESS in 1h 10m 21s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3640c3520edd4dc3aed1ff2d97723d16 : SUCCESS in 1h 00m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"072360ecdc4da172e454542dfe545224703b0bd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-09 11:42:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bddbb142c9f792978800147894f42993aac7f1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 13:23:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95e7bd1538a42cbb3c19fea4b0e4ad0\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9096afe659ea4227b5bb80c705611cc1 : SUCCESS in 1h 01m 58s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/904ee400b6034e4d908944897c2c810f : SUCCESS in 40m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41c9b489f2f48fc8370b8afb939cd1f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57219fe0dd2f4ec1a091951f09c8aab6 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2215e9be76d43f39c419c1ab05dd456 : SUCCESS in 4m 02s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/477420a22a2f4d3e9aebf445a1aeddbe : SUCCESS in 41m 23s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7cd133fba8f4444f968cfc11ee0d9f19 : SUCCESS in 3m 53s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/364a17fafbf74a5dad1815556481b8e6 : SUCCESS in 4m 56s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/70ad3d15115549c5b5857a5214ff4272 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ac317f06944b80b65fabc6949fa245 : SUCCESS in 12m 19s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b50aad34f2834c4d9c0a4ae57969840d : SUCCESS in 1h 01m 42s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0bae77496bd44c71a87c65c8fe7bea65 : SUCCESS in 15m 24s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/79423f95349841deaabefd1600729346 : SUCCESS in 32m 47s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0eafb83bc43040b9bfe621562d4a5bcd : SUCCESS in 42m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b7c7718401b04914beedeb4c38d72e23 : SUCCESS in 42m 34s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ff9a90866cbb49aeb8307e9ab93f9ad8 : SUCCESS in 46m 42s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/cd14fd0835014ab988e1edb6a3063cb4 : SUCCESS in 34m 23s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/27c5fc778e204a81b8122c76fa440a72 : SUCCESS in 27m 26s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ef077077ebf44455b7f7319304fcd214 : SUCCESS in 21m 48s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/05654021391c4536b882a81f15fa8b7e : SUCCESS in 33m 08s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/13268ce416be434ca8fbfcb3563b1f0e : SUCCESS in 48m 39s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9a14456888c14ab985c5370acb45809e : SUCCESS in 57m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cb0921f84d6942e284078c49a2126d3c : FAILURE in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6a455e38198ced940477a3739d3f0038b2f8206","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-01-30 08:03:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5033d39c38e82af92957e14370a4e6280671672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 08:03:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0cffa5e841b5d3047f94e81e8fae06faacd79cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 09:09:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7de10153b9b9461aa3292ac3b8e99ad2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e3eab1f6b6455f8f9b7b767c40f55f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c907e4bdf84a2aaa56984d36c01ba4 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f6ea1041f514813914396177ebd67b9 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f86e701df3f849e488aee4c982a8a033 : SUCCESS in 23m 46s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4e0683a574564cba96056117f70ed3da : SUCCESS in 4m 44s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3dd6aa6e6dd94dddbf6a9af1781f3176 : SUCCESS in 5m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9de70c664db94e168c71bc85ec8c0ee1 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0613f95eb6de408093d3e70456158c4d : SUCCESS in 9m 17s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5d0660dc714f49e1be728d0094048474 : SUCCESS in 45m 18s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ab2d3b96db2a4aff894ac5548c8db5c8 : SUCCESS in 13m 52s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/96e4be6d8a2645e298ba35723872bca0 : SUCCESS in 26m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c19e872e86847c4884117850fa8869d : SUCCESS in 34m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f30df59283444106a7dafce66b564bec : SUCCESS in 37m 49s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/0f7673f75e2b4b16ab6474157a074a83 : SUCCESS in 31m 10s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/f178fc09ea474eee9628d89dd32843e5 : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":2},{"id":"15296814e052737d31e7311962ac7e9a5d588de1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 09:09:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d506b9ff49e2f5d2826722117bd0481e4859483e","revisions":{"b05630484429ce85650944c90f344be4ec19630f":{"kind":"REWORK","_number":1,"created":"2023-11-11 18:11:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/900719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/19/900719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/1"}}},"commit":{"parents":[{"commit":"0620684f7291e3f60ab8bd30c65620db63744ca0","subject":"mypy: Address issues with remaining service modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0620684f7291e3f60ab8bd30c65620db63744ca0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-27 10:40:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-11 18:10:46.000000000","tz":0},"subject":"mypy: Add typing to openstack._log","message":"mypy: Add typing to openstack._log\n\nA nice easy one to start.\n\nChange-Id: I9a08341002cbf90ba9dccaacaa5f2d3e7d7560a2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b05630484429ce85650944c90f344be4ec19630f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b05630484429ce85650944c90f344be4ec19630f"}]},"branch":"refs/heads/master"},"d506b9ff49e2f5d2826722117bd0481e4859483e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-09 11:42:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/900719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/19/900719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/19/900719/2"}}},"commit":{"parents":[{"commit":"554dc6284c15c7a5c8bca52ed148f88f04344564","subject":"mypy: Address issues with remaining service modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/554dc6284c15c7a5c8bca52ed148f88f04344564"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-27 10:40:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-01-09 10:54:07.000000000","tz":0},"subject":"mypy: Add typing to openstack._log","message":"mypy: Add typing to openstack._log\n\nA nice easy one to start.\n\nChange-Id: I9a08341002cbf90ba9dccaacaa5f2d3e7d7560a2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d506b9ff49e2f5d2826722117bd0481e4859483e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d506b9ff49e2f5d2826722117bd0481e4859483e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
