)]}'
{"id":"openstack%2Fopenstacksdk~909109","triplet_id":"openstack%2Fopenstacksdk~master~I160d95eebfe4de5bbea77820e40632f2a1dfd67a","project":"openstack/openstacksdk","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-11 11:36:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I160d95eebfe4de5bbea77820e40632f2a1dfd67a","subject":"Fix typo","status":"MERGED","created":"2024-02-15 12:18:14.000000000","updated":"2024-04-11 15:28:51.000000000","submitted":"2024-04-11 15:27:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909109-trivial","meta_rev_id":"c3f0deee63f3d89c3c315009482ea4f82df27385","_number":909109,"virtual_id_number":909109,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-11 15:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-02-27 14:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-04-10 09:15:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-15 13:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 14:56:53.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"4e15a963d7f8c6dbbc8fecd9382cff60d0ff3da7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-02-15 12:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f3f6740c71dc482289ac17ad126a10ce823a25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:41:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc3984c50274871a62a3610826cb043\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/89257a3ddf42460981a92d77527ad7a9 : SUCCESS in 57m 07s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/68d002d723cf45cd95ea0f3d6b9aae6c : SUCCESS in 40m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92159ab99f7942c7b04c34cfb2044efc : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcf3f9dbded84c15b29aaf6456a8b497 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc076c4e225945e6b367a2460b778fe7 : SUCCESS in 4m 26s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/368c494cadd64462ae00e2cae731213b : SUCCESS in 37m 12s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8134420cccf149a2b3093b12febc8814 : SUCCESS in 3m 35s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2b38d22870564f8a82bc1d94d5e5827d : SUCCESS in 4m 45s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/4d11485ce29d45c3b88a9d5b5b44a07c : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f35a8f6d27488d9ebcd4cc829c2e20 : SUCCESS in 7m 50s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/885303b0d56241b6a34441a2c1b486c2 : SUCCESS in 1h 03m 47s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/071f72dc82be4bce844b11da8058ef4f : SUCCESS in 17m 20s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/5d3994e405b74c6fa96ef150c439bf99 : SUCCESS in 33m 10s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7b20c50bc88461baae3f805c1b1d96f : SUCCESS in 37m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/586f51065e374c2cb94a3582d132dad9 : SUCCESS in 22m 16s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8adbc462b76948bf8f5affe4987e955b : SUCCESS in 24m 09s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/6e48ae27fdbb464db699d27c15bd52f4 : SUCCESS in 25m 02s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/8c85ea73bbdc457b97350caf0572e63a : SUCCESS in 35m 16s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/19514db987234ee3a48fb2b1eea8a8ba : SUCCESS in 24m 44s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ed9263a4caf446e2885db4aeb60f53e4 : SUCCESS in 40m 37s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b922de3e1dd34eeca6044f5e68d2ea16 : SUCCESS in 33m 25s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/55e55548b59a43d3ad27aadd80884ac4 : SUCCESS in 1h 12m 28s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/edc83e1b38eb4be9a5847c6092a98c00 : SUCCESS in 1h 05m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"375b211588881a395bb03e8139eee50588aa41c3","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-27 14:56:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daf5688814da0afd3f08b33933381f1fcedf3ef6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-04-10 09:15:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf0e75a974ce03b78c7789ee91f4206862c93532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 09:16:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9ee6ef4a1b7f5ff60bf35254ab7c72c3740e341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:03:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98b9c751f574497b77407e7bf93bdb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89fabc768774d1ebd8c28458cac26fc : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b95b26df114a208626dc01097bfde6 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f01b57a013a948a59a3ed65c2da6668c : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/91e3155f29564038bf1ebc8aadd85211 : SUCCESS in 51m 32s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/42cc3f7a8d314b41a1f9537c0d6cc2a5 : SUCCESS in 3m 45s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/28f70a7e02a44b7281014f973fb08bbd : SUCCESS in 4m 13s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/417de64f31b447b18fe3ff7f52ce0dd2 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445f683bdbfc4e888edd9dc6592d82c9 : SUCCESS in 7m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/daad2892274847e8a44f55ffc695e5b4 : SUCCESS in 1h 09m 33s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/885cc165bc034a0fbe34287cf8073655 : SUCCESS in 13m 03s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/f7519f185958485ab4779233ac42b7d7 : SUCCESS in 33m 48s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9e60e2c634946e39a3ff94856236b6e : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/08bde5d722ff4caa9f2a1b8d83066430 : SUCCESS in 1h 23m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e06a8b23da9442cd99776c4b7bc7230e : FAILURE in 1h 38m 46s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/809b513a14334de7af71f58bfea63a2c : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"598039e487a2fcba8af11f8e2fcce7fc9e606e80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-11 11:36:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce2354d92e89cb42dcc4710005442a770ca72668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 13:28:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67d252360f44dc89d16e8af92aa7bd4\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/30cc37ba5bf24651a3aac34db626719c : SUCCESS in 42m 59s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/e6744979c107469392487cb77503d180 : SUCCESS in 31m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad36dcc0d9746bd855f3d98f78ba436 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc9991f8a8a64967bdb831c60f6d5a18 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83862a22586542038c5e881f905232d5 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8ce388fcf2949fc916ab79988968a1b : FAILURE in 7m 08s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/704815a7f1884ddda016d760e56a5b25 : SUCCESS in 1h 02m 22s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4fd522c439694fae88789964606499f8 : SUCCESS in 3m 40s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3420b23d96e6443c9968d50c00f39dbd : SUCCESS in 4m 17s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d58b707cb31e45ce8d24f8659b041509 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb251580c1de49a7bf13e23af107a3e4 : SUCCESS in 8m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/361f07d9442d4595879774da3606ce7d : SUCCESS in 56m 40s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/5184bbaaf39f4407bfb73d8317fc7741 : SUCCESS in 14m 13s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/caa137cef2114740a0c3a1904e56ec0e : SUCCESS in 34m 22s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a1e0c65efe64222ba21dfec4eefbb6d : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/81d76557a5bb4ba1b4f904135c24bf85 : SUCCESS in 1h 09m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/33cc15f87be044b48caa824c2e799774 : SUCCESS in 1h 14m 55s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/1f0892ebd2074ff593cbd94dd618bc2a : SUCCESS in 54m 25s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ff0e7d5a8c5244eda88f1e7ae9a84127 : FAILURE in 40m 31s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/139a7142d4884c5e8188083f682a54de : SUCCESS in 35m 30s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/26e4662adc934706958b37bbac879b1d : SUCCESS in 20m 10s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/db2e4dba34a34b1bbb5a782d009bff7a : FAILURE in 24m 49s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bbe37c7edd9e4be5a7dc2adfe17b5e07 : SUCCESS in 47m 17s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ac8f24c3c7cd4efebd6fc67dee6cfcde : SUCCESS in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bf9ff740e7d4ea41fae6e2e1a5e5dea6cf1a83f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 13:29:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d56a40735140602f2e95a691365bcb87b53a0d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:27:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d3221ee41a45caabf2c7780fa69290\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad6b49bd6dd48a48aed4d78014d0141 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6130560c1a384a6f9c55f72b108f28fa : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d2ae8ab695348dc8c6db83777e81b7a : SUCCESS in 3m 50s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/85bf55a6245b4e20881a1a377214c52c : SUCCESS in 1h 05m 39s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/16d5b797031d4cce9e406b282309a5b8 : SUCCESS in 3m 43s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8c22f970419f4fc49baa970fc348663f : SUCCESS in 4m 29s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0bc1198c91184a8f9db9ea7ab7ea6d5a : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e14631036bc4123921ec7bfac6defcb : SUCCESS in 8m 54s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0bf9c873430545408b664212fa71d533 : SUCCESS in 59m 55s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/16e7fdac6cdb4bf7a162803bbf01202a : SUCCESS in 14m 09s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/9c330a1b4d9d45f5853c694440ddebeb : SUCCESS in 31m 38s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4354efa546c4412bb222a2889ccd792 : SUCCESS in 1h 00m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/937a286d418a4d838eb798c2f7fb1733 : SUCCESS in 1h 24m 12s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/fa3fd2f39e6f4c59a2441b4b46f3a43b : SUCCESS in 1h 26m 36s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/23e8ea1f3ec242d4ba2641d72fbf2deb : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bf267d75118b8a4f36778ac819d152371dccebcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:27:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3f0deee63f3d89c3c315009482ea4f82df27385","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:28:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245b215c5c084392be93cb3d8777f539\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1115d96f14a94696aabdf578edceb0df : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1ff60d125fad059ffd8b45fea70b76e61f72acc","revisions":{"d1ff60d125fad059ffd8b45fea70b76e61f72acc":{"kind":"REWORK","_number":1,"created":"2024-02-15 12:18:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/909109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/09/909109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/09/909109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/09/909109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/09/909109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/09/909109/1"}}},"commit":{"parents":[{"commit":"c2baf1db41cce65ce1aa8ac8908c4e0d98f7a94c","subject":"Merge \"Incorrect protocol type in create_security_group_rule()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c2baf1db41cce65ce1aa8ac8908c4e0d98f7a94c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-15 12:16:58.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-15 12:16:58.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I160d95eebfe4de5bbea77820e40632f2a1dfd67a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d1ff60d125fad059ffd8b45fea70b76e61f72acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d1ff60d125fad059ffd8b45fea70b76e61f72acc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
