)]}'
{"id":"openstack%2Fopenstacksdk~919884","triplet_id":"openstack%2Fopenstacksdk~master~I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb","project":"openstack/openstacksdk","branch":"master","topic":"compute-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-06-18 15:06:59.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-06-18 12:00:10.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb","subject":"compute, block storage: Add support for quota class sets","status":"MERGED","created":"2024-05-16 16:28:21.000000000","updated":"2024-06-18 15:08:18.000000000","submitted":"2024-06-18 15:06:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919884-compute-gaps","meta_rev_id":"25eb717fe7bd767feae009a719133a195eb78478","_number":919884,"virtual_id_number":919884,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-18 15:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-18 12:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-18 12:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 17:59:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-13 11:23:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"2cafa7b154c230767bf45a5af396222f23ed68ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-16 16:28:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"741c7a8c6e736715184d8686bfbee0229a4484b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 17:59:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4673d86b36d4e8dad50e89fd19aa3ba\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/478c48f2120344579759f0482fdf10ba : SUCCESS in 46m 48s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d4c6a19f9dd84ab1b0326e8d8fec56d9 : SUCCESS in 42m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac8d6f12ca541608a1ea9d84d15c5bb : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6ced3194d7f44f08ba790f65729c865 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7b9f7f845b641cea288f81c7a3fe4b6 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3292b49faff74ff2afb0f6c7d4a7a1a1 : FAILURE in 9m 40s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/29f123ee04364c869c34ffbd0887eb17 : SUCCESS in 1h 02m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2285a68a7fc742f9aa8bee9f5787d634 : SUCCESS in 3m 51s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/07955e3cc4a4480aa5510d9e6b57ee53 : SUCCESS in 6m 06s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/c1abf8c90394483093bdc36c2ece56b0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d0419eb69e4eaab73ff7aa0675e739 : SUCCESS in 9m 34s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f12361d0beca4e32bcabb9a71301dcc9 : SUCCESS in 53m 32s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/c86a81deb2c74a45b9b5b96bee00cd37 : SUCCESS in 13m 44s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b6319d01a85247a4a125321297e93ae4 : SUCCESS in 31m 28s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1107479ec47a4e67a29d9ef97fe06796 : SUCCESS in 1h 00m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0f1672191a364c008e0ea94071d95763 : SUCCESS in 1h 21m 34s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/fc6d1bc0767f4341ae319e384c404921 : FAILURE in 1h 25m 17s\n- openstacksdk-functional-devstack-senlin https://zuul.opendev.org/t/openstack/build/8e501139b226452ba0ce8b007c3563cf : SUCCESS in 44m 01s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7c8b8ad741524730b2d17c9730016f01 : FAILURE in 1h 03m 17s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/794a4970a8c84683820fb0c7e45fdef4 : SUCCESS in 30m 43s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/82716bc774d14ab49ec6b582d88a12ad : SUCCESS in 36m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1eaf28d9b2154fdf9ca116bae6a985b7 : FAILURE in 29m 45s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d02b14ac7fb44fd4a41962c0e6553676 : SUCCESS in 59m 23s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/030e0ee9060a4460a732d2a689399271 : SUCCESS in 50m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef435e818ddaea86b93886bc1deb9c5d83c27329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-17 12:21:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcb6ed4837c515759feac1b48f8269afe25df5e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 14:02:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a90c5ea9c5f04de5b24da469bdea8ac7\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/73c20e0de2004ab8ad15760799316162 : SUCCESS in 1h 00m 10s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/cf32fbcc17274cf484d25cec8d21337c : SUCCESS in 33m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e75772ca6a74e27b1c86d23ba3e0440 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a364542d6da54bf7b9d54d942e4cc9c7 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f8922e6f587406daae594af1e3b4e99 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91889a1e181f46329ff7e9c8a6ec8a9b : FAILURE in 8m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c9fdfa44b72a4193bd8d9a60281145a7 : SUCCESS in 1h 00m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d3cadec868394ee4ad0817e64078ebe0 : SUCCESS in 4m 30s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c7ee24c9840447baac51da6ca5fb8e17 : SUCCESS in 5m 25s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f4814798183e4095a000a25de8a3174f : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c082a61ddd4caba2c8204d6cbbe690 : SUCCESS in 9m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/973cdf4c23414628ad9cc799048bfee4 : SUCCESS in 43m 59s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/eb9d2e98fa9442a2a796af401b3f2398 : SUCCESS in 13m 01s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/8889c52860b84c1a98f8e840ac2fa22d : SUCCESS in 25m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01ae5d7dcfe94aa39d3e32e3128bbac3 : SUCCESS in 51m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/31f917feb50c480da1b2dc6d5630ea2e : SUCCESS in 1h 31m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f48b2c5d47a540fda0181f0d540f709c : FAILURE in 1h 23m 23s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0463b500c0464dfbb207e5a2cdeceee8 : FAILURE in 45m 03s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/dc2304e191004cc49d311eb6e8494c9a : SUCCESS in 27m 52s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b24e58b12f8e40609ac91828da5bbdc6 : SUCCESS in 37m 09s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/25cfec91b7e94ebfb56f8261de50e231 : FAILURE in 32m 09s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/87ea732c090f4175af1ceede7dcb226e : SUCCESS in 1h 01m 45s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a234a1dbcc374a03867752ca306d793b : SUCCESS in 1h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe3ec633f826e92b1d45660078f4aa03b506e10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-11 12:15:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c3c42a76fa848c42b3d9f81c407eff504bcbdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:10:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b4e6472d0544aa88b0893d989b5995\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2197eecb0e8546bf89f9517eb3ecc4f6 : SUCCESS in 46m 00s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/abdd142759c44a52ac1a310375581f1c : SUCCESS in 40m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bff07886da4029b05c34e263f2f7af : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/272a9351c51d45de80eed1d6ce5f3f67 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e5ff49f3e9542ea987a6991ee35bc00 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f19860106a44a71a46a514b41445855 : FAILURE in 6m 19s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c50e55efe52d4295a6b61458244e5846 : SUCCESS in 1h 03m 50s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2b623f592bfd4f0ba58916797900820e : SUCCESS in 4m 59s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/422de34a476b4c668febf994be2494a7 : SUCCESS in 6m 23s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/7f048153160b4bbdaa40d7f2f07a2cfd : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b64bcf929b4ad0bd9f557e9ab451c1 : SUCCESS in 9m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ee18ab18713b4dfd9667438c608e5610 : SUCCESS in 1h 00m 57s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/37bd46a5c3584c1e8a669dfc1dbbf7db : SUCCESS in 13m 50s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/4641bf79878046e38609a9ec1fc1923b : FAILURE in 19m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1023c45d6ac34056bb6c2d13571a096a : SUCCESS in 58m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c27a104db0a2493db0c91c8c2ce0e45c : SUCCESS in 1h 43m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9c6e2cb3ea1949a9b45f195e7954b99c : SUCCESS in 1h 30m 04s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/602f29d838984aaab8eaa41802f8d60f : FAILURE in 1h 03m 49s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/dd59d59e5d69455e940b09c3bcaca004 : SUCCESS in 37m 58s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b2d8d5095a0e4c93a665b34f35581063 : SUCCESS in 22m 43s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ae0695d795c64ec1a58ef0896ea9a6bc : FAILURE in 45m 49s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/333113eb822d47679406006985efa444 : SUCCESS in 1h 09m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ab2624337b32456eb123266f92e6676d : FAILURE in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"911024b7edfb0f1f1fded85715615222a5018c4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-13 11:23:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"354fde0a0924de81d6d402706ee8f975d624fccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 13:06:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86a7f5396b9e4792b15c295442ce70f9\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/93369582c4d54404a7f4ca337f721a45 : SUCCESS in 55m 07s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8130f9b7622741a6ad6ea51775c06224 : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70a0358878e4344b7230db1b9ab1afa : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/887b43ede22149019964f4920d9b2bbd : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd28ad2bcf174c5aaf410c5736cfc09b : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6c4e6b70c014193b6af60448d99c7f5 : FAILURE in 8m 12s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a34bcf03089048b49888f26efc10c6f1 : SUCCESS in 1h 15m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a37d96293737477d9fb71462a6daa817 : SUCCESS in 4m 10s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/caeac556ac6b447f9891844f6555d919 : SUCCESS in 4m 27s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d10ee5c843cb4d91b6afd2766e5302fe : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0056344ce84842d0ba7d63160771f785 : SUCCESS in 8m 22s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/690edef7dd2647b091200209e9ef545d : SUCCESS in 59m 20s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/367108cdcc514c92847c95dd462e3a31 : SUCCESS in 12m 38s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/00ce8f43e7154a488f9affcc10a1c9e2 : FAILURE in 20m 03s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c03819629924b1180b5f6d0c6739d90 : SUCCESS in 1h 04m 20s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0683fbb9b594508afd79d06ab54d7cb : SUCCESS in 1h 33m 07s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/c48ead3196864444895e3474f854e767 : SUCCESS in 1h 36m 47s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/04c89088a3134352b03882d1742ebcf2 : FAILURE in 1h 05m 36s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2b937073232b4b40bd87696f8ad507c1 : POST_FAILURE in 18m 42s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f7d30d82270046a3ae0f96e5ab9d462b : SUCCESS in 39m 38s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/608a6a46915f4576abf2bb5c7bf710f5 : FAILURE in 39m 36s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/666a9c5871dd4d10879418289af94db6 : SUCCESS in 1h 03m 49s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a1fdd3305dd24261825f42633b577851 : FAILURE in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8df3428efd036b79e95561086db18ed554ff1324","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-18 12:00:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b53b16c7b91f349cf46398a316a05e2edf6aca08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:01:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8d0cce481bf28be30cd1ca49cf5f37b94c45d941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:06:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eb15f1fd1024f6fb33f24249f75967e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25ccd78c91849a5ae5cbbf11fc58a31 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc576d59e188441894b2329d3abdeee6 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d43e01ae91704518856849a5824dd6be : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0d5ea99b834c4741bb5ff06b7e8a9b4d : SUCCESS in 1h 14m 51s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/770c7a8b3b4146bbab0d87f0f7ee1264 : SUCCESS in 3m 37s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/671e741ea3ad494ab389a58601336197 : SUCCESS in 4m 26s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/8a89764b7d774259a38386c219861658 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34616984eed4fe5961936047a5eecc5 : SUCCESS in 9m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6e48de62d9d4476db8dfb7e1f92f449d : SUCCESS in 37m 00s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f2975141781d4f5b8d17575773ff284f : SUCCESS in 11m 10s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/44a4323635174eaead461aaa005679f3 : FAILURE in 21m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60dfc23879f84aa7acfaf87d5c2ca502 : SUCCESS in 1h 09m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f7b3ed0c5874cdb87b964ab92ae1b62 : SUCCESS in 1h 26m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4cc11b7558264d96843ef0edf121e1e4 : SUCCESS in 1h 35m 33s","accounts_in_message":[],"_revision_number":4},{"id":"96b9c637f2a7236711b1f34342ae58c8488d3ebb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:06:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"25eb717fe7bd767feae009a719133a195eb78478","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:08:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e963c7bbb43f4d33b04ed64943b4a2e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c80f11c70fe4f81bc451ded1be49766 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4275dfc53c745e584103b7d5ce1c8fb0c5cdc46f","revisions":{"add88f0d41cd92de2e341560ba13b7ed4e722381":{"kind":"REWORK","_number":1,"created":"2024-05-16 16:28:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/919884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/84/919884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/1"}}},"commit":{"parents":[{"commit":"afbeb781d6a51d6cff85d3752d83065e0b3a719f","subject":"compute: Add support for legacy \u0027onSharedStorage\u0027 param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/afbeb781d6a51d6cff85d3752d83065e0b3a719f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-16 16:13:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-16 16:26:12.000000000","tz":60},"subject":"Add support for quota class sets","message":"Add support for quota class sets\n\nChange-Id: I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/add88f0d41cd92de2e341560ba13b7ed4e722381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/add88f0d41cd92de2e341560ba13b7ed4e722381"}]},"branch":"refs/heads/master"},"5b198b4361b48beb28e72fab68322d70b637857f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-05-17 12:21:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/919884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/84/919884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/2"}}},"commit":{"parents":[{"commit":"22044646f57202a6b31d6ab2bf4a46bc7f25b1f2","subject":"compute: Add additional server create paramters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/22044646f57202a6b31d6ab2bf4a46bc7f25b1f2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-16 16:13:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-17 12:20:46.000000000","tz":60},"subject":"compute, block storage: Add support for quota class sets","message":"compute, block storage: Add support for quota class sets\n\nChange-Id: I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5b198b4361b48beb28e72fab68322d70b637857f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5b198b4361b48beb28e72fab68322d70b637857f"}]},"branch":"refs/heads/master"},"2bd116b7f6e81f822b990a3ebb203bea2a609c11":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-11 12:15:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/919884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/84/919884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/3"}}},"commit":{"parents":[{"commit":"2a684fcf0a37623406c1d00af6b7425363c2036c","subject":"compute: Add additional server create paramters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2a684fcf0a37623406c1d00af6b7425363c2036c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-16 16:13:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-11 12:14:44.000000000","tz":60},"subject":"compute, block storage: Add support for quota class sets","message":"compute, block storage: Add support for quota class sets\n\nChange-Id: I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2bd116b7f6e81f822b990a3ebb203bea2a609c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2bd116b7f6e81f822b990a3ebb203bea2a609c11"}]},"branch":"refs/heads/master"},"4275dfc53c745e584103b7d5ce1c8fb0c5cdc46f":{"kind":"NO_CHANGE","_number":4,"created":"2024-06-13 11:23:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/919884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/84/919884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/84/919884/4"}}},"commit":{"parents":[{"commit":"0f311ff3e2e57bf3659cef77e98551b6c0c7e3c9","subject":"compute: Add additional server create parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0f311ff3e2e57bf3659cef77e98551b6c0c7e3c9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-16 16:13:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-13 11:23:21.000000000","tz":0},"subject":"compute, block storage: Add support for quota class sets","message":"compute, block storage: Add support for quota class sets\n\nChange-Id: I2d0b9a900d60d666ba07b759eccdb9a8e1c164bb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4275dfc53c745e584103b7d5ce1c8fb0c5cdc46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4275dfc53c745e584103b7d5ce1c8fb0c5cdc46f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
