)]}'
{"id":"openstack%2Fopenstacksdk~923214","triplet_id":"openstack%2Fopenstacksdk~master~Ic6762a131e03f78f1d290b0065f3628d3e478f8d","project":"openstack/openstacksdk","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-08-01 11:50:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ic6762a131e03f78f1d290b0065f3628d3e478f8d","subject":"cloud: Make _OpenStackCloudMixin subclass ServicesMixin","status":"MERGED","created":"2024-07-02 11:15:45.000000000","updated":"2024-08-01 15:05:57.000000000","submitted":"2024-08-01 15:03:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923214-typing","meta_rev_id":"5b0e4a4ef2a7ef507860a62239d680e781b42428","_number":923214,"virtual_id_number":923214,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 15:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-07-30 11:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-07-30 11:07:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 12:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-30 11:07:20.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"c632712a667acd61150e33091d7abb55306461c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-02 11:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5eed7ded08c43fc4fd7aa3ff7bb0cf20ceb7c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:42:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c071ac19bd5415f8f1aeafdcb1c63ec\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/74147b9a3b664b579da5aa63e2419713 : SUCCESS in 52m 17s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2925e1cb93a443b18b76968db86ea916 : SUCCESS in 36m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b917e8414b4b289984d96c12c22d87 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e1dddc352854d9497ad36bbd8cd69ee : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7232626e10a14153b0f8a4ab59615de9 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d117bfa86a0e48bdb716d787d172bb15 : SUCCESS in 8m 10s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9e582a35e40b4c61ac0e51ed5deacc28 : SUCCESS in 1h 07m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ac8a3ff2df394a268a9b431574711b72 : SUCCESS in 3m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/38817541cd4a45e08346138078577616 : SUCCESS in 4m 19s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/bb59f95695274f4c96ca178feb6fc4f4 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ba04adad994e5f9ec0b5d5e1c48be6 : SUCCESS in 8m 32s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4d6617befab84415802e38194cac6be3 : SUCCESS in 1h 01m 53s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/611695dd016b448b91be6b9e96f30496 : SUCCESS in 13m 08s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/82ca48c2a9754da4bfa6f82b1eeed57f : FAILURE in 19m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f45c2be7e6e54d4391a77ac65fb3c8cc : SUCCESS in 1h 04m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a8d11193de614495b945505d2a4e62c2 : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5f7bd7ea0c0943f9937e275074a1ffdf : SUCCESS in 1h 16m 04s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/3017c94a5d7f45acade1638971981aff : FAILURE in 55m 11s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/40a10538663c42c087f7989eac686484 : SUCCESS in 31m 27s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/176128076eb147f9a22528f8b977927c : SUCCESS in 30m 06s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7512ea575b944a19b07d0d7ab2ff0a2d : FAILURE in 42m 12s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/59678b3ccb804d8293322072052d5a31 : SUCCESS in 1h 05m 28s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1c4d8301fa4c4e93b9e188af75110bdf : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"636430e7dd93b5d56e69cd62b468fe34960b2c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 13:34:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8bb4c1072b41e046009d6d8f7bbc0e2262ee578","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 13:58:53.000000000","message":"Topic mypy removed","accounts_in_message":[],"_revision_number":2},{"id":"edc1adec4a5e7229445ec0a3c5e04801a2f871ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 13:58:55.000000000","message":"Topic set to typing","accounts_in_message":[],"_revision_number":2},{"id":"02a19e964dbbba5f808f4e623fcde31e4a09f2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 15:01:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f977e012afe347fbbed8ceb8e152dc6c\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/dab72efed04445a0bdf5dfbc81fea2bb : FAILURE in 28m 46s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/60df0ee4f3ef4a9ba9526d4ce51f8522 : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6581f934d6ea49e18e93fe1f38790b1b : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b36a159d7de54a6f96ff865ab97ac637 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f328637b4ba453f93b7381c555364ba : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd50bbe0ff5f43ce8b783927d95ddea8 : SUCCESS in 7m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1b29c4f7b69f493db51b781871b2996f : SUCCESS in 1h 07m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6c86a858753141358afb9fd9916b99b2 : SUCCESS in 3m 45s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0d90086113a3473b9217cd61ae752212 : SUCCESS in 4m 39s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/b6b68d22c0c447e8a1f21cc30af71b1f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d77b0c25a34c95bf6178e85e41e91a : SUCCESS in 11m 17s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/06974ca7f5c4400d9e964e4493b93452 : SUCCESS in 52m 04s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/736ca52a52234467ae78b5d4b56dbcb9 : SUCCESS in 11m 57s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/70e157295e2643929834b6882e136a8a : FAILURE in 25m 07s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c2d2ef7fd75499aaac1f30aa57fe2be : SUCCESS in 1h 02m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce3562d327414f239725b1a39cef4e49 : SUCCESS in 1h 21m 41s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ed98543a8c994e0eb8117b10b910c23e : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e75cedc813de4c3593fe02dd77b0cb31 : FAILURE in 58m 53s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/601b4d2a1578451a9f27c623d321d776 : SUCCESS in 29m 48s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/396b0dacca0d4fb3965e327c0e025ab5 : SUCCESS in 26m 29s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1d9b6cac464a4cb0845f80423b28c2e8 : FAILURE in 41m 18s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5e354a6ddc5c49c1aad0c09ba75b3b00 : SUCCESS in 58m 09s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/68ece93b86e849ae83061c4877f7a352 : FAILURE in 52m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4251d4cd8eca4b825d110dfb6b5f996adeea9c6c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-30 11:07:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"339f83d042f96dd68b05b1143fe36129f5b9a63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 11:08:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ecb4b8fedbf7952862a154f27a83835291549ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 13:30:17.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/923225 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ce8ecd32e570932573933593c6a91864a6c8febd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-30 15:51:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63f45acb4debe8251579315a449df91f7a01b716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:19:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1118904b3283472ca63c45f2b176dfc9\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/11c24274ad684fcab5a81878c560cb0e : SUCCESS in 42m 14s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b64757dcb581482cafdb899ef8a31379 : SUCCESS in 43m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec6594d79084c40b52ff28ebdafdf6a : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4367681fdcca461f876ee32beb446212 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6ba8eb2763744b784be3a610531dc82 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c5e779232884153ad695177b583c2ac : SUCCESS in 6m 34s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cbcf1ebbed0c48d99037e463ab01703c : SUCCESS in 1h 14m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f10d8fdc79014b099b7faccd0090ae25 : SUCCESS in 4m 00s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7e56b529182b4a4a82395a4b890efdec : SUCCESS in 4m 01s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d9d10188039748c3b86a8aef7fce5155 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e45aad08b246bc9da9b6aa81878b6b : SUCCESS in 9m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cc42e3e6dcf8474896528b7844ac1395 : SUCCESS in 50m 24s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/c4e3dfd771384b82af02b3fc6b8c2c56 : SUCCESS in 11m 25s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/f566b9cb2f7a4de2aa14a7207c430baf : SUCCESS in 35m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e55b8fb90947389cd20637d0ef562a : SUCCESS in 58m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b972b180934f43009ac5bbaf16e9a8b5 : SUCCESS in 1h 23m 20s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4b030d161a274000a33d3a23d1a5c338 : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/70ed14c9eeff4297b7a0faafb0c3d3b9 : FAILURE in 50m 41s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/600e1ad01029441c9e5b652080d3f2f3 : SUCCESS in 28m 12s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/dc34f538d47046ceba635ee9c2d7fa89 : SUCCESS in 24m 52s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/eef91acec8474013a5bde1527c18b2a0 : FAILURE in 34m 01s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dfd3c02e0c314cbc857642232b018453 : SUCCESS in 59m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/861d4732fa35441599fa5715ed833efb : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc6917ca2b38621329d4e4e3f3353a5b40197a1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:23:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd168c57d863ca9d16fb3450c2cbd94fdd18706c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:59:30.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/923225 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7166716235d8e0d762a22ab38f84d46f01268627","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-31 17:48:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26255209cff41afadc969a679473333c7ac4aecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 19:34:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a4b43e424be4a21af501ce4c808cf42\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9162e618c25f4dc1ae38bf73a2d75efd : SUCCESS in 55m 08s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/75b3a90ef09e46fa82aa92d58ed3eeb1 : SUCCESS in 36m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3689bc58d03a4070a7dba44b5533f06c : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13da7612d88e4b6a87b6ef5c844584dd : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32a1e6573feb4683946abd8f6d386985 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/855bd0e16ee64f4bb4037f58e374ba2e : SUCCESS in 7m 04s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/12fdf3ac397143d08c0d46d4d17d20ee : SUCCESS in 1h 00m 30s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0760a75481184bcd93be34cbd3fc247d : SUCCESS in 3m 58s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/660cbb2ae4274d7fa448dde3f241991c : SUCCESS in 4m 10s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/8337bce11e3d4a959455b449f6d8aebc : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4bd76bee034e26a5baa27dcecc4f4b : SUCCESS in 8m 45s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/09636c4ca0b54b0cab2fd5aab3dcb82a : SUCCESS in 45m 25s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/27d0a9d65d45423e863ff75975198b16 : SUCCESS in 10m 12s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/9106badb2bcf4b5d8abd8f427d7c5617 : SUCCESS in 29m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f194148f340a4197badaee8b097b53a7 : SUCCESS in 51m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/97ad1584fef049f69e907b2e2e7a8aae : SUCCESS in 1h 22m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5e5ae3b652e0472ea230a99205322ac1 : SUCCESS in 1h 38m 39s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e5506b19846f48c591075e8dff074243 : FAILURE in 59m 12s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c86812a1c80047958a4ecec2df131169 : SUCCESS in 24m 33s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/510fbfad6a8e4c4ba63a337dc2562863 : SUCCESS in 36m 38s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8bc9f5c5532149d3869709a56ad46b86 : FAILURE in 34m 05s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/471f2e06a429404bb10e5d858efc4861 : SUCCESS in 56m 11s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/17897db5a1904448becda7fd4d0f114c : FAILURE in 35m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ff215108c3dd305286813fdefa2cc5cd65c46e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 19:35:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f58afb557ca2de3a7e837788b65dd1e5211e41e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 21:42:17.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/923225 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"8311419f10ff1e88b511212a9edbe2299f6fb903","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-01 11:50:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac405c88de88397f69daec76587020d988318d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 13:26:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ba4122347f4dc4b4df910e4fe77894\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/52726c1b3a26410a95fb26c9687916c5 : SUCCESS in 51m 09s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2ac20f4326f94fc1bd79076f244431a6 : SUCCESS in 44m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a12056d82d42d59eaff3cb52fb14d4 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8991c4e705d470b914005e84c343e5e : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40f67d378a1743208b03e17ca98fca7c : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d830d7436dd494dae7c9cadffe71916 : SUCCESS in 6m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5d597547d4334d4e8bbfc85bc76922d4 : SUCCESS in 53m 28s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c0f31c1d2580489793718aa35f649232 : SUCCESS in 3m 08s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2bdf8f55e6d94bdbb21964ee8c611f8b : SUCCESS in 3m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/5046c579147b4f0fa3bb5b3957f16b9c : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e440d184784bf6a8b73fa2f558775e : SUCCESS in 8m 49s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c8bbe33c72f9473a867c905c6de14177 : SUCCESS in 56m 37s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/6354700b6f9e402694d47bc1e7224962 : SUCCESS in 11m 58s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/34c136e8bb8e4ab4a550167593632a60 : SUCCESS in 31m 26s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0e4e6f704d44ccb8f9cdd7118d5714 : SUCCESS in 57m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8db09bf649824bc4b8cbb8512a0f7dd0 : SUCCESS in 1h 29m 14s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/277ae46cb1294259a23dcf37f0bdeec8 : SUCCESS in 1h 29m 17s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/1e21a06a95a34fcb8aa21fa24c1a4ac5 : FAILURE in 43m 50s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2ebc441b6bb7458e97a20d63dc54d90a : SUCCESS in 33m 12s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e8413c0a4c59472384764d90ec0f4431 : SUCCESS in 19m 47s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/afd0ad6ffce44ed3b7b1b32fc2bc5be6 : FAILURE in 27m 23s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3191efed89df40f49e9fbe2080b43a7b : SUCCESS in 1h 12m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d126c0f9104881b2ec38ec59627f21 : FAILURE in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"137b08c57e75419744cb7dff3aba80bfac61237f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 13:26:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"feb710e638a3969161a4b6b4cc969641fa8925e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:03:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b9b33da65242b799cbaa1fe6c210ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acce07c80cf448718388ca409892086b : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73e3645a8e89474fb9f7d285e36fe278 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4312c14bc9384680992212d4a8beee52 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2d07b2281d544f83bd1a1d68fde665b3 : SUCCESS in 56m 28s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bfd35cb4ab3b49b38afceddee91f8bb2 : SUCCESS in 3m 44s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6ec36e3c485f4ee8ac27d54c4024ed29 : SUCCESS in 4m 16s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/baacad2f247f499c8147793f9f4eb04d : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6aedc951b8f4c789bffbf9958925365 : SUCCESS in 8m 26s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/dd319c10aa8843f7847047058aa46bad : SUCCESS in 3m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a61a2aa5700f4223bc37fae7bc8b9d36 : SUCCESS in 58m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7372e7bdd5be4c56974eb0568c1b5de2 : SUCCESS in 1h 21m 15s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a584db8cb7924f248576bd11dd60996e : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"44599d32ab21d57be883da29b9fe4453509f3d76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:03:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b0e4a4ef2a7ef507860a62239d680e781b42428","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:05:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e184bf279e3e4cdcbb9c7ec5aff06e6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7adfe4712bee478d93930bf844846474 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c842154091969b5dd56b500421f60d6c49857ee5","revisions":{"bc32ca7df260eb1322ef33adb9e4f151c3ffa5d6":{"kind":"REWORK","_number":1,"created":"2024-07-02 11:15:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/923214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/14/923214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/1"}}},"commit":{"parents":[{"commit":"89aca18c74702322b060537dea56644d01015bdb","subject":"cloud: Remove check for nova extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/89aca18c74702322b060537dea56644d01015bdb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-13 12:15:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 10:16:25.000000000","tz":60},"subject":"cloud: Make _OpenStackCloudMixin subclass ServicesMixin","message":"cloud: Make _OpenStackCloudMixin subclass ServicesMixin\n\nThis is the next step in normalizing the class hierarchy in\n\u0027openstack.cloud\u0027.\n\nWe also fix some docstrings while we\u0027re here.\n\nChange-Id: Ic6762a131e03f78f1d290b0065f3628d3e478f8d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc32ca7df260eb1322ef33adb9e4f151c3ffa5d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc32ca7df260eb1322ef33adb9e4f151c3ffa5d6"}]},"branch":"refs/heads/master"},"c842154091969b5dd56b500421f60d6c49857ee5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-24 13:34:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/923214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/14/923214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/14/923214/2"}}},"commit":{"parents":[{"commit":"200e1524acd24a6c2bddace781c78f5a6891406b","subject":"mypy: Enable checks for openstack.tests.fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/200e1524acd24a6c2bddace781c78f5a6891406b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-13 12:15:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-24 12:13:59.000000000","tz":60},"subject":"cloud: Make _OpenStackCloudMixin subclass ServicesMixin","message":"cloud: Make _OpenStackCloudMixin subclass ServicesMixin\n\nThis is the next step in normalizing the class hierarchy in\n\u0027openstack.cloud\u0027.\n\nWe also fix some docstrings while we\u0027re here.\n\nChange-Id: Ic6762a131e03f78f1d290b0065f3628d3e478f8d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c842154091969b5dd56b500421f60d6c49857ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c842154091969b5dd56b500421f60d6c49857ee5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
