)]}'
{"id":"openstack%2Fopenstacksdk~923223","triplet_id":"openstack%2Fopenstacksdk~master~I6b74ad9cd8f32793d24c89caec5b08347d7d563f","project":"openstack/openstacksdk","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-08-02 15:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I6b74ad9cd8f32793d24c89caec5b08347d7d563f","subject":"tests: Remove TestClustering functional tests","status":"MERGED","created":"2024-07-02 11:15:45.000000000","updated":"2024-08-02 20:40:37.000000000","submitted":"2024-08-02 20:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923223-typing","meta_rev_id":"cb6962d925f61e04d22fc47a86bd7152b59b3b46","_number":923223,"virtual_id_number":923223,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-02 20:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-02 12:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-02 12:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 12:55:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-02 12:57:17.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"91687fe9608e0fd2912e2c0b5f92820306cfec4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-02 11:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d881a30e3ce536a06d10182c9ad9917afca083f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:55:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d0f5a5edaa42ca898913628cea8e7e\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6c66b885f41d42b089e1a160ccc3f1ad : SUCCESS in 51m 15s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/559886f40f9947c0915c9461bb9a656c : SUCCESS in 41m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ba7d881d1a47d880478405d32d5739 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f623cb5ee6664e0ab093b2427b31fb0a : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f85978303294f218814933351f709b8 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ff21e58472845ccaf820b6e7a602306 : SUCCESS in 9m 14s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f9d5bc2805b74f679f91abab3b7ef979 : SUCCESS in 52m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c2342f3730443d6a9d591e5f5b4c6f1 : SUCCESS in 4m 01s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ada1a782242249349d50143a8e231f21 : SUCCESS in 4m 23s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/215b031a282e404d80c68e43db7ed25b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1e9337afc04a86948b29f29e1a03f9 : SUCCESS in 9m 25s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/269e15a645054bb5866fa8fc0ddf83fd : SUCCESS in 53m 54s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/60297c27906643e78443926ea0c956c7 : SUCCESS in 14m 35s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/8dc128f1e139442facd6249121305abc : FAILURE in 18m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9962194c7e3c47658c2c8e2cfab41f1b : SUCCESS in 1h 00m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/928938d2e1304d9eb62595ccbdbe916a : SUCCESS in 1h 28m 09s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/096038dc50d1433a8103e3bc20389be5 : SUCCESS in 1h 19m 41s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7e144e5092b24984a78981a2815a99e2 : FAILURE in 55m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/79538ad0e61d43d9a795c74cdc3c87c3 : SUCCESS in 39m 47s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/5cac684a8be940a995bd25186036f607 : SUCCESS in 20m 36s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/41f739ce770941cea48262b87b7c88f8 : FAILURE in 39m 48s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/02290c482c4f4c88a90a05c1c80514fc : SUCCESS in 1h 16m 25s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cdfeb2bbaef04a37927cef72d285cac2 : SUCCESS in 1h 09m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aeb0a2055b2dbda60a0c2f012d29ddb788041394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 13:34:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6983ebb4dba85abca7b09a4c1c32907a569a6f46","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 14:01:17.000000000","message":"Topic mypy removed","accounts_in_message":[],"_revision_number":2},{"id":"bc16abe00f49354fc875254b54334769f38c3ffa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 14:01:19.000000000","message":"Topic set to typing","accounts_in_message":[],"_revision_number":2},{"id":"68f93e5172932001c7c1d67bde19fff995a2a837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 15:32:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8a2681862b499ebf856e8af34993e1\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7b2d50b7468348ea9138ef39bacd3e0c : FAILURE in 24m 05s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/be6227edae844bc3bc787495503c2e12 : SUCCESS in 42m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c9a0b68ede415f8f2c0e5d4fc0e9f8 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/048b3a51ee2349f5929b4c4dd266efa0 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa0b53d3f87b42bda8d60b32c0568cd2 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74b74c1e84174004bbcac2b96e516379 : SUCCESS in 7m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ccbafec52bfa4abdb19856cf415b093f : SUCCESS in 58m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/40c8b4cc659c42d9b77a62cbe6e306b2 : SUCCESS in 3m 41s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/99884d78402f441794e9412372b1c0be : SUCCESS in 4m 52s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/fbaba37b0ac045ac9bdc280808e8ddb7 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238e61316a814212936c40a2dbf7a777 : SUCCESS in 9m 03s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9c82dca385eb47dd9bd691ce42f28a93 : SUCCESS in 37m 29s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/cb62f90370fd46b8bff0f1a4694ba4b9 : SUCCESS in 11m 31s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/f59fd9ab9b80447dbc6bc5beeea241bd : FAILURE in 19m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c82b0bc322f54816ae719f9f0db96be7 : SUCCESS in 1h 00m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b6bf48bc3e6747fc83c40220ea33d71f : SUCCESS in 1h 26m 30s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/099e6d56a7c04e76969d61eb9939e7ab : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9e09a19927ea4b47b61086c3070e0180 : FAILURE in 54m 21s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/332c76ca40304fa6a87e29d9866e0080 : SUCCESS in 39m 43s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/774e584045a04c30a9ed9792ea42540c : SUCCESS in 28m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3e4d468756d64630a4b5bed023148427 : FAILURE in 29m 33s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d7c9323a9c58431ca23b1657016c7cd6 : FAILURE in 56m 54s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cf6cee5ab7e34c308ee840229801aaff : FAILURE in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88589aa2edaf594d1e32033af287a7e0fafaf8a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-30 13:05:37.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e9fcaa65343be270c1a3a0385e48229f4bf7b065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:34:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959e7900f2d14dd79602d9a4215dcf15\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3fc4bfef119f4800b2e05e9d105f5772 : SUCCESS in 48m 45s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1525fb8dc07d4c42ab91fe1b3a273c6f : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ddcde88f1d4c44865679fbff3e8d3e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e41b153eea9442389d2c8ee90fadfbc : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/382f7f9a0ef946c5a8c525d4366cf20f : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f510e5f354c40b0b70d7e825bc207f8 : SUCCESS in 6m 31s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7cb6720b48594c3cab81539847580ce3 : SUCCESS in 1h 01m 09s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ff8faec42a534a57901408ec4cf21141 : SUCCESS in 4m 20s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0a98393879704c479aa4809f7c6fe7e0 : SUCCESS in 5m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/92066df877f643b5b173300b2268a17d : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f521f498d846e8a2442cebac338f81 : SUCCESS in 7m 58s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bb4212fc952e4eee817e4f2559cec30a : SUCCESS in 42m 17s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ef094f3b1b7c463ab2561bc6530295b4 : SUCCESS in 14m 14s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/2cc8acdc16064f6c9de959c267db78e2 : FAILURE in 19m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f4a3b1d3782422e804360fd0aac2020 : SUCCESS in 58m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/acd0115d43814d31bad98bec2a7853f2 : SUCCESS in 1h 18m 29s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e3ad2cfeb7ba42ac80de94271bc11f90 : SUCCESS in 1h 22m 37s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5dbc368fd18b4ed9b4ae135a938dd438 : FAILURE in 53m 06s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/9461095ebb5249b3800261444cde4375 : SUCCESS in 35m 53s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/cf7cacc4ff134246b687670b87a94381 : SUCCESS in 24m 40s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b949dc0365a64b519292022dc2edd44e : FAILURE in 31m 08s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1765d4a96f05427ca558d7afbf857d75 : FAILURE in 1h 05m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a4667e881bbe449bad531d27a8dc3bd8 : FAILURE in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa343567b65edf028a83dbf118c3513525aab762","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-02 12:57:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"44cc10c6e509028a703b72ae71d5b60d7c7733fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 12:57:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"754645e8631971c6c248c4cdd2a4e75fc2f4ac80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 14:16:56.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/923219 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"4af2d9f4437406a0254f9d97ca2c4d7c497d7dc1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-02 15:35:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90d13c29fdbfcfe3d178b9249186b1264723fd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 17:05:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b3d49af5934ec0952f7df8f195d58a\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/49708af8329b40819af7a7aaf9eae169 : SUCCESS in 50m 56s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2c03870ae0064608a05de68a2aa2ca9f : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce9777a6e594b4ab3c80988e8e3aed8 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27f604c033bd484c8eb215a1d532262a : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c12173b403540d7a654b8a546ad66ab : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aaaa60df8c814db99f39391970b7c707 : SUCCESS in 8m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f4177c757e914fa89f212f9b1047aa85 : SUCCESS in 52m 15s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8f394c57087e454da27afcba1fafea78 : SUCCESS in 4m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1306255021214e398d7a4146dcc571d7 : SUCCESS in 3m 56s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0a59a730d6d0421d8d39ab046c90de7e : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9782b25d4e0245cc92f6dd9779cef02d : SUCCESS in 8m 32s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5aa69e4bebf6425e863b66433ea5e498 : SUCCESS in 46m 45s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f048fcc99d2340efbd1fc03fb350a89a : SUCCESS in 12m 00s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/c78233df0fca423780a3d988ddeb995d : SUCCESS in 30m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1182b3dc3844615b10efb7e199d0807 : SUCCESS in 46m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/255078d49cb14e51a620a72d375efc61 : SUCCESS in 1h 22m 53s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/bf93b9c110284f71a1fe0c74ac34c4ed : SUCCESS in 1h 25m 03s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d5b79f1fa6914660867ccdb5ffa70e09 : FAILURE in 1h 00m 34s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7ddd4faf402640a69b15dab4a2dca793 : SUCCESS in 26m 42s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0a418b34740f44b786c080d2b81e10ff : SUCCESS in 36m 06s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ab70486ff97a4f7eaf8c4b13858c8761 : FAILURE in 39m 27s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a25a4486385447f982e4742af78bfdaf : SUCCESS in 1h 11m 29s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/71bdb416a87e4f8c807b7c588f3f6e0e : FAILURE in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f8a541db2a189b4e2d2ddb89a1f256da1fbcdb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 19:04:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"878b218b38dd248d7e12abd8802f8379df44bc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 20:39:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b33b449d46544e5b3e4fb7770d6108c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad657ce37d0421784aef573e48fcd32 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4e10bda60bd4458a551931401c932cd : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7651ed2a9faf41fbbf902aeda53197d9 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/420fb21072704c88973dc1ba31538c65 : SUCCESS in 54m 21s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/297b85f213ee4be28789f0c3fb94b645 : SUCCESS in 3m 37s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6e0c2d7c4364468d85ab921ca8300bf0 : SUCCESS in 4m 48s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a4bb50b2eab44598848997d2e8f59502 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2f91790bc2454580b7cd7c8a157e7a : SUCCESS in 8m 36s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/87c4caa018214e8d8f19f6d02374d536 : SUCCESS in 2m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d394a160e6554350a456549b1075fc11 : SUCCESS in 44m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7d2aecefba43475cacc94df5a0ab1d7a : SUCCESS in 1h 27m 01s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/b8be961e2475491f9a2be77b120dc2fe : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":3},{"id":"85b613f25497d9f0560195238812733b78a6dde5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 20:39:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cb6962d925f61e04d22fc47a86bd7152b59b3b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 20:40:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33cd0aea62924126b1e6c899cff9adcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f7466b37b84bc9acfe7ef2e8b5a641 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4b9a44a1e7ce6c28e399e981456d6b5dab0f58a","revisions":{"14797fd9bd5d3c65722f3366b7baea4c03656cfc":{"kind":"REWORK","_number":1,"created":"2024-07-02 11:15:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/1"}}},"commit":{"parents":[{"commit":"a110911ddfc11c46e03ba370608306bb3fb4c991","subject":"mypy: Enable checks for Ansible tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a110911ddfc11c46e03ba370608306bb3fb4c991"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 14:59:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 15:00:54.000000000","tz":60},"subject":"tests: Remove TestClustering functional tests","message":"tests: Remove TestClustering functional tests\n\nWe remove the mixin that would have added these methods recently, in\nchange Iee56f6841355bebe1bc21a05bd5ce594f2d7b4cc, meaning these tests\n(which are currently skipped) cannot pass now.\n\nChange-Id: I6b74ad9cd8f32793d24c89caec5b08347d7d563f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/14797fd9bd5d3c65722f3366b7baea4c03656cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/14797fd9bd5d3c65722f3366b7baea4c03656cfc"}]},"branch":"refs/heads/master"},"64bcf100457589c7698454e51d0d90cf73392d9c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-24 13:34:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/2"}}},"commit":{"parents":[{"commit":"94a30f1389639aba6663266a4d36c1a09b770ed8","subject":"mypy: Enable checks for openstack.cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/94a30f1389639aba6663266a4d36c1a09b770ed8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 14:59:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-24 12:19:59.000000000","tz":60},"subject":"tests: Remove TestClustering functional tests","message":"tests: Remove TestClustering functional tests\n\nWe remove the mixin that would have added these methods recently, in\nchange Iee56f6841355bebe1bc21a05bd5ce594f2d7b4cc, meaning these tests\n(which are currently skipped) cannot pass now.\n\nChange-Id: I6b74ad9cd8f32793d24c89caec5b08347d7d563f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64bcf100457589c7698454e51d0d90cf73392d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64bcf100457589c7698454e51d0d90cf73392d9c"}]},"branch":"refs/heads/master"},"c4b9a44a1e7ce6c28e399e981456d6b5dab0f58a":{"kind":"NO_CHANGE","_number":3,"created":"2024-07-30 13:05:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923223/3"}}},"commit":{"parents":[{"commit":"5704b3ef767d7be7d592c9dd62a4d4c63c9cac42","subject":"mypy: Enable checks for openstack.cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5704b3ef767d7be7d592c9dd62a4d4c63c9cac42"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 14:59:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-30 12:48:12.000000000","tz":60},"subject":"tests: Remove TestClustering functional tests","message":"tests: Remove TestClustering functional tests\n\nWe remove the mixin that would have added these methods recently, in\nchange Iee56f6841355bebe1bc21a05bd5ce594f2d7b4cc, meaning these tests\n(which are currently skipped) cannot pass now.\n\nChange-Id: I6b74ad9cd8f32793d24c89caec5b08347d7d563f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c4b9a44a1e7ce6c28e399e981456d6b5dab0f58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c4b9a44a1e7ce6c28e399e981456d6b5dab0f58a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
