)]}'
{"id":"openstack%2Fopenstacksdk~923226","triplet_id":"openstack%2Fopenstacksdk~master~If92f5a3b4348cb2e97115fcb4c8d309a7fa7c40f","project":"openstack/openstacksdk","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-24 14:54:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If92f5a3b4348cb2e97115fcb4c8d309a7fa7c40f","subject":"mypy: Enable warn_return_any","status":"ABANDONED","created":"2024-07-02 11:15:45.000000000","updated":"2024-09-24 14:54:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19493cd934f9825bb4144a4833ba470992e4db5c","_number":923226,"virtual_id_number":923226,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-30 14:42:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 12:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb05a54b86f4583a61609537b741ff30a5a7284e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-02 11:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f4373b398b1b2ed4aa8e8727340167a87eeb1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:23:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a376d56d88b84976ab93f6ea8dc6236d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/064f1ba99e994386afd560dfe54ea396 : SUCCESS in 2m 06s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/28d366ec28da4235a3e44140d33ebae6 : FAILURE in 18m 08s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f7faa28887b6476ba9ebd6445437818d : FAILURE in 9m 03s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/5104f18664394b08875aacb15980d015 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83aa08b514d445d4bfed165878047350 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d06e78ee51094e908c97ed6b22c5aa3c : FAILURE in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/521e988f81aa40f694e6a5a425c759ed : FAILURE in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d6fcaf7fd8d4f59864cb209614a1853 : FAILURE in 5m 12s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/23f34cab9e2d4ab7a5f6bb69c2fcd1d2 : POST_FAILURE in 18m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/31fbc7efd8a64898a20acead599db1a3 : FAILURE in 3m 13s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a43ff7ff627b4eda84e9fcd5ceb30f1c : FAILURE in 3m 45s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/19f2bc8c5eb4417a8eb2d790c0bfc977 : FAILURE in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f586f3603a284a20af17d5f79a2c847a : FAILURE in 6m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3226b2a710364f69a8c5ef33855cf64c : SUCCESS in 56m 10s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ab8096de205c43e192265a436d0db25d : SUCCESS in 15m 52s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/5da0ec1cc3104918857af19c593b2aff : FAILURE in 19m 29s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69fd26a9534949c581e034a3ddd92a5f : POST_FAILURE in 16m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a563cc8ebff7409a83293c6da2c4f8b1 : POST_FAILURE in 18m 27s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/08cd624fee14424182ecd8e7de1464b2 : POST_FAILURE in 17m 21s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/8eb7ccbff437492bad7d0e48cfef959b : POST_FAILURE in 19m 53s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b8e41bb496a84e4e9f22355ddb99fb31 : POST_FAILURE in 16m 32s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ffcc818e79614cbeaec9d43977df9bfa : POST_FAILURE in 22m 46s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cedd832bd75f4c89bbf972a15a232583 : POST_FAILURE in 18m 29s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/181ceb44347c402ea62d7b40849cd0a3 : POST_FAILURE in 21m 02s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c685268d2712462390f78ad4d5e332ca : POST_FAILURE in 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f2c2ef48d6fbb2fcc57a8c909f767d83945f436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 13:34:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f278c45d5006ff89a1c06743ffdb477d737032b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 14:01:26.000000000","message":"Topic mypy removed","accounts_in_message":[],"_revision_number":2},{"id":"275b48632e6b567f92505ad8364bda42bbb12287","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-24 14:01:28.000000000","message":"Topic set to typing","accounts_in_message":[],"_revision_number":2},{"id":"ac129bf49bb7f8eb09bdbfb7a93f66f79a1e5faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 15:13:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da3f39046e564b5dace43b4e5ad6a204\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/07cdf390082341a2839cac0a735bc4ee : SUCCESS in 2m 12s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/00d04520639b49b58d68aca60db80056 : FAILURE in 16m 33s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3d6506e9986a459c9c79ffc47457cfa3 : SUCCESS in 42m 12s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/22b0565070fa4d74a81f8c412d5c2418 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ceec2b6e8d34723a779c7bd5ac7c2f0 : FAILURE in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08b404c99db140ee9a65896a59757cc9 : FAILURE in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2476c1d226424910b2cf4470895308e4 : FAILURE in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/093c5c4bc8b546f5b1f8095cf33e39ac : FAILURE in 7m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d402a781ea434d8999c1859fb8b461d5 : FAILURE in 58m 17s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b54df44c04e54309852733f9e3b38781 : SUCCESS in 3m 39s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/afa6de55c3ac4612b0e9946681fb7040 : SUCCESS in 4m 15s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/93e34b3766a64e6182547b2ea740ec61 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160241625aa641de8e02421a70f50df0 : FAILURE in 6m 43s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a369a31056314a5cad1b691cfc4b8836 : SUCCESS in 44m 06s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b9a781d0333d4452b9b2b834960574ad : SUCCESS in 13m 36s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/84fabb80839a46659b304a6f0d4d63fe : FAILURE in 21m 49s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df60935b7bbe474faa5f7cf8dee6bf2a : FAILURE in 54m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1762a895d85a4b85a8fbc25837c49fd7 : FAILURE in 1h 22m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1cf04b68b2e0498fa68f2e0b7aba092d : FAILURE in 1h 21m 39s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/17c08e885a84401cb1a3ce55ce25dba7 : FAILURE in 42m 12s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7efa3fc9646f4fe88d657e00d67f6eb2 : SUCCESS in 36m 32s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d73e98bc72f044058ababa8aa5d1a6d3 : SUCCESS in 31m 00s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/006ae37086804c19aea5e3d02afd9577 : FAILURE in 29m 36s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/467c822dcbe0494cadd87ac03b7e3858 : FAILURE in 59m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ef073267254242be9fc2abcc9f7c7fd6 : FAILURE in 42m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file openstack/block_storage/v2/_proxy.py\n  Comments left for invalid file openstack/block_storage/v3/_proxy.py","accounts_in_message":[],"_revision_number":2},{"id":"d32d0499f80ddff5ae5b782980d6aab53a29ef91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-30 13:05:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c0c08bb5e120ed929480801dfba313e062f526b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:42:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/029ced9bb37a4bd3ae386d369c3253d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2550fdb57aed417d9fcd798c53e6243b : SUCCESS in 2m 55s\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/83b348b2b4404027b71c46727f4f51d4 : SUCCESS in 49m 32s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/baf96f47435844f688f8a3f34dc6a1ec : SUCCESS in 38m 51s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/be1766e946f34795b2a11f9d3b9ec1fc : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39038c0a73924a2ead9f27c795c5043e : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de1710624c2d4e1c886089a76fc3d32c : FAILURE in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7552a1347554da984243818b1d11df1 : FAILURE in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e46b35ad4ca94bc4b1d387ed41ccca82 : FAILURE in 7m 52s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7c08908ba2694d5e88cf7f2af9f3e517 : FAILURE in 55m 02s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/54cb9413e3784bc4856e5374ef329ecf : SUCCESS in 3m 14s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/253c835bcc23475f8e588ffbdeabac65 : SUCCESS in 3m 59s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/bd12b298b072498ab64eca84a3ebdc5b : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9685b96d03743aaa8980f9946f1cab1 : FAILURE in 5m 36s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7a184817a5ea4beda554421c5ad9b3f0 : SUCCESS in 38m 10s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/dc6d95fd418946989ecd27d2e2165af5 : SUCCESS in 15m 11s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/2694d1ab18da4528b7f05766aa6d9b81 : FAILURE in 18m 18s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e392488d8bb4a1aade8b4ae4b286f15 : FAILURE in 54m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5825a4b73e54abcbfa61f56522b8b1c : FAILURE in 1h 15m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e8dd3f71b5194370b17b12735a6cd294 : FAILURE in 1h 28m 04s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/803ad8ce12cc40efad911355269a56a9 : FAILURE in 51m 25s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/0f528085f3e245ce901a23f09fd2666f : SUCCESS in 25m 09s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/feaf6c13cac049918ab1c5c1efc0808d : SUCCESS in 25m 45s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b3bac1ccc54341af896441e29eea24f4 : POST_FAILURE in 23m 45s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e5e6b3b27af24b60975c4d16c5a3b145 : FAILURE in 1h 10m 45s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/19609675f84b40eb8743563e55eca332 : FAILURE in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"19493cd934f9825bb4144a4833ba470992e4db5c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-24 14:54:22.000000000","message":"Abandoned\n\nI\u0027m tackling this elsewhere","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f67827b8331311bce6d8a42940f832f15822acd9","revisions":{"6f28e44e485dc16b1f327274c239b124c0234206":{"kind":"REWORK","_number":1,"created":"2024-07-02 11:15:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/1"}}},"commit":{"parents":[{"commit":"0327de0128f25cb37b3c1d3b56b80c9ce6bc0ba6","subject":"mypy: Centralise disable list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0327de0128f25cb37b3c1d3b56b80c9ce6bc0ba6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 15:57:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 16:58:02.000000000","tz":60},"subject":"mypy: Enable warn_return_any","message":"mypy: Enable warn_return_any\n\nMake things a little bit stricter.\n\nChange-Id: If92f5a3b4348cb2e97115fcb4c8d309a7fa7c40f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f28e44e485dc16b1f327274c239b124c0234206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f28e44e485dc16b1f327274c239b124c0234206"}]},"branch":"refs/heads/master"},"6ad0131d4350f7c05adf7dde37b46436409f2a0f":{"kind":"REWORK","_number":2,"created":"2024-07-24 13:34:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/2"}}},"commit":{"parents":[{"commit":"b1041f6fd64f58535b7d857be3846c62a64f5c51","subject":"mypy: Enable checks for openstack.tests.functional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b1041f6fd64f58535b7d857be3846c62a64f5c51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 15:57:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-24 12:19:59.000000000","tz":60},"subject":"mypy: Enable warn_return_any","message":"mypy: Enable warn_return_any\n\nMake things a little bit stricter.\n\nChange-Id: If92f5a3b4348cb2e97115fcb4c8d309a7fa7c40f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6ad0131d4350f7c05adf7dde37b46436409f2a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6ad0131d4350f7c05adf7dde37b46436409f2a0f"}]},"branch":"refs/heads/master"},"f67827b8331311bce6d8a42940f832f15822acd9":{"kind":"REWORK","_number":3,"created":"2024-07-30 13:05:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923226/3"}}},"commit":{"parents":[{"commit":"fc18b74c9ebfe2dd1cbcd1f138c5ae4656605862","subject":"mypy: Enable checks for openstack.tests.functional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fc18b74c9ebfe2dd1cbcd1f138c5ae4656605862"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-01 15:57:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-30 13:05:31.000000000","tz":60},"subject":"mypy: Enable warn_return_any","message":"mypy: Enable warn_return_any\n\nMake things a little bit stricter.\n\nChange-Id: If92f5a3b4348cb2e97115fcb4c8d309a7fa7c40f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f67827b8331311bce6d8a42940f832f15822acd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f67827b8331311bce6d8a42940f832f15822acd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
