)]}'
{"id":"openstack%2Fopenstacksdk~923379","triplet_id":"openstack%2Fopenstacksdk~master~Ib92913204bd90c2fb56268f58b089a2d3ba45644","project":"openstack/openstacksdk","branch":"master","topic":"bug/2071784","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-03 21:42:23.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-07-03 21:42:23.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-07-03 21:42:23.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-03 15:47:36.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Ib92913204bd90c2fb56268f58b089a2d3ba45644","subject":"Run auto_allocated_topology functional tests in the own project","status":"MERGED","created":"2024-07-03 10:01:34.000000000","updated":"2024-07-03 21:43:33.000000000","submitted":"2024-07-03 21:42:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"923379-bug/2071784","meta_rev_id":"a1b061bafb7fc52e01fae8c50a914b2e94e8d7ac","_number":923379,"virtual_id_number":923379,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 21:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-07-03 15:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-07-03 15:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2024-07-03 17:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 11:17:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-03 11:56:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-07-03 12:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 12:34:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-03 12:34:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-07-03 12:34:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-07-03 13:35:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"7a5a975ec19a5fc01a6ca2d472c288c1e9ed6994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 10:01:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c96b2a1e7e495d2933d743cafe9e0eee5c07f11","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 10:02:43.000000000","message":"Topic set to bug/2071784","accounts_in_message":[],"_revision_number":1},{"id":"f3cd7458d96121ffca8512c868ce2ad4dc25903e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-03 11:17:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24b06967ea666485676afaae4f84c857d8b13f39","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-03 11:56:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab0dceb7439793c835857e0c7142534a49f0b167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 12:18:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76f80124f304491fb074ade8dce299b8\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3279ac249759416499233297804af022 : SUCCESS in 50m 44s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/38ff0293fa154c9abea5b80761e1a5ed : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01d71c5a3234e67b758bac18e43dbf5 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d466dec1f5d4850acef2c45d2e55545 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33104705d0774cf0beed222b52897fb3 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30dff1509a9c4012a05e94078bddbf9c : SUCCESS in 7m 29s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/64e39b2a929544b1a646949ff65d1b32 : SUCCESS in 59m 40s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d6450f8ead0c4307a11b89db4d4340e8 : SUCCESS in 4m 27s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/56591f385d544fcbab1173687e6fb426 : SUCCESS in 5m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0fa3ad0121574773944e67c9c2ad4aa0 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00f6ac719444c5a9ec05c5e522a32e8 : SUCCESS in 9m 26s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b377b8fd6daf4745b7f8b6af1f98e4be : SUCCESS in 56m 56s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/9db48506d59a4f939d6836ebc74c5e7e : SUCCESS in 13m 04s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/da1f3e79a0e9458fbc439ca0be1594ba : FAILURE in 20m 24s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25cc06f97ea9486ea3408ae3d965aef0 : SUCCESS in 1h 03m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c644a5ab739743f4a8c082e026b834c6 : SUCCESS in 1h 39m 20s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2f0de7097b654995b2c7b8070547e101 : SUCCESS in 1h 29m 37s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a9347525c135449481b8d69dc802f231 : FAILURE in 52m 27s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/40501ba8cbc84db8aa3fe8c1a4d9792e : SUCCESS in 26m 45s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/89aa460ec6fd4385ace24d5b82301d1b : SUCCESS in 22m 42s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c1b25007072641eba3aad88783c0c71c : FAILURE in 30m 07s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/07b0f887c17c4635825c2198c61b577d : SUCCESS in 1h 11m 12s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2529647f5888416da56a0aa3e629072d : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f390a00f3a206e394d58b766cb6acd60f149e16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-03 12:34:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cda0c9d313e17c2d5b4650a0692c79698afd7f31","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 12:37:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6956f1b865b91ccde5af514bc8d59854869632a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 12:41:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a53be90e875290b4ca26f721f0d8b3ee4af2c7ac","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-03 13:35:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"600844ad9a7c69b8fe784fea9f7687e200a594db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-03 13:40:37.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e39e795fa86ea4617d93be610f465bbda2d9f740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 14:00:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ef3462801a5d9f2259c81eb6c2af78cdbf53dafe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 14:00:36.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ddcfdd508383dd76a74e1e20e7fbc832c832f17f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 14:02:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f259230678323a8aeec85fd8f2834ffb6c90ef78","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-03 14:11:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08f892a467075488c47b7bae80e5d766452ee626","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:42:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21e11ab8dd57a0b4b8948fca621a591b80845a50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:44:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4abcb71626b2c6ea752513ff8362a01068235f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:45:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d3049bec94ff61751c16187e059f7fd011dda49","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 15:47:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb30d4ef3ab50075d8614694989359e236ccfd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:43:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e702e4346cc4034b57a68ab5621585b\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b3bb376a478d4c5a9235c02bb9257bbc : SUCCESS in 1h 06m 45s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/67a8eae17e9648dcbb13512613b8e9e3 : SUCCESS in 41m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac334ce8291482ca8354d88ab0e3006 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca781bc8e10b494eb92310a58dabcd83 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98a8286ab1ec4e189014aefa4e017680 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dae3dbbb99344955ac8cd22f3cd321f3 : SUCCESS in 7m 01s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f216f313c3de42c8b062b319b39db8be : SUCCESS in 54m 15s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3a2bf4466d59426ca33f9b84b28dd4ba : SUCCESS in 3m 04s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/01e5c1ccaaaa4475a83fc4a9d4d69674 : SUCCESS in 4m 38s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/563df883f6a84016a7101a663abcb5a9 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73043952eff543b3b0536db3991730d7 : SUCCESS in 8m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e5d6cfd010f4440da62b254295f3e15c : SUCCESS in 56m 36s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/b7bff709f47e44fb9f54d01086abbbb5 : SUCCESS in 11m 47s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/66433afbd39d4b279907fe25b48ad525 : FAILURE in 23m 07s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/148324725c8446d9a4d36c3dd9c2c866 : SUCCESS in 58m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/959cc51cc0cd4028a6971488c68e7291 : SUCCESS in 1h 24m 02s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2ddca9f98b8148b29391fd5d392b503f : SUCCESS in 1h 32m 03s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/74413caee2404e1d83d6bf36a24b0b5c : FAILURE in 55m 59s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/09eb87e4186f44f88dcd186df192ffe6 : SUCCESS in 28m 15s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/24a67147a4bf48a9943ebb8f778e9ab3 : SUCCESS in 30m 49s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/488108bfa8e646f9b670df9ea5eeb8ec : FAILURE in 39m 23s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7f0c6f1c0205435b98e12f0974bc69cd : FAILURE in 1h 09m 47s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/838a053cf11649ef9c1ee13c5e008470 : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61f3ed9e61d5f37103bd5ca2df167cdf52be09b7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 17:46:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04a7ffe6dd60d8bfd975fb3f3e9ee55a3e3ef800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:48:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9883d489cd95c1bd2e7dadda3394ee455ddcc944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:42:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b03810dc57f7472ebf6ce3931462619c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f85a4454a44d4cb3d859fda374f374 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d446de66e03472ab98c1289a98284ec : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd0b4fda3f4d48e5ab08b78b005b3914 : SUCCESS in 4m 08s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e7e8a90ab82f4950a753d1f4afea8a60 : SUCCESS in 1h 00m 26s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/087c793fb07e4aaf9daa70293071e8b9 : SUCCESS in 4m 19s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/06d9cd702c344302b77c18c9ef546d31 : SUCCESS in 4m 27s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/70d1cf2f363e422082331e29070a790f : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7d3227bc3e4676ba1afeb2c3d4920c : SUCCESS in 9m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ca4f3c26e6734ed0a1b8842503dd9fd2 : SUCCESS in 40m 53s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a1c2bcdc0cf0453ab316b5d1d1682104 : SUCCESS in 13m 01s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/b2a1b93107b445b89172a398bd246e00 : FAILURE in 19m 19s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a126f18df00486e8c726f91087b56e6 : SUCCESS in 48m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7bdab0057e2b4e18a12307773f6b0c90 : SUCCESS in 1h 26m 42s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/627c936987f141278038c6c4f664373f : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":3},{"id":"fc9967db0949d9a7d6da6489dbf0aa74dbdb6f6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:42:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a1b061bafb7fc52e01fae8c50a914b2e94e8d7ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:43:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00af3500e74143eb8508077bd9da2696\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a101c539238e4c6287444b5163436c41 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3ee18eb1711c6e76188b58c0eb520443fd34750e","revisions":{"db6872f258500b970d0d961ad69ce4766c10e7fd":{"kind":"REWORK","_number":1,"created":"2024-07-03 10:01:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/923379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/79/923379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/1"}}},"commit":{"parents":[{"commit":"10e5e20fc0eb5264080f0cfdc0523d65883dc693","subject":"Merge \"compute, block storage: Add support for quota class sets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/10e5e20fc0eb5264080f0cfdc0523d65883dc693"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-03 09:58:37.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-03 10:01:31.000000000","tz":120},"subject":"Run auto_allocated_topology functional tests in the own project","message":"Run auto_allocated_topology functional tests in the own project\n\nTo avoid accidental usage of the \u0027auto_allocate_network\u0027 which is created by\nNeutron automatically when \u0027get_auto_allocated_topology\u0027 API is called,\nby other tests those auto_allocated_topology tests are now run in own\nproject which is destroyed after tests are done.\n\nCloses-Bug: #2071784\nChange-Id: Ib92913204bd90c2fb56268f58b089a2d3ba45644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/db6872f258500b970d0d961ad69ce4766c10e7fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/db6872f258500b970d0d961ad69ce4766c10e7fd"}]},"branch":"refs/heads/master"},"a83ba99dd6b7ed2f46ee2fde0ecc6ddbfdee32ff":{"kind":"REWORK","_number":2,"created":"2024-07-03 14:00:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/923379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/79/923379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/2"}}},"commit":{"parents":[{"commit":"10e5e20fc0eb5264080f0cfdc0523d65883dc693","subject":"Merge \"compute, block storage: Add support for quota class sets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/10e5e20fc0eb5264080f0cfdc0523d65883dc693"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-03 09:58:37.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-03 13:55:49.000000000","tz":120},"subject":"Run auto_allocated_topology functional tests in the own project","message":"Run auto_allocated_topology functional tests in the own project\n\nTo avoid accidental usage of the \u0027auto_allocate_network\u0027 which is created by\nNeutron automatically when \u0027get_auto_allocated_topology\u0027 API is called,\nby other tests those auto_allocated_topology tests are now run in own\nproject which is destroyed after tests are done.\n\nCloses-Bug: #2071784\nChange-Id: Ib92913204bd90c2fb56268f58b089a2d3ba45644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a83ba99dd6b7ed2f46ee2fde0ecc6ddbfdee32ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a83ba99dd6b7ed2f46ee2fde0ecc6ddbfdee32ff"}]},"branch":"refs/heads/master"},"3ee18eb1711c6e76188b58c0eb520443fd34750e":{"kind":"REWORK","_number":3,"created":"2024-07-03 15:44:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/923379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/79/923379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/79/923379/3"}}},"commit":{"parents":[{"commit":"10e5e20fc0eb5264080f0cfdc0523d65883dc693","subject":"Merge \"compute, block storage: Add support for quota class sets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/10e5e20fc0eb5264080f0cfdc0523d65883dc693"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-03 09:58:37.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-03 15:43:47.000000000","tz":60},"subject":"Run auto_allocated_topology functional tests in the own project","message":"Run auto_allocated_topology functional tests in the own project\n\nTo avoid accidental usage of the \u0027auto_allocate_network\u0027 which is created by\nNeutron automatically when \u0027get_auto_allocated_topology\u0027 API is called,\nby other tests those auto_allocated_topology tests are now run in own\nproject which is destroyed after tests are done.\n\nCloses-Bug: #2071784\nChange-Id: Ib92913204bd90c2fb56268f58b089a2d3ba45644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3ee18eb1711c6e76188b58c0eb520443fd34750e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3ee18eb1711c6e76188b58c0eb520443fd34750e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
