)]}'
{"id":"openstack%2Fopenstacksdk~923420","triplet_id":"openstack%2Fopenstacksdk~master~Ic3adaa90d80093b3ae6f8227a6c218ef4174a830","project":"openstack/openstacksdk","branch":"master","topic":"use-proxy","hashtags":[],"change_id":"Ic3adaa90d80093b3ae6f8227a6c218ef4174a830","subject":"tests: Remove errant print","status":"MERGED","created":"2024-07-03 15:39:14.000000000","updated":"2024-08-07 12:13:52.000000000","submitted":"2024-08-07 12:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923420-use-proxy","meta_rev_id":"3edb402323b44ac4656f3748bc1d52293b56cd25","_number":923420,"virtual_id_number":923420,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 12:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-08-07 10:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-08-07 10:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 17:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47dc82ac4456fde60b73fd92415dd9ef50507ea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48099f40f1d32c1e9f1ef61474de8ae67dbd7191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f655a0f3fb684d41822c5c6d8987785c\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/627ad41a70914844a3c53b22dd3cf12c : SUCCESS in 48m 23s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/53eff57421ad4a539abd7545ff32fe76 : SUCCESS in 42m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f78401359794c4cae3b441eadca5c43 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db6a6b80e97d4ae0aa52693be3de1439 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8f71ec2e9dc4aa187b47ebc075b384b : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ad6b1811a7b49ada281386a682de4ed : SUCCESS in 6m 54s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/68e771ae80034fc1aff7d8f66db06c63 : SUCCESS in 1h 09m 06s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3e6072460d054594b4aea9bd401b6114 : SUCCESS in 3m 58s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/709af3a857d944feb9267b96b0697931 : SUCCESS in 4m 36s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f1fc50c0a1514d42a0b9606dc85ae670 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4d759d31e14ddf89545b6a157a6a86 : SUCCESS in 8m 23s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0e5cdbeee5684a7084a63d333d6a362c : SUCCESS in 1h 08m 47s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/d58e53f992f94b16bc7e95d660d42bb1 : SUCCESS in 12m 39s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/22c3388b63084631a9820d40a66a98be : FAILURE in 17m 51s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c82aa257d975471aaca96c467d46c943 : SUCCESS in 1h 04m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d15e6681ca774820a79cd588dd273545 : SUCCESS in 1h 23m 51s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/09e7be8ae35945318d534e763fb772e6 : SUCCESS in 1h 22m 41s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/32db53731860410a9b8e9788e1f1b341 : FAILURE in 1h 10m 43s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/113cfbac644f4fabaf970838a456f50e : SUCCESS in 27m 25s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1f9e503a6421443cbda04b9ba92caa9f : SUCCESS in 27m 02s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d3c8637911d649c9b399ac97c8e38dbe : FAILURE in 29m 27s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c43570adf9b94e25b07a487f17ae1d45 : SUCCESS in 1h 19m 40s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/59013a8e9a2e4302ba9839c9f21219ee : SUCCESS in 1h 04m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d26b934c7a136deedc634200ad072a00e5c6124","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-07 10:43:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b396cf76515db431ec9ef2944f6e286044ee282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34456718ea8a98283889491f351afad91540c73b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:11:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2514f429e24971ba997888c8d0281f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6826da3c29416489d2762025c1249f : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ba1208908d4e8ab9ff7c0b5b8cc349 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9353143f1c240a0ac10625d87f44339 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/719d66ccdee84540aebbcbabc9312523 : SUCCESS in 52m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a516c4508b7947d29204f78926e27788 : SUCCESS in 3m 26s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ebe5ca7839ad4b52983bf6fe4f03b57d : SUCCESS in 4m 16s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/165b4f109a7847f9a63040db196b390d : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb247127d6d04763a29d13240130f68b : SUCCESS in 8m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2ac61416f1f0442f98e9036ec974ba68 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ad604d063ea429b8bfa590bb8772bde : SUCCESS in 49m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/849c5126c3674206bca4aac9b047ff2f : SUCCESS in 1h 16m 57s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/63021659438144c2a6a054662ad03ccc : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e3306a28ee8c28000190698bd65225cc0da131e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3edb402323b44ac4656f3748bc1d52293b56cd25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:13:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ae912424fe4ef785bae954838a9353\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a203a953ce4271a92ee75f5984349e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a8c12a80b883a314ff3bcbd2874fbed698b69ab","revisions":{"4a8c12a80b883a314ff3bcbd2874fbed698b69ab":{"kind":"REWORK","_number":1,"created":"2024-07-03 15:39:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/923420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/20/923420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/923420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/923420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/20/923420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/20/923420/1"}}},"commit":{"parents":[{"commit":"10e5e20fc0eb5264080f0cfdc0523d65883dc693","subject":"Merge \"compute, block storage: Add support for quota class sets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/10e5e20fc0eb5264080f0cfdc0523d65883dc693"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-26 12:52:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-03 12:29:07.000000000","tz":60},"subject":"tests: Remove errant print","message":"tests: Remove errant print\n\nChange-Id: Ic3adaa90d80093b3ae6f8227a6c218ef4174a830\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4a8c12a80b883a314ff3bcbd2874fbed698b69ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4a8c12a80b883a314ff3bcbd2874fbed698b69ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
