)]}'
{"id":"openstack%2Fopenstacksdk~923423","triplet_id":"openstack%2Fopenstacksdk~master~I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e","project":"openstack/openstacksdk","branch":"master","topic":"use-proxy","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-25 09:10:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e","subject":"cloud: Switch remaining block storage functions to proxy","status":"MERGED","created":"2024-07-03 15:39:14.000000000","updated":"2025-02-25 10:55:27.000000000","submitted":"2025-02-25 10:54:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923423-use-proxy","meta_rev_id":"e01772fae445762c9006fb786469ed315bbea009","_number":923423,"virtual_id_number":923423,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-25 10:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-25 09:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-25 09:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 17:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a571dafd981b2adda68bcea5b9357d115943d405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a587dad74400bdfca50cdf14ae27ea74c214ec9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:22:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3fdadbf46df4a6582865872c93b28b9\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4f18c9f3356f47618e064cfd3085bf4b : SUCCESS in 44m 15s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b564c1501217462d8bd7cb9f9ae055e4 : SUCCESS in 47m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e48bc31cdd48459bc8d6a906bc684c : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c206483d4da24f66af67023a0e4f4e9d : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f714e48087645ee8f8d9cfa72fe91ed : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a6bcec8b78430e9fbb6272fcd7d3af : SUCCESS in 10m 18s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5563c20cab2d46dca872fbfd7e0efb72 : SUCCESS in 1h 03m 06s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1535e8766e424b0bb8716ff0e27b46ba : SUCCESS in 4m 43s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d037d54f5046440d90b053b6c84af114 : SUCCESS in 4m 39s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/8bfd6229aa8d4359a9a26f3fcfc791fc : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b1f11b73304979b8289148bff2d4ee : SUCCESS in 9m 17s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7c5d13d7d5f64ad98097cec138092a60 : SUCCESS in 1h 05m 13s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/2f0cec45c1bc477e8e4e2c9a1331e360 : SUCCESS in 12m 47s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/9f6520d6000f499c8d2f3bd805b02769 : FAILURE in 21m 28s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10571105891940f3b65f9cfed40b3e2a : SUCCESS in 1h 15m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/625a8a7dd4144b2ba41008930f583cd9 : SUCCESS in 1h 27m 33s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/131e7df1fa8343358c281bbdf671cb44 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4fe2704a266f4d238d983df980d8da45 : FAILURE in 1h 00m 04s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/be5fdc4954624995882bc189146cc30f : SUCCESS in 35m 50s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8ab539f44bc24e949144d6202c660f3a : SUCCESS in 30m 22s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/16bca17d6ffc4755b18c91746c62b5a4 : FAILURE in 36m 49s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/68c60401afca467da2b05bd35ff78eaf : SUCCESS in 1h 09m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ed078771f44841498524e3d00d39ea5c : SUCCESS in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29d67967b3e831032f85e6f6bae5bc09acff2fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-09 10:02:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"28996d0553bfeaccb964400c541f308cde3ec9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 11:29:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b880dd832d3461fab5565aad6aedca9\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4a54f7785be64a3f92d566ebc2137639 : SUCCESS in 50m 53s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8a7390510efd478498d17f80e070bf71 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1ad459bd464e4ca412209122466c15 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97562be4a0b54f5a852e44e38d0c49a6 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fce5cfdc044f4c8ca37674d8814c19a0 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d93f1d0d7bb44d6f893b14327cf88c39 : SUCCESS in 8m 51s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4f2c8c5fb39247469d9d6282fc952785 : SUCCESS in 50m 28s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/799b98099b754f3cb163c731c3953487 : SUCCESS in 4m 48s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e4f19cbadd144146bb7ef9b53c467faf : SUCCESS in 4m 16s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/90771d81ad894c4684f45dccb8251a82 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9aef4dfe05640eca1762cedea4d7065 : SUCCESS in 8m 34s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/640c01ae1cf245f79bc785d7887b0d02 : SUCCESS in 1h 00m 34s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/7f858cde965c48878ffc8c9c8cef6e06 : SUCCESS in 13m 22s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/3d2697cd81d54bd29b14f401ba028c55 : SUCCESS in 31m 44s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c1eedc6f2448ba89f0c3e81b64abce : SUCCESS in 1h 00m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/996a97c6bfbc4110b5617e0ec0af6b9a : SUCCESS in 1h 16m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/33a3a9eda2d14d56b522869f6765c6e2 : SUCCESS in 1h 25m 11s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e1db3251144b4ea38e849ec3c18d48fd : FAILURE in 57m 31s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d51d625959c348ff9fd6443819f42b07 : SUCCESS in 34m 10s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0cfbe328dc6845ce91129340be5c4c61 : SUCCESS in 30m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ac8756ca0eee40c2b406e0fcc5989bc9 : FAILURE in 35m 48s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/eabc70ffbf0a4d169c4452e2a5eb0a09 : SUCCESS in 1h 08m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/42fc3d75b6d54ab9a806fa237ad986dd : FAILURE in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f99168c37a65baa18e7629ccac36aa053ae5ebb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 11:29:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e5827409838b6476391342c14773dfd82aea7907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 12:53:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32933b0f8163463885019e84eab78562\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/10403c5a4508416380afa7a077ca6b00 : SUCCESS in 57m 17s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ac336bcca37d42b6a32b7576f351c923 : FAILURE in 43m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee86bbecde844551ab849368bcc606a8 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d9505a738904225bd6c2b73551a5b8f : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff43c7dfa77847889c8119e634d87538 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/096f5d6a3a854b72bfc9bea6b23c5380 : SUCCESS in 4m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3a49002ba9b3497c9702666f85b59e13 : SUCCESS in 1h 04m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a7f929c47d06416c8a2cee557a63466a : SUCCESS in 2m 55s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5d5cc67f86af4270aec2213a40ce5d74 : SUCCESS in 4m 12s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/971cdb134abd47b488835edb11506305 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bd0277b7cb4bcfad6504ec64be26f7 : SUCCESS in 10m 00s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8b1bdfbe83ed44ed887a71b487f478c7 : SUCCESS in 54m 09s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/8ac21528c0bd4d368e544d232375d65b : SUCCESS in 13m 37s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/f2fb1e9e06f44946b30f74ab428dfce7 : SUCCESS in 34m 19s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/415e38265551424e928a1695dea495c1 : SUCCESS in 55m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b1a9b2b54592429c8cb81cb18343baec : SUCCESS in 1h 19m 34s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/3ad9d6d04d2a43f393b859ffb2eadd90 : FAILURE in 1h 07m 06s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/eea8b6ed901642d4830ae2bd5bc8792d : FAILURE in 51m 45s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/a783433f8b6b4f4e8e9546f13d6fb1f9 : FAILURE in 35m 58s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/42f31a1b55b64586abfb7878c4081426 : SUCCESS in 30m 19s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/edf59e45836b440882fc3bb5f40ed16d : FAILURE in 29m 28s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6e793bf8b58d4e25b948dfc60b90af09 : SUCCESS in 1h 07m 07s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4b83b88937d448e7a4ec1badbdda2e40 : FAILURE in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f89c94b53604f20f7f2817c537aedd5d379354b9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 13:05:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2486a4deebf2c55979a078bb7be7133a23892382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 14:46:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a2f8618a1234d96a198a561ce91f99c\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f6669fc2e3d2435cb005a8a98f69a713 : FAILURE in 49m 39s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3634285bc6e2438681cd08a1cd4fc5f5 : FAILURE in 44m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75bc2300b3a4458acc5c6f96e701a7a : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c98bfb7cbf745c88834e6c4d9a10c70 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85c4297a0ab547ef911862bff6f42f00 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cf5dcf1e7fd4e2896f8fb8150bb19de : SUCCESS in 3m 53s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ab76c4b861fc4f74b25012769836dcdf : SUCCESS in 53m 52s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9bfb69f34a504278b061a78050190b4d : SUCCESS in 4m 48s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/11525e2969a64f22bca25847d906eec0 : SUCCESS in 4m 29s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/23e72d378c6b4c0aa6f9056e59687642 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f1f8c83cb8425194691cca8ce40acd : SUCCESS in 9m 46s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a47e28cb592647f69717619bfc588a3a : SUCCESS in 1h 01m 31s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0fa2d211a36145be86f08a6c7dd84c8c : SUCCESS in 13m 50s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/957b7e0b5bbb4e6c81f327d4a53f6f08 : SUCCESS in 33m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e53ba5f76b9f4efb9f06f9ddac8904d2 : SUCCESS in 56m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0eed22d0180246e1bb7db68071913777 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1be3aeaf20394223bad8bbb9615b4dc1 : FAILURE in 1h 22m 30s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/1ef7a25c8152408f8df15f6c08606652 : FAILURE in 1h 00m 17s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/12eb9789dde0418d9e3e810341a9b871 : FAILURE in 33m 34s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/47474b0d5d944b0382e4fdf816befd59 : POST_FAILURE in 29m 04s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8d35ca2b748b4cb7ad5f4f0391ba9df1 : FAILURE in 29m 52s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/75ec5ba366ae435baa23fb9057f9f1d4 : SUCCESS in 1h 03m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c5bca68ce0fe4083878d7d6e2b6664e1 : FAILURE in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2364bf2dbe8b7bb13c0a004fac4a0f2287c98a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-24 15:58:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3cfef57c421c6c809d410ed5123d5a9baa01e5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 20:06:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09a96b83886e4b9b8230a1f0f06b5256\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f4107f0d2a174edbbd16ec766e7deee0 : SUCCESS in 30m 11s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2666abbd33b14aa880ebe3153ae4d701 : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ecb125003fe4c22a4a3bd5e934e7893 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf57de147ad4eb3bfede8c7d3a973b1 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79a5e90d0fa4492facba1b4dd51ca5c1 : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2803ea74e088496ea04ece3f48b0434a : SUCCESS in 1h 07m 08s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8a4b3707d201428d8fd36f42d52e70c0 : SUCCESS in 7m 47s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fd3ccf5987a845dea5d525a1469f473a : SUCCESS in 5m 39s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/fd618e7c7ce44c9cbc27b2cb9dca3fc1 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed3a6e1bb3a44d3b54e49db4c7db83b : SUCCESS in 8m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0a5735384358473ab0066d5317e7819b : SUCCESS in 1h 51m 31s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ef08075aab524648a315a733f25b0eaa : SUCCESS in 13m 07s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/ce040858b2cb4008984551d636a0092a : SUCCESS in 39m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5926af8d03184d18b41749f71704c14e : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f6d4f2d4ad9b444289669b2bba03a256 : SUCCESS in 1h 01m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/63402e0cc5f545308432e85d756d051d : SUCCESS in 1h 33m 02s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9f14327514e24b9ca986da9d2bfbdece : FAILURE in 57m 02s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7b76e7c251fa46f78f3815596291eb96 : FAILURE in 28m 57s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e5d8b2d9b6724b4aa7b942778a58bea2 : SUCCESS in 26m 30s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/14479729322a4cbc9a669615f37311e9 : FAILURE in 30m 01s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b5c19b99e501492ab26c46963b2cd96c : SUCCESS in 1h 07m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/636f2c985f2246b69d28f5ceb3189ce4 : FAILURE in 49m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"729f6db18cea00641012bcef2c0b494e2194aed7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-25 09:10:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5921d6706590af0cc86ce98a8195f9438fb62da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 09:11:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"801f5d5139b9916af4d9305c8980dc485f8cec9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 10:54:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8f7fc915588488eb6c03e48a0cf74c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2ff39f30f243dda12c6e56a22681d3 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c2c77cc0b924935848fc6dd46449094 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8367f7ff84434cd9abf0ef7c38c208f9 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fe98d7359e3c46fbb6820a65c8ee0ae0 : SUCCESS in 1h 12m 12s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7f1af318bab54d33b11721971ee50234 : SUCCESS in 3m 57s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/509f1de4f88a42e498434741f0f3791b : SUCCESS in 4m 26s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/6ed2203bc6e940c7bf92cc16033b17ce : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c08c44608044e14a486c88917f4a14a : SUCCESS in 10m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f4514c3be6834446b79ecf2de817a332 : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fd1160f2cc94fb9bf2f2b4aa85b17e6 : SUCCESS in 1h 05m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/361b7b07bda844a682413a6537ead3bd : SUCCESS in 1h 16m 09s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/7e3f73dd349447ea84d83ab286a3637f : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":4},{"id":"2dc5a400a4a808387e682ba028cd73ea74f484d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 10:54:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e01772fae445762c9006fb786469ed315bbea009","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 10:55:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9543c80c9806421f80f81cdfa6dfa899\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cedea6e1051426d8c3e5529b9393409 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e4a09ed8a469c88fb34f45d54e6e6b28b5231efa","revisions":{"6bb0f6acbce18cfcf64b15bf3ce7e3a5ad7cd835":{"kind":"REWORK","_number":1,"created":"2024-07-03 15:39:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/1"}}},"commit":{"parents":[{"commit":"5ea4ae1ceea80acd239dcc4edfa723270642379e","subject":"cloud: Switch remaining image functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5ea4ae1ceea80acd239dcc4edfa723270642379e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 16:07:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-03 12:44:00.000000000","tz":60},"subject":"cloud: Switch remaining block storage functions to proxy","message":"cloud: Switch remaining block storage functions to proxy\n\nThere are a couple of changes we want to make here:\n\n- None of the \u0027get\u0027 functions should allow specifying additional filters\n  as this is the responsibility of the \u0027search\u0027 functionality. We\n  deprecated these.\n- A couple of parameters are no longer used. These are deprecated for\n  removal.\n\nChange-Id: I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6bb0f6acbce18cfcf64b15bf3ce7e3a5ad7cd835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6bb0f6acbce18cfcf64b15bf3ce7e3a5ad7cd835"}]},"branch":"refs/heads/master"},"a4b20981992a5c423bcaf41db214b4bce3c9eae1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-09 10:02:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/2"}}},"commit":{"parents":[{"commit":"d1037ee64f6ad5f725aa8b26f28a507b864450d6","subject":"cloud: Switch remaining image functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d1037ee64f6ad5f725aa8b26f28a507b864450d6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 16:07:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-09 10:00:38.000000000","tz":60},"subject":"cloud: Switch remaining block storage functions to proxy","message":"cloud: Switch remaining block storage functions to proxy\n\nThere are a couple of changes we want to make here:\n\n- None of the \u0027get\u0027 functions should allow specifying additional filters\n  as this is the responsibility of the \u0027search\u0027 functionality. We\n  deprecated these.\n- A couple of parameters are no longer used. These are deprecated for\n  removal.\n\nChange-Id: I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a4b20981992a5c423bcaf41db214b4bce3c9eae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a4b20981992a5c423bcaf41db214b4bce3c9eae1"}]},"branch":"refs/heads/master"},"bf741f33644cdbd4b17a3b0144904847c830755b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-12 11:29:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/3"}}},"commit":{"parents":[{"commit":"3f4044f4d24fe0befa1aaa24cb85a47429943a65","subject":"cloud: Switch remaining image functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3f4044f4d24fe0befa1aaa24cb85a47429943a65"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 16:07:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-12 10:52:31.000000000","tz":60},"subject":"cloud: Switch remaining block storage functions to proxy","message":"cloud: Switch remaining block storage functions to proxy\n\nThere are a couple of changes we want to make here:\n\n- None of the \u0027get\u0027 functions should allow specifying additional filters\n  as this is the responsibility of the \u0027search\u0027 functionality. We\n  deprecated these.\n- A couple of parameters are no longer used. These are deprecated for\n  removal.\n\nChange-Id: I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf741f33644cdbd4b17a3b0144904847c830755b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf741f33644cdbd4b17a3b0144904847c830755b"}]},"branch":"refs/heads/master"},"e4a09ed8a469c88fb34f45d54e6e6b28b5231efa":{"kind":"REWORK","_number":4,"created":"2025-02-24 15:58:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/923423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/923423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/923423/4"}}},"commit":{"parents":[{"commit":"37bb57f4d54ba886c447b1ef8f4c86546b374929","subject":"cloud: Switch remaining image functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/37bb57f4d54ba886c447b1ef8f4c86546b374929"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-24 15:21:09.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-24 15:34:39.000000000","tz":0},"subject":"cloud: Switch remaining block storage functions to proxy","message":"cloud: Switch remaining block storage functions to proxy\n\nThere are a couple of changes we want to make here:\n\n- None of the \u0027get\u0027 functions should allow specifying additional filters\n  as this is the responsibility of the \u0027search\u0027 functionality. We\n  deprecated these.\n- A couple of parameters are no longer used. These are deprecated for\n  removal.\n\nChange-Id: I2c9a54f9dd7fa49f2422c5270bb860dbce2aac1e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e4a09ed8a469c88fb34f45d54e6e6b28b5231efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e4a09ed8a469c88fb34f45d54e6e6b28b5231efa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
