)]}'
{"id":"openstack%2Fopenstacksdk~923426","triplet_id":"openstack%2Fopenstacksdk~master~I921a2eea00efe393b611d4c5d97cbc18aad9fad1","project":"openstack/openstacksdk","branch":"master","topic":"use-proxy","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-26 18:19:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I921a2eea00efe393b611d4c5d97cbc18aad9fad1","subject":"cloud: Switch remaining compute functions to proxy","status":"MERGED","created":"2024-07-03 15:39:14.000000000","updated":"2025-02-26 20:47:06.000000000","submitted":"2025-02-26 20:45:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923426-use-proxy","meta_rev_id":"43c778faff7250788f6fba7ea640cc4c9002bb8c","_number":923426,"virtual_id_number":923426,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-26 20:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-26 18:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-26 18:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 17:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"537f60734ab8bd619e6ef7a10b4464db9e85c606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 15:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a0b54850d784f99cf922efcdb024fbe450b4a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:36:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61ef57998e2b4da6a2cd063ce1cf1bfb\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4fa7c9e4b3c94798802fe6e647bd986a : FAILURE in 42m 59s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c3d6ef55636e454d9c56c790d5a92326 : SUCCESS in 41m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73b397c4a1547779edf587dbf27f225 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcc1c4aa021c49c9ae99ab8888eea328 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5a4199ce0ac4e7e9eb809bb294abc98 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cc82a2bbf1843d9b6bd93e832b089a0 : SUCCESS in 6m 38s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0cce9e44840c49008d8bad35c9ae9f43 : FAILURE in 58m 58s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f8ec4c9118c8418bb872e3773fb176de : SUCCESS in 3m 38s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/da6701dc29dc4c4c8f528f114c6a9dbc : SUCCESS in 4m 34s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/b93e3aa208974d3fa1ee61460c2ad86c : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4132e406fae4b18ac98287d2455c9d3 : SUCCESS in 8m 52s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d9c7cefa059f481b9e3a4e65f7a5fd27 : SUCCESS in 1h 03m 11s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/624a8c7a08b6413ca9716a2c60b5348e : SUCCESS in 13m 25s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/40c8038430d640849c80a0c5536c2800 : FAILURE in 17m 45s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d9387f07b0f45538099fbdee3ede3f4 : FAILURE in 1h 06m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7a18d68bb20b49cf8bb81524b17941b0 : FAILURE in 1h 34m 18s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/6caa2276237e433e80ccddf6942e7f26 : FAILURE in 1h 31m 52s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e80bddba5647484690d0993032e532f3 : FAILURE in 59m 35s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/eae3f2163e7344829451c10a33db42e6 : SUCCESS in 43m 55s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e37e329f9a494f67a686eae390afe3bf : SUCCESS in 29m 59s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8175767aaa844f0683407f3c77b94226 : FAILURE in 38m 49s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c9483ee6132a4138825c285766b0cc1a : SUCCESS in 1h 11m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/734163270e5e4595be179c9ec5a693d8 : SUCCESS in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ba376f7042e02f1ed9e85496bd06836e6d3516c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-09 10:02:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d607ed45cdd4160bfd67a4224acca278511b3bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 11:33:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc3d450731b4d888c2015c6f3258612\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/467645ae825d498ca2e34309e94e789d : SUCCESS in 52m 11s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/913f2ae221514fd589ada3218286c13b : SUCCESS in 36m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cef6bcf0e924966926c8bc14486b7ae : FAILURE in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85205b9268944971b002f069bf83d219 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8affd1f397a4364aba707979f051ec3 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bebe8668f76d41c7b91890cb35dc755d : SUCCESS in 7m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/14abdc9eb6144548bf375586aa6d6299 : FAILURE in 1h 01m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1adb706bc6304603864fddcb356ec0f7 : SUCCESS in 3m 20s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f8f3f26580654a76a068b346d1620174 : SUCCESS in 4m 15s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/b24a23f5a4fa4c9c8e8f2ce3901824d8 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60095642f0f94944865582fdc794459f : SUCCESS in 8m 13s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/89cca3de2bb04b37b6f3e6d55dcd3246 : SUCCESS in 52m 15s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/61bbdd7855ba4c589df6bb58eba1e75a : SUCCESS in 10m 27s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/4585735d2c3743aa8d0b490267b0bf51 : SUCCESS in 36m 00s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41b85eb08c574fab927f8094e49b0aba : FAILURE in 1h 05m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/31c505d5b03846b29f73315c12d4445f : FAILURE in 1h 21m 11s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/be2952d176474a8ba86e85961b5b5b37 : FAILURE in 1h 24m 59s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c196eb63e57942e89e704a9de8f0c9c5 : FAILURE in 54m 43s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5d982475317f4a939777a226efff1e9f : SUCCESS in 33m 16s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/36c70f952430495cb2f65314da1c1b2f : SUCCESS in 29m 08s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c42687128a3c4a0985676ab71ca384a3 : FAILURE in 28m 04s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/493a7659c9554e5cb9aac0b6b334e8e3 : SUCCESS in 1h 08m 20s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/34c2e4ede85c4f2a9f01ce85c46ba1a0 : FAILURE in 43m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file openstack/cloud/_identity.py","accounts_in_message":[],"_revision_number":2},{"id":"d9780e450b4800d133c5f0265b247ea1a8f23432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 11:29:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d93bdc18ed49d2ab2487ba165cbf9ae8b66f250a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:12:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89fed56e93d8482b8315fe82f87d31b5\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/9c607a70019b48d2aa2f43d795c9f411 : FAILURE in 45m 01s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/467bd0ef39b64250bae386cfc93e900a : FAILURE in 43m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e450d2962b31434087a76b3ba799762c : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9fda9cccd1d447689e3737767bbab8f : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d36270b87e624c60a353e506c41526cf : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/334e7a88e29a435da2d76f52f45dee2f : SUCCESS in 4m 30s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/652d2df170494c8386a210d50041735c : FAILURE in 1h 01m 01s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/531b116c18224d6f8f14edb57dc63933 : SUCCESS in 3m 35s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ba45b28413b64d68bcff7a161db598ef : SUCCESS in 4m 01s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7c638abb08cf4eb3b6d201398cce90a8 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115deff0a586441499fb8c7d8ece4e62 : SUCCESS in 9m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fbe02baead0340f58964e7d5d5ffd8ca : SUCCESS in 47m 38s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f5ca08a7615f4e49b54198129437aebe : SUCCESS in 14m 31s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/628604d74a684a999acf2702aa844f2e : SUCCESS in 29m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db007f1b4b7c4f929032afccea1d18e0 : FAILURE in 47m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f0f18be14dba4e0386ff3d3d7e6f850a : FAILURE in 1h 21m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/526ce5ebaa1d48f8a05259cffcae0028 : FAILURE in 1h 36m 47s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a1cb75a1f50a450799cad9541570e4b8 : FAILURE in 57m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5f45f755dabf4349aa56b681ca19b32c : FAILURE in 35m 05s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/178766d686b1417ab5d0c4fd5a7a750c : SUCCESS in 27m 15s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/58d7803062ac45be92a19103b596ae23 : FAILURE in 38m 05s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f33952e6079f4e1e8df63cd19473b905 : SUCCESS in 1h 02m 29s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d4d3e3758fdc4004ac829a9cb2b25b5b : FAILURE in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe8f64f7e6cde4ab308b402573cd97300c4469ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-24 15:58:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"172cb3028d6d3e41f24e8f29fc548158c392f408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 18:32:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef93e353504f4df88af3370494f7b8f8\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/684eb7703c524ac1a8f9a77b8345ff79 : SUCCESS in 36m 05s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/10c77cea59f9400594e61c854e47c8f5 : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455c9bdc13c348f18f15f5f1a83893f2 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46434f6c09694e678d850d94b90793ef : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a25daa5a51c64e23acb2b6898ba0c215 : SUCCESS in 4m 18s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b8f0e1f9a04d4cdcba8d10d5c304740f : FAILURE in 53m 31s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5ce24a71a89d479db1a67cb690126298 : SUCCESS in 3m 33s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ecda6858bead455581d1997ec2a9307b : SUCCESS in 4m 45s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8d4a6af6a3994a9db206d97e56a84404 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d47d0e48d1a48b4812bf5ac9566a4e0 : SUCCESS in 7m 44s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9eeea3bc6dd24920b3b7c5bb2245102b : SUCCESS in 18m 18s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/33f2a12e813244f5a7250c7f8f6d8a8e : FAILURE in 7m 30s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/ea1210ebc23242dda63541abe9c37b54 : SKIPPED Skipped due to failed job nodepool-build-image-siblings (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18389a2002d2466ab5fd9a4bf5b9b1e5 : FAILURE in 1h 33m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/94db54fded5146c2bc0467114ae5c7cd : FAILURE in 1h 32m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/c3d6ea88232e4578b45e359dc43c863b : FAILURE in 1h 11m 26s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ee7f8b7590d145ae9e89538748467319 : FAILURE in 40m 16s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1726490f48f2451c80d63d714de9852f : FAILURE in 37m 30s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/28358c2754c64e93ad998b97c86542a8 : SUCCESS in 27m 18s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7a4292871c6045e8a239b3e6e4504342 : FAILURE in 28m 02s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d060600490bc4ba28588c199f46aecce : SUCCESS in 1h 39m 10s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a413ea33ad1343cf9b932d86cb5d6bb0 : FAILURE in 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4bc404bab48b6f1ef73a43f8fdbd67628befad69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-25 10:04:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"80da8435d8abd2effb476a7d12f84177299b6d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 11:49:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd8ed2187804281acc9530506ee2f56\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f6d013da0f1c4c3dad1e8a4962755daa : FAILURE in 27m 33s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0cbe1da670de4b30bd08db49de078742 : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb233eb614e94af5801841f260f7a69c : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29ce94a64db4e5eae70cc8292b7bf17 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bf39e3d1f2c43838c2838c40f66feac : SUCCESS in 3m 59s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3d1e0a88e862426baf57a075505e7eb3 : SUCCESS in 1h 05m 33s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4a3e7a22e5ca4c3b866e9d9582bb19b4 : SUCCESS in 3m 46s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/46158c7ee88a498c8ba474d0b67c26d1 : SUCCESS in 7m 24s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/054e51afb7de4277b57eb51ccbe2dcdf : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0755084dccca4d6b952ca1338b369027 : SUCCESS in 9m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2ae7e2fa56ca4bf6b1a06219d59e168a : SUCCESS in 1h 22m 16s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/c1d623ce8e64471192a8fe4f71912b88 : SUCCESS in 13m 05s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/614388ce613842e29889145dce9aad43 : SUCCESS in 45m 32s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/568a2ae7363c446db3fc21542b9c1bff : SUCCESS in 51m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/76db053dcdf34953941d7148962ebf45 : SUCCESS in 1h 21m 57s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/280e1f251c134158989d6ea76540c371 : SUCCESS in 1h 39m 02s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/76f7675b71c3454381f38a77be5d3bf5 : FAILURE in 55m 02s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/71acc986e7864d47a43939eee46472c1 : FAILURE in 36m 16s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/7d2e04b41e704ae3995426061ae4a86f : SUCCESS in 33m 32s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/120af66cada644498f664596f618dd47 : FAILURE in 38m 32s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/46027667586b4788ac6a5dfeeaca266b : SUCCESS in 1h 14m 33s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/843098211940404292944342e530adf9 : FAILURE in 46m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dd6260d517fae8560bd8a8239fa6a28378f975d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-26 18:19:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1c93006fdee90d8a3ef7d25035e8a453219c861c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:19:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1bfd92543c2a83ef0754aae4b82a6aba76699b45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:45:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544ca5176b424aab8b34e4aa08dc7235\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bc3c7a486342e7a742d768bdd14bd2 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86749c00d76c4a50a491e22f7be8ca4c : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b09498252104f7ba421978708f04855 : SUCCESS in 4m 30s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/891e8d52bbda4c3fb5041ef7cb1c6140 : SUCCESS in 1h 29m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/40031561fff548158bdd87595074c197 : SUCCESS in 3m 45s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e88afe5cd2df4ba7b956fe04ba04fb7c : SUCCESS in 4m 36s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/81781eee0a0d409d9d1897118eded247 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b2bbef7b8c45ab8f6d6b2e50d6e445 : SUCCESS in 8m 58s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e307029e79964b64adc065074301e919 : SUCCESS in 3m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a95d69a527d426a8aa38ad6b89a7df2 : SUCCESS in 1h 14m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b6dce615443148098726ec306aff7e6c : SUCCESS in 1h 33m 16s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1d54d89637624752a50fb091980500f3 : SUCCESS in 2h 11m 16s","accounts_in_message":[],"_revision_number":5},{"id":"f6cb7a9a8aa94f7a74ef8edb04bc9d08d480b8a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:45:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"43c778faff7250788f6fba7ea640cc4c9002bb8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:47:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de2c2c79179c4a3e950d9d53ca1c05b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c651aaeec64d1e8a3912e5b9047d0e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"76010b1fceaa8e034f5ac4b75c5aaa3b766cf74b","revisions":{"e50e85bdd52ebb02bb80f2a56dbbabcdf8c9d45e":{"kind":"REWORK","_number":1,"created":"2024-07-03 15:39:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/1"}}},"commit":{"parents":[{"commit":"7810fb293c063a3f537ec06b19ae8846a1bdbca3","subject":"cloud: Switch remaining identity functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7810fb293c063a3f537ec06b19ae8846a1bdbca3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-25 15:48:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-03 14:55:28.000000000","tz":60},"subject":"cloud: Switch remaining compute functions to proxy","message":"cloud: Switch remaining compute functions to proxy\n\nChange-Id: I921a2eea00efe393b611d4c5d97cbc18aad9fad1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e50e85bdd52ebb02bb80f2a56dbbabcdf8c9d45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e50e85bdd52ebb02bb80f2a56dbbabcdf8c9d45e"}]},"branch":"refs/heads/master"},"aff5409044d13e36f190278e72f511ad7db78e75":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-09 10:02:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/2"}}},"commit":{"parents":[{"commit":"d19abac3dfc5e30b95ef046ef136c7fa743b8d3a","subject":"cloud: Switch remaining identity functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d19abac3dfc5e30b95ef046ef136c7fa743b8d3a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-25 15:48:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-09 10:00:38.000000000","tz":60},"subject":"cloud: Switch remaining compute functions to proxy","message":"cloud: Switch remaining compute functions to proxy\n\nChange-Id: I921a2eea00efe393b611d4c5d97cbc18aad9fad1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/aff5409044d13e36f190278e72f511ad7db78e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/aff5409044d13e36f190278e72f511ad7db78e75"}]},"branch":"refs/heads/master"},"4fe6faac60e76f7f38fdf9f1a1ca9a232f31cf2b":{"kind":"REWORK","_number":3,"created":"2024-09-12 11:29:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/3"}}},"commit":{"parents":[{"commit":"2a81daae4be2310f0896dd68a81240503967230b","subject":"cloud: Switch remaining identity functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2a81daae4be2310f0896dd68a81240503967230b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-12 10:53:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-12 11:27:25.000000000","tz":60},"subject":"cloud: Switch remaining compute functions to proxy","message":"cloud: Switch remaining compute functions to proxy\n\nChange-Id: I921a2eea00efe393b611d4c5d97cbc18aad9fad1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4fe6faac60e76f7f38fdf9f1a1ca9a232f31cf2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4fe6faac60e76f7f38fdf9f1a1ca9a232f31cf2b"}]},"branch":"refs/heads/master"},"85169107bfa34017eec0282aeeafb1e45264359d":{"kind":"REWORK","_number":4,"created":"2025-02-24 15:58:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923426/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923426/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/4"}}},"commit":{"parents":[{"commit":"51806366c9644090f56139a882ebd0397368437f","subject":"cloud: Switch remaining identity functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/51806366c9644090f56139a882ebd0397368437f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-24 15:24:56.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-24 15:49:33.000000000","tz":0},"subject":"cloud: Switch remaining compute functions to proxy","message":"cloud: Switch remaining compute functions to proxy\n\nChange-Id: I921a2eea00efe393b611d4c5d97cbc18aad9fad1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/85169107bfa34017eec0282aeeafb1e45264359d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/85169107bfa34017eec0282aeeafb1e45264359d"}]},"branch":"refs/heads/master"},"76010b1fceaa8e034f5ac4b75c5aaa3b766cf74b":{"kind":"REWORK","_number":5,"created":"2025-02-25 10:04:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/923426/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/923426/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/923426/5"}}},"commit":{"parents":[{"commit":"51806366c9644090f56139a882ebd0397368437f","subject":"cloud: Switch remaining identity functions to proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/51806366c9644090f56139a882ebd0397368437f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-24 15:24:56.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-25 10:03:38.000000000","tz":0},"subject":"cloud: Switch remaining compute functions to proxy","message":"cloud: Switch remaining compute functions to proxy\n\nChange-Id: I921a2eea00efe393b611d4c5d97cbc18aad9fad1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76010b1fceaa8e034f5ac4b75c5aaa3b766cf74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76010b1fceaa8e034f5ac4b75c5aaa3b766cf74b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
