)]}'
{"id":"openstack%2Fopenstacksdk~923505","triplet_id":"openstack%2Fopenstacksdk~master~I3e2a4083df5d72db2f9afb5f3d2e322ce4963566","project":"openstack/openstacksdk","branch":"master","topic":"fix_add_ip_list","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2025-01-31 12:47:40.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-01-31 09:46:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-01-31 12:47:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e2a4083df5d72db2f9afb5f3d2e322ce4963566","subject":"cloud floating ip: fix add_ip_list() for single ip","status":"MERGED","created":"2024-07-04 12:19:04.000000000","updated":"2025-01-31 12:49:39.000000000","submitted":"2025-01-31 12:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923505-fix_add_ip_list","meta_rev_id":"6856e0dcd68c523534fd0a86d7f5a1702239a397","_number":923505,"virtual_id_number":923505,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-31 12:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-01-31 09:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-01-31 09:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-01-31 09:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 12:19:44.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-07-04 12:19:44.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-07-04 13:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-31 09:38:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"5e721743d06645e9f997f3425b6dc0736c700c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2024-07-04 12:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64e13258a145e564f70318c5ea9a5290ba4d0d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:56:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6afc6ba8fbe641948c6601f66025c4eb\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8ff0bdbcb2964995a68427fc979284e6 : SUCCESS in 44m 51s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/cf18a26611584b16958dab1b0d221fc9 : FAILURE in 39m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/454d18530efa48d0a17775eb3ee0e97a : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dfa71c66433473497c37a2aaeaf05ae : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1f67afd217c417c8945386b4c6fa885 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4bf6a5aa4934ac2aac4d111c78d5359 : SUCCESS in 7m 05s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/418a09820c3a4589962b2ce42389a532 : SUCCESS in 1h 08m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6f713e703ae441638f7e53f50048ff0d : SUCCESS in 4m 05s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e5ff64e640744722958cbb609dbdf28f : SUCCESS in 4m 31s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0af1c83a0b044280856e4a40e853b5d3 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e78c0c89f144cfa55300189775afc2 : SUCCESS in 8m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c443d1b9d462437bac5b3ebc0b9d4c5c : SUCCESS in 46m 04s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a04c33f89e9d44e4ba66610c9e3bb976 : SUCCESS in 13m 30s (non-voting)\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/af610ac35f974d6d87f515b3e78eb0d5 : FAILURE in 18m 51s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45724cfc3e494f139dc112e7e49153fd : SUCCESS in 58m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9ba30548feb84e84a7d06aff13d2d3b0 : SUCCESS in 1h 30m 37s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/622aedc233e14826a44758b1b343989f : SUCCESS in 1h 29m 43s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/31d72d730e8e48b2b844e55273b6bd61 : FAILURE in 55m 02s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/59c91399106b4754b66b47acb0b47421 : SUCCESS in 31m 48s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/411e98ec8488423d972fd6966a05d6bb : SUCCESS in 27m 20s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ff1eff3da9124060a5f44bbfccc9402b : FAILURE in 39m 33s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0945440498db473b85b39b04f8496ede : SUCCESS in 1h 10m 13s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5ede883f78a14f7cae3dae268086a6af : SUCCESS in 1h 09m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f71da0425e29e3732484426eb7994cf160b9fb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2025-01-29 14:24:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5bb431b257b899bfde5be8f295a59412784142d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 16:21:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58ee3b4d51a6433583baea28aa943732\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/fbfe493024bc44cfbc72db962956ff68 : SUCCESS in 52m 00s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/046b4f277a134976bd39b5c946782683 : SUCCESS in 38m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48472e8b196540ee954cd022eed21814 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90f19920322b4829aab3621e70be293b : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e829aa48a0d4b499e8944d07a6904f1 : SUCCESS in 4m 26s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ef7899a5642c47b0b4aa1a87b7516570 : SUCCESS in 1h 31m 11s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a9abdce1c2f64050bf73c789bc363cb7 : SUCCESS in 4m 11s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5207b2eaabf344488f4715136075389b : SUCCESS in 5m 17s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/b5aa869aa6114cdbadf7b2f3e6d4d530 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81812fa41cea4333a6fee531e9f914b3 : SUCCESS in 9m 00s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/71bd7e6a41c84023ac64583d38d41787 : SUCCESS in 19m 37s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/812712d1ead0437cbe7665ea2ce60184 : FAILURE in 10m 31s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/c41942294c2b445ab816cf9927b29a9a : SKIPPED Skipped due to failed job nodepool-build-image-siblings (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59fc6a824da481e9411759645655413 : SUCCESS in 1h 06m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ebdd036b571349158ea0ed43d5a07b6e : SUCCESS in 1h 33m 24s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/48c1783250724ded985f450737340039 : SUCCESS in 1h 51m 15s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/75fb27ee856b426d84a25085a6e094ca : FAILURE in 46m 08s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/8dcbb45c3bc1436fad92b36b55c0d599 : SUCCESS in 25m 45s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ab597f28375446f7999704df1a96a27e : SUCCESS in 17m 39s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5fe2df40d04a46af836bd481068e6f70 : FAILURE in 37m 22s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a97affb5e9314f8e939adf9391eb207a : SUCCESS in 1h 32m 29s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/bf9ffdc031ac4a0f8125936b98f62d79 : SUCCESS in 1h 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ac05022aae7c20685ca44a3b3e9236246bba099","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-31 09:38:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3030d3de6128f49ab8fe216f531365661e7f1bea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-31 09:46:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7102efec80853a1307cb22bc0390c16b9f54a5d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 09:47:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0d733a7f7d33eb569ebb0f855f769d6ae597209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 12:47:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/078ae72b36b444b8a8cf0342b9267317\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9925f17bcbc4da79b29364996d38656 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/699264aa0f7f45eca8cb9b9008b8ea56 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce45199ef5bd46cc805550abf67c8b35 : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f8bf6fc63f2f4f9889d4f8e33ea79dad : SUCCESS in 1h 47m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/87621fcc32fe4f958893f4f3b5daea0f : SUCCESS in 12m 15s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1fc653123ed7455aaefb91cfb09bcbf6 : SUCCESS in 15m 53s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/553974805bca4f239d4bc4c9dd45da56 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e198ce19f774205be056d3eb8675346 : SUCCESS in 10m 16s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d55e57a1bee8431080063d6769b12c44 : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/398138bda8434708aeb63b1a7d843309 : SUCCESS in 1h 12m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/878398bf778c43a19f377ec66e2352ca : SUCCESS in 1h 26m 56s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/155faab0ff404936bdad97f2997df74b : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":2},{"id":"177e7c10b0ed62eb94c66da95bbf4d7c3948b0b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 12:47:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6856e0dcd68c523534fd0a86d7f5a1702239a397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 12:49:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738e603e7908402aa96c2e2cf1667a86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a269329931ed45eabdf552ce85e54ba7 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e85eca80e2956737b823abe21bd090332faeb67","revisions":{"a85c63bd635ff61d5e9f4835f49694735b13ac12":{"kind":"REWORK","_number":1,"created":"2024-07-04 12:19:04.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/05/923505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/05/923505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/1"}}},"commit":{"parents":[{"commit":"3ee18eb1711c6e76188b58c0eb520443fd34750e","subject":"Run auto_allocated_topology functional tests in the own project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3ee18eb1711c6e76188b58c0eb520443fd34750e"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2024-07-04 12:16:49.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2024-07-04 12:16:49.000000000","tz":120},"subject":"cloud floating ip: fix add_ip_list() for single ip","message":"cloud floating ip: fix add_ip_list() for single ip\n\nFixes bug introduced by I44385c25fcb010f3e62b4098fd34ae3290292630.\n`!\u003d` should be changed to `is not`.\n\nChange-Id: I3e2a4083df5d72db2f9afb5f3d2e322ce4963566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a85c63bd635ff61d5e9f4835f49694735b13ac12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/a85c63bd635ff61d5e9f4835f49694735b13ac12"}]},"branch":"refs/heads/master"},"2e85eca80e2956737b823abe21bd090332faeb67":{"kind":"REWORK","_number":2,"created":"2025-01-29 14:24:41.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/05/923505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/05/923505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/05/923505/2"}}},"commit":{"parents":[{"commit":"c6459ae0b0983530d8db75178bbfe213c8022396","subject":"Merge \"identity: Add support for endpoint/project associations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c6459ae0b0983530d8db75178bbfe213c8022396"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2024-07-04 12:16:49.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2025-01-29 14:24:09.000000000","tz":60},"subject":"cloud floating ip: fix add_ip_list() for single ip","message":"cloud floating ip: fix add_ip_list() for single ip\n\nFixes bug introduced by I44385c25fcb010f3e62b4098fd34ae3290292630.\n`!\u003d` should be changed to `is not`.\n\nChange-Id: I3e2a4083df5d72db2f9afb5f3d2e322ce4963566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2e85eca80e2956737b823abe21bd090332faeb67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2e85eca80e2956737b823abe21bd090332faeb67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
