)]}'
{"id":"openstack%2Fopenstacksdk~925605","triplet_id":"openstack%2Fopenstacksdk~master~I2a48a3793ebd2db9dd9ca02b75b599c266ad1c24","project":"openstack/openstacksdk","branch":"master","topic":"bug/2075347","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-12 13:21:45.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-09-12 13:21:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a48a3793ebd2db9dd9ca02b75b599c266ad1c24","subject":"Fix parameter name clashing in resource lock query","status":"MERGED","created":"2024-08-02 15:35:34.000000000","updated":"2024-09-12 13:31:57.000000000","submitted":"2024-09-12 13:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925605-bug/2075347","meta_rev_id":"78d552386da88053a5457dd48547ea541e904ac0","_number":925605,"virtual_id_number":925605,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 13:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-09-12 10:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-09-12 10:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 17:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 19:49:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-08-13 19:49:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-09-12 10:51:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"33fb815e5816eca4a8b25d2a718834fee09a6d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-08-02 15:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdc4cee2ede5d1b1ab4bc79796ec3429e3c3a644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 17:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99d21301f13443659e34a46a9c5c2e15\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a77e2fd25b6b40f694b291f02841b535 : SUCCESS in 48m 00s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/346f3e72f0464cd3acba5cbc1c3e9a42 : SUCCESS in 47m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621ed5ed4ab741959e5cf1e66d0a159c : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a08cb23c6974d3287882c1e14ca57ba : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/976c4348136f4a4f85ca0ecba1f312b0 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f121c00b3b5f4186a3f24a994344c915 : SUCCESS in 7m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1c7222a5b714421fa270e510b1d4ec30 : SUCCESS in 1h 02m 36s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d3b54b0dd2d040e08853be0a9366f438 : SUCCESS in 5m 03s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/491fab33c9d04a33a4ba1d34bbd336bf : SUCCESS in 4m 23s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d0e15f95088e4b1a89c43b0a800f5657 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6c6aa05ea94bd2b484351bd72fa39f : SUCCESS in 8m 48s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fb1cb34710154de0a5e49dc99f07534a : SUCCESS in 50m 30s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/7ddd7f768bd74fa6bea7d5a939e26404 : SUCCESS in 12m 23s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/db5a3dd2e86d4b17bb08b8fdf71808d7 : SUCCESS in 34m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1fb1371527c4c4eb1982cd33f5af2c9 : SUCCESS in 59m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/705887813c354dce805bb9342091126a : SUCCESS in 1h 30m 22s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5d2a61cf4e5c41b88616f8a3e812c2d5 : SUCCESS in 1h 29m 32s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/eb958299bff8493cb50ae40796ab7484 : FAILURE in 55m 56s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c1a9bcda64db40c7820467b85d31898d : SUCCESS in 23m 26s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/75b1f261ab0046ee881b9623550e9b7c : POST_FAILURE in 27m 41s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ff392ea9f2a74cdc8673a44ab981c7e7 : FAILURE in 39m 41s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/49b1fbb048f0429e9e1baecc1e1bfe2c : SUCCESS in 1h 03m 23s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/507a32536f3e400587126cefb3c07894 : FAILURE in 32m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"848182b0e2a9f4f00185d2825a3fdbf3e9d75afb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 10:51:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c6d31f4bcb96049319c56616cb1ad0e0de99c23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:54:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b608c075837d95872639167e8d0673c1952284e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62143f3f894348d1a5f05cb2c26dc9f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d54b04a134a4d3fa4269c6233a31ff3 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40c5938ce6674b86b9485de0078202f8 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fe1d11fdb164399926ee6b259c4f6ad : SUCCESS in 4m 03s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6cd0e88e37ee4bf9b4665c75528eabe2 : SUCCESS in 48m 31s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a34da2587fec42de84b52e7a5006e7ae : SUCCESS in 3m 16s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/37597ea34ee54c4fad42c1e7b6dd0480 : SUCCESS in 4m 45s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/406e1a3d233c480fa21442df88c08c21 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a46dba5348f4293afdbdbc02aa0f050 : SUCCESS in 7m 52s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5541fec5def3462a9743c149020e3623 : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/264361eb11414c8f8d9d40bca06b004a : SUCCESS in 50m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a1959e3b9fba4007a87ab366a10d6209 : SUCCESS in 1h 23m 29s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a1ab791f510b4b469c614c82bd37202b : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f8bbdb9aea6fff46c3f75643279879097be4e576","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78d552386da88053a5457dd48547ea541e904ac0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d60101c7924958a960b0ef4a2ffa9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45efa359ebba45cf84a56094697b077d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef052e48807e0811859d8a5fc720846ea93d0f2c","revisions":{"ef052e48807e0811859d8a5fc720846ea93d0f2c":{"kind":"REWORK","_number":1,"created":"2024-08-02 15:35:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/05/925605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/05/925605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/925605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/925605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/05/925605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/05/925605/1"}}},"commit":{"parents":[{"commit":"939ec03ab389c7aa897328a54d9d8b036e8ac89e","subject":"Merge \"block_storage: Add support for project_id in Limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/939ec03ab389c7aa897328a54d9d8b036e8ac89e"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-08-02 15:19:31.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-08-02 15:35:30.000000000","tz":120},"subject":"Fix parameter name clashing in resource lock query","message":"Fix parameter name clashing in resource lock query\n\nTODO:\n* add test coverage\n* look at other methods taking resource_type as kwargs\n\nCloses-Bug: #2075347\nChange-Id: I2a48a3793ebd2db9dd9ca02b75b599c266ad1c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ef052e48807e0811859d8a5fc720846ea93d0f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ef052e48807e0811859d8a5fc720846ea93d0f2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
