)]}'
{"id":"openstack%2Fopenstacksdk~927854","triplet_id":"openstack%2Fopenstacksdk~master~I4d7b9a3c0f652f9a07fbf70df35e3c780462514c","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-12-17 17:49:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37209":{"account":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"last_update":"2024-12-17 19:50:14.000000000","reason":"Change was submitted"},"35119":{"account":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"last_update":"2024-12-17 19:50:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d7b9a3c0f652f9a07fbf70df35e3c780462514c","subject":"Add getting info about the Limit in a DNS SDK.","status":"MERGED","created":"2024-09-03 16:49:18.000000000","updated":"2024-12-17 19:51:44.000000000","submitted":"2024-12-17 19:50:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927854","meta_rev_id":"91ffc53e2feb1c1a11a920ace7607ceb56712821","_number":927854,"virtual_id_number":927854,"owner":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 19:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-12-17 17:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-12-17 17:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 18:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 09:26:23.000000000","updated_by":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"reviewer":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"state":"CC"},{"updated":"2024-09-12 09:10:47.000000000","updated_by":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"reviewer":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"state":"REVIEWER"},{"updated":"2024-09-12 09:37:49.000000000","updated_by":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b429086be51d00ed0eeba27a5ec6f5177a89f4fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-03 16:49:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4589031c40e7ed758a2cf1034383634b856603e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 18:49:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d083dafdc5d419480a918c38a22d574\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/079cb1a6bf5045f39b527245495ad36e : SUCCESS in 45m 46s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/dc192c608fbf4051890dfe0aa710e8ba : SUCCESS in 40m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbe709199e446ff84101e6f13915181 : FAILURE in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e86b40976f14923b71fb1b27f06cbb5 : FAILURE in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28e2aa17bd26426ca46ef33590f838c3 : FAILURE in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fe720b2d3324bd58b13301d45fcd088 : FAILURE in 5m 29s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/636ad8eec80b4509828223027a98ac23 : SUCCESS in 1h 00m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4562cc9af4474068a09f4d9050c143ae : SUCCESS in 3m 07s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e1ed911970284c7bb2ba9d36c6f514e2 : SUCCESS in 4m 05s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7333983c154e4b579f67fc9f4cd2f048 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca839ffd3da94b90940b73216e7bb5ee : SUCCESS in 8m 38s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4d97f2f3132148cbb4ac55a4e1918c31 : SUCCESS in 46m 30s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/efaf844f50914b63bad1166cedfe0134 : SUCCESS in 11m 06s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/0ad3ffc6204747ffb156cab1853154c9 : SUCCESS in 30m 59s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7274e2ddcf15415798859b9f02647978 : SUCCESS in 1h 02m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/54e34ad3c01e452d820608e4725c76dd : SUCCESS in 1h 19m 43s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8cd702bbffd848f5a52fc53f7a684db1 : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/198f79916134436298391f214ae564fd : FAILURE in 48m 26s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/45fa253dd04a4684a4154844522d8b9e : SUCCESS in 38m 35s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f596f99771ae438cb890f93ce42c3d9e : SUCCESS in 21m 58s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/fdf8ac646e5e4005a1b5d3d5b9d8193c : FAILURE in 42m 00s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4e6a777297a04a7b82c7900f1cf39650 : SUCCESS in 1h 00m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cd72a5cdc2824be6b30e57b63a2d4f06 : FAILURE in 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b179fe8f9010ae8cfbb130c75a13394a152c9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-05 11:58:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"120687b9472e796889b608ce40f18b943fcf3f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-05 12:13:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b0f13eb239357e3474019ce3b68b8e95a5dbf53e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-05 12:35:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29040636d4a658f34c8ccb166df6366f2fb182b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-05 13:17:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"84e13f779ef22bc07861992a7da6f8064597adda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 14:44:25.000000000","message":"Patch Set 5: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00c59f95c4c64aaca944da87b1f0d3ad\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1b909e0a34984c8d8dc902323db21cd0 : FAILURE in 45m 15s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b6584c18cd0f41d28c2294f59e572ebe : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f827a8959d47fa89a432f6aa817748 : FAILURE in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7948c531a59b4573b61f287303ea129c : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9660ffa9faab41bb9d314115170f2762 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5f84e0749714e6eb9d25935365ebe44 : SUCCESS in 5m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cac651b1d1cd4aa4860a51ea47eed0f1 : SUCCESS in 1h 04m 56s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b22341c88d764560a348b7184f772c25 : SUCCESS in 3m 23s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9c94c9dd17ab4398ad539812791f2c66 : SUCCESS in 6m 14s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/89a2a3601db045ec81357cd53f632f24 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9027437016f2463db413c987a919e49d : SUCCESS in 8m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/55fe7a8a22e141c3b73fd5fa748801ee : SUCCESS in 57m 08s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/4e55100b3ae94713877086685ffeab3d : SUCCESS in 13m 25s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/457c52faa86941eab05f593a1dc09c38 : SUCCESS in 37m 30s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61fc1939d978422a90e93aa64fba8df7 : SUCCESS in 59m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/87eaa9fe4de04f43b17ba659c6eae2de : SUCCESS in 1h 25m 56s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/af3bbaec26f045ca88e6ba3fd29f35b3 : SUCCESS in 1h 20m 18s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e586f759f5b44db5982e36981de4b83d : FAILURE in 52m 47s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b02b442f438b4b85b8a396b474dc1417 : FAILURE in 25m 39s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9077a90c1f0c4dbea358b43666087a67 : SUCCESS in 17m 55s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3cc4d37d69fa4be6b28858999c8f132b : FAILURE in 40m 34s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/46a43bc4417c487598754e8253058764 : FAILURE in 56m 21s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/87ed810cec1549fe97846568e0eb2a80 : FAILURE in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0509167c4e99b25a7c624b88eea44105c9ffae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-06 05:26:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6e43c6a7b2d4f58b8c2b834a50b5d16e9e10c662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-06 05:40:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5f74a33f525a5d45b79682b7d17a587d1bc00177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-06 06:16:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7b12c54062e00628a5287aa8935fbde6716748a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-06 06:27:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7e841f697f47847d49f23e4682ff3026fedd7895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-06 06:42:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ec8add4a979349fbd2f0b692ed0c44671a646579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 08:23:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65ce2edd06404f40b42a3d5f05c104ff\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f6997643007a4f10ac0f580176cbacd4 : FAILURE in 49m 46s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/26af7b23c83744579821cf126d8cbce5 : SUCCESS in 37m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cbd36f24e54029aebd2e61012590d4 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ac87a19f2794b6f96e15623cd6b9e75 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e2a194b0b5c4f12812e8baf744b9caa : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32b26970232245beaa8cfc7fc716cfc8 : SUCCESS in 4m 34s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d020d477e3c345a99ebde6578c63c0cc : SUCCESS in 49m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/21e278aca40645e39fdb687c3632dd06 : SUCCESS in 3m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/53e00a304f584cdb820b1dff21732afa : SUCCESS in 3m 54s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/e4399843d54248a4804b212f59bdee3e : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24225234fa847ecb4963be01d2a15ec : SUCCESS in 7m 42s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f2bce14ce0ef4b97b6c7a0653f61d307 : SUCCESS in 53m 12s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/6ccb1864a3a949e5a0dad708d8c09648 : SUCCESS in 13m 15s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/599de0bcfe114db184df338b90a95ecc : SUCCESS in 35m 37s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/502401517e4d4e06b78cf3c504491975 : SUCCESS in 57m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/38db7ada77c24413bcd835b66d9a0ff2 : FAILURE in 1h 31m 50s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9ba71ff003b04ff2b78fe3f29d503b87 : SUCCESS in 1h 30m 22s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e712751b172f4edaa2d3284c84940753 : FAILURE in 56m 04s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f90b06573ab5465d8087db2f3319dabb : FAILURE in 26m 49s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/63f2f08038b040e591b1a9d6a2c77701 : SUCCESS in 39m 46s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0c554d21cd86454b82e4ba86a149778c : FAILURE in 41m 29s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1fcd6864707549bda9264f25939e05f6 : SUCCESS in 1h 09m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/681e8321a08444bf85faaabd887ae125 : FAILURE in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f2468afe7d92105a8d38581058b0e61cc7b67c0a","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-08 07:17:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5e05f2f94d674ec674f30c87c95edc399d4e408f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-08 08:34:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd802746b8c743c1a345c9a0b6ec4e8a\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a1dca4a57d8e4f62a8353f73c724c35c : FAILURE in 33m 23s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f06d961ade1740d0a06ce10d98a264da : SUCCESS in 31m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2938eeaf52654ef48663a0a3c7e2f820 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a64926fd384e8cbb0fd3edfe688e17 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1cd46ea68bfd401598f1be56e2e22f12 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/424ba231da8e4759a19c2ae0151f23cb : SUCCESS in 4m 19s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1fe776686c1348948f46b453a83ee6cc : SUCCESS in 40m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d090aaeb016549e289d56e153a74ab6d : SUCCESS in 2m 57s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e86ec17707a84ff8bffd26eb66e6270f : SUCCESS in 4m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/b7d59cdc85b149609a5ef89197de9a70 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060b61fe45dd45bba8e89a0ab8e8f707 : SUCCESS in 8m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/952abd14f863458f93d075f916f22459 : SUCCESS in 42m 04s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/50fe0941a5274d2a9c4c71c11823a933 : SUCCESS in 15m 08s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/b529aaa1cb3c424ba66f9226c013f1ed : SUCCESS in 19m 55s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b399363b777249c38ac0f27a25795934 : SUCCESS in 1h 00m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4e1ca0cca6534b1c8b46897edee7b025 : SUCCESS in 1h 10m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f7d4ab4b545d40749895aa488a8a8503 : SUCCESS in 1h 15m 59s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ea5b800708d04b83974813d311be5a99 : FAILURE in 57m 48s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ae46dafab7e2465b813e29f93f9a05fb : FAILURE in 24m 26s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8be23c79faaf4ecc82d2afefb1f7e909 : SUCCESS in 30m 45s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/10db59e9c41f457cb5b0fa858b89fc5e : FAILURE in 22m 14s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e9029fe067d04473a6ed99cb17be50b3 : SUCCESS in 47m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c2dd202d95254e5089e881e80aac43da : FAILURE in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"14e38e2f336e4109a80f0e45d891d42399649957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-09 05:36:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"0015276056000c27ee7dd952af4e96a25d0cb57f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-09 05:48:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"410945230917e40d14da7f8323ec4228c7cc3c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-09 06:29:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a55a90182fb62adf27959f09f06a8bc16102bc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 08:02:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a07a5a350b4d7291895b6b4a91eb6f\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2b5edd4fa0b247f0a7c77d3339babd25 : FAILURE in 45m 32s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ffa7392991db42278507d96da19d1aa4 : SUCCESS in 27m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bff14d7dee74211ba3043b7e2560b6c : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62eac9167f254d2c84d0bb1d5d306b5c : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae3e067141f54bbb8e1810ecadf6a6cf : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dde7603e95094b2984d41a430bf0ed77 : SUCCESS in 6m 21s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/aabb6e0c5f854f65bf865b7720f47c3e : SUCCESS in 48m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1a609e370d74d9ca48bfaacd95a3112 : SUCCESS in 5m 56s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/98a2f9cdbdce492e940bf544fb464a38 : SUCCESS in 4m 21s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/1d9274fb7a624c2d84c08dc9bf30d73b : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2af050475a4b2a938fd732b6cb016a : SUCCESS in 8m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/da380b7f34e74594b68df43ebd056e92 : SUCCESS in 55m 53s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/33ea959bfa3d41989671672c22638c21 : SUCCESS in 12m 36s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/cfcec00e9e794ba2a0a91fb7db2605c1 : SUCCESS in 31m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71a5d1216f114a21aadb381bc24fd47b : SUCCESS in 48m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/59d61f6166134b17965196a085ab7847 : SUCCESS in 1h 21m 38s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/c6d0dd6070e4472fb37e3f284ec80b6d : SUCCESS in 1h 31m 43s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/95669d805bcf47569c08b6cc6f861117 : FAILURE in 53m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ed73ba53143946e99693bac1846e6a69 : FAILURE in 24m 10s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a30e425f25ab4578adf62224d3790700 : SUCCESS in 25m 09s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a11dd44ed24b4136b3f6f87605a73c90 : FAILURE in 33m 26s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c328076a51034f56ae02dc320537e606 : SUCCESS in 59m 15s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/efdb7c376d974ad2b28c176ce5cf8fb9 : FAILURE in 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3be350fa57523778dfba22ce897b61d7675be3ec","author":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"date":"2024-09-12 09:10:47.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"462e4ddff6e50afbf0b82efaf57f1c2b95749c26","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 09:48:15.000000000","message":"Patch Set 13: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6c73e9e5fb1b92c0bf205bb6484da1a82f0d9a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-22 07:16:11.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"4250a2fa677774e087c9a3f7a63b82b6d81eabc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-09-22 07:27:09.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"cd3422a46636a6be056983c8a16a8515143facf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 08:20:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2450bc1d20264fbc8015071321286b7c\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2ef6e208f20d41fea092b1f24ba7ec94 : SUCCESS in 46m 32s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/787536cd94534d6a904a83ea5596f56a : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af806e3211634014902a72dfb741c6ff : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f4b3804b4cd4b57a2e3361e5dbe1e94 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c611941d22e4410b7f2dabad4fa7c52 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba93c234aace4d20aaf96cda1101645a : SUCCESS in 4m 26s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7280ca7cd2b44670961ce957a64b3dc7 : SUCCESS in 30m 34s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9fe7f05cb7a34dd492b89cfc93c936c2 : SUCCESS in 3m 42s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/698a64ff6b3148fda0261354f74a4ee7 : SUCCESS in 4m 48s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/1fd007a2d43545c9a92c6335bdf6b919 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6d991a42124059b525d08292163133 : SUCCESS in 8m 49s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/dee989c36b564478ada429a16b135e1c : SUCCESS in 39m 58s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0f8d5ff5827f4a8b9dd1ad50ca05654e : SUCCESS in 11m 44s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/206628699a79449290a12f1b6c868516 : FAILURE in 24m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0521f7923de64607923db02bfb5b373e : SUCCESS in 29m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/24e8aec523dd470c83266a69f3973332 : SUCCESS in 17m 20s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a6cce07f4a8f4432bfe5ca26d1c64d0a : SUCCESS in 36m 48s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4710c27da73748ea9e144465e0ba9f11 : SUCCESS in 31m 05s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/c4e95ec6a81747cfa9bafe821930029f : SUCCESS in 14m 19s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/cd316af465ba44e3ae19d3e883b5b41a : SUCCESS in 33m 11s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ab04beabf60740b4829ed75408a2849a : SUCCESS in 40m 06s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e6edf5a6e76143c9b22ecd99f6d80671 : SUCCESS in 48m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/65f6b2c849274cdb9e9d8ccb9a180f4c : FAILURE in 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8eaa2ce6462ddf4204a6ac10fe51e5f75f8e9c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-10-05 07:19:12.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"139962176a39e3d5df54fed4e73857a04e568bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 08:44:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5494d219540547ab86120fdf7873013a\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7b7edf24d3824011a070c9f85371551d : SUCCESS in 35m 12s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6916177189554e1d90c4e9a201ecf44a : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61cb317045504bb7a404056dd10726d2 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e29931b9d69d436b87706603c120844c : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa31fd6dea9b4126a06b3291a834dd16 : SUCCESS in 3m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/945a9dbc5d2244dca8dd46da34e58daa : SUCCESS in 1h 00m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f3cfbdb1694e4f0d9d667cbbcd6ded9f : SUCCESS in 4m 27s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5db3d7b8c7f949cc87e12a05ab148b57 : SUCCESS in 5m 15s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/1a40544548f14d79b3d42a4bc5d3adbc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35420e33c414ba2bebfec25e5b9d3dd : SUCCESS in 9m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8eba8f6f024c4a5790d03b0e0dfab368 : SUCCESS in 30m 42s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/ef1e86b954d04612931f642d6a010827 : SUCCESS in 8m 38s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/a07629a15e2443dc94659298befd8c34 : SUCCESS in 18m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d44052f79e73438880328de18091cc01 : SUCCESS in 1h 00m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/073f94ae973a423da71a2cdb6614d1ad : FAILURE in 59m 07s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/69aa12e0577248d1adc8713f995884b5 : SUCCESS in 1h 18m 45s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/482ef1d28b0d47b7b35910968f765aa1 : FAILURE in 54m 22s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1150713f32a343ada6c47dce5faf06ec : FAILURE in 23m 11s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/dd02b9b977da40cdb325d077dd2bfe54 : POST_FAILURE in 25m 02s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/2b4c891876e64d79b5c0ecc6a8a58f0d : FAILURE in 23m 32s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/146db460b15e47b48cab882b64776493 : SUCCESS in 1h 13m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d4323c053b2844578fab88710c21665b : FAILURE in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"61a1db00c07c033d61624c12532dedcf10a4f0be","author":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"date":"2024-10-05 09:10:34.000000000","message":"Patch Set 16:\n\n(8 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9a0e5a2954e59f2f6b7a12c697b19be4a44a3b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 10:38:02.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb06f3986f59450fac753a174f1f384e\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1e774b6c8f5e4575810f978b00ee0af7 : SUCCESS in 38m 22s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/39494c7164b543a29a108f83c420c3da : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7af43d60504fb0bf6701bf83f620ea : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5736255a8fb4d62871beed8f2c2ab16 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5cc9eb3150f41efb8b1b041e919efaf : SUCCESS in 3m 58s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e3559b9979654601a9dcc1fe21db1c1c : SUCCESS in 1h 04m 26s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a44c74a2730b4087990025aa503fccba : SUCCESS in 3m 21s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/42771fd416e84498ab1605f72c2e0c70 : SUCCESS in 4m 19s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7702240c6c6f430cafad310da81ea32b : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b20f0c9b7f647faa2ee695fd4cf896a : SUCCESS in 12m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0a7b9d4879be4582ac2af18d14795310 : SUCCESS in 46m 17s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/44c8d238fe5d4e9f9743aa7e5015a5e6 : SUCCESS in 12m 25s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/63d83627303c4179b6e02bcaa67663af : SUCCESS in 28m 39s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5655bc9b381b4eee9b07f7f4b6b5d1f1 : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bdf6d581a7f8496cb897f1124dc550e0 : SUCCESS in 1h 25m 54s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d13e2efc60b04940850d7aef0c4a9f44 : SUCCESS in 1h 24m 50s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/f293c059e88540229dbb23d36a323676 : FAILURE in 50m 57s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/02f29e9817e14246b964af4af8b709e9 : FAILURE in 37m 27s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1c73bfe87ae542bea32f599d5da3cd59 : POST_FAILURE in 14m 39s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/87363de12e7343a1ad815daf842f06d2 : FAILURE in 34m 02s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/10f2711f7e384ff8b24e68bdd40d8a74 : SUCCESS in 58m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0fb3478467a048db9817852098110256 : FAILURE in 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"db6382844978c8f553bd97fe4992b2907dfbf352","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-17 17:49:14.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1872da6a97e27dfddd8ddd2efaf1ee8432fb1114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:50:20.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"ed4da26c19d31ccb8e0f7bcda7aef50acc1edcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 19:50:14.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f51fc5bfe1494d952c8bb0b0d54a17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47dc8a8f15a04814b424df484541a121 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f77e2256f54440988c08d2d6ccca7f7f : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e2937751e9a4716934e642e177c6ef1 : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e55912dabd3c4d1f853b7005d71a6dc6 : SUCCESS in 1h 04m 22s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4c2f765585824690912834677b6aca89 : SUCCESS in 3m 25s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fbbdee0ff13548298b225ec7c7b7bb41 : SUCCESS in 5m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/3406949127f94fccaf8675b7e0234d89 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198e49e55904407681242ecf529004e9 : SUCCESS in 11m 54s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/98186e1d99234a21a60c5474ea8d7cb2 : SUCCESS in 3m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e38b787d19d54f0e9dea667dcf19d2f0 : SUCCESS in 48m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c734b987251f4db1aa382988d08660b7 : SUCCESS in 1h 09m 53s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/7e659e155f2848b69f179f457fb88bad : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":16},{"id":"e2a45baf1c0c3a12103bd2041b26c431d88c3730","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 19:50:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"91ffc53e2feb1c1a11a920ace7607ceb56712821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 19:51:44.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc08cc1be188469aa8db33bbdf331bec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e3a150abfd43cbbfa9e29ba60c5e64 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"e53b1b7606b178d7d66ba0d46357a1c061ef8bb4","revisions":{"e7dc50a1925aaa86aab548c83cc1d8a9ced54ff8":{"kind":"REWORK","_number":1,"created":"2024-09-03 16:49:18.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/1"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e7dc50a1925aaa86aab548c83cc1d8a9ced54ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e7dc50a1925aaa86aab548c83cc1d8a9ced54ff8"}]},"branch":"refs/heads/master"},"57d18c2f2ee1634646b3a6b98b36ed0000fc441d":{"kind":"REWORK","_number":2,"created":"2024-09-05 11:58:24.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/2"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-05 11:57:48.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/57d18c2f2ee1634646b3a6b98b36ed0000fc441d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/57d18c2f2ee1634646b3a6b98b36ed0000fc441d"}]},"branch":"refs/heads/master"},"39367faa580d3dfdeb1d4aa0d1784cc5247c2298":{"kind":"REWORK","_number":3,"created":"2024-09-05 12:13:21.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/3"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-05 12:12:40.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/39367faa580d3dfdeb1d4aa0d1784cc5247c2298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/39367faa580d3dfdeb1d4aa0d1784cc5247c2298"}]},"branch":"refs/heads/master"},"131b39155afb28b4bc0bac702a05bbec02e2d4b6":{"kind":"REWORK","_number":4,"created":"2024-09-05 12:35:08.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/4"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-05 12:34:41.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/131b39155afb28b4bc0bac702a05bbec02e2d4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/131b39155afb28b4bc0bac702a05bbec02e2d4b6"}]},"branch":"refs/heads/master"},"358780d6b67223a82e8d3e2cc48fe7ac07d15f63":{"kind":"REWORK","_number":5,"created":"2024-09-05 13:17:29.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/5"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-05 13:16:53.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/358780d6b67223a82e8d3e2cc48fe7ac07d15f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/358780d6b67223a82e8d3e2cc48fe7ac07d15f63"}]},"branch":"refs/heads/master"},"df4aaf8a274fd193676c66a08e47c6de0d85c7bc":{"kind":"REWORK","_number":6,"created":"2024-09-06 05:26:01.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/6"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-06 05:25:32.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/df4aaf8a274fd193676c66a08e47c6de0d85c7bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/df4aaf8a274fd193676c66a08e47c6de0d85c7bc"}]},"branch":"refs/heads/master"},"69902a1ce4ef281046d88ae6467786d48cef4e44":{"kind":"REWORK","_number":7,"created":"2024-09-06 05:40:35.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/7"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-06 05:40:03.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/69902a1ce4ef281046d88ae6467786d48cef4e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/69902a1ce4ef281046d88ae6467786d48cef4e44"}]},"branch":"refs/heads/master"},"b462195c136965edca07660fae45307a92f03834":{"kind":"REWORK","_number":8,"created":"2024-09-06 06:16:16.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/8"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-06 06:14:54.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b462195c136965edca07660fae45307a92f03834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b462195c136965edca07660fae45307a92f03834"}]},"branch":"refs/heads/master"},"12b1c2f39ba78182b26bedd2ecc8eeeaecbc4c35":{"kind":"REWORK","_number":9,"created":"2024-09-06 06:27:26.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/9"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-06 06:26:55.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/12b1c2f39ba78182b26bedd2ecc8eeeaecbc4c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/12b1c2f39ba78182b26bedd2ecc8eeeaecbc4c35"}]},"branch":"refs/heads/master"},"d2b2213807003b910ccae47f4353e8cbb00c5173":{"kind":"REWORK","_number":10,"created":"2024-09-06 06:42:06.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/10"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-06 06:41:33.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d2b2213807003b910ccae47f4353e8cbb00c5173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d2b2213807003b910ccae47f4353e8cbb00c5173"}]},"branch":"refs/heads/master"},"4ba6887a3cac63a83725833058fdbe7c8fcfb8a2":{"kind":"REWORK","_number":11,"created":"2024-09-09 05:36:14.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/11"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-09 05:35:39.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4ba6887a3cac63a83725833058fdbe7c8fcfb8a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4ba6887a3cac63a83725833058fdbe7c8fcfb8a2"}]},"branch":"refs/heads/master"},"f4983f5549724b6d9abc6fa82afbd888e26c5748":{"kind":"REWORK","_number":12,"created":"2024-09-09 05:48:54.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/12"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-09 05:48:19.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f4983f5549724b6d9abc6fa82afbd888e26c5748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f4983f5549724b6d9abc6fa82afbd888e26c5748"}]},"branch":"refs/heads/master"},"2568b17573c3f1fb77f1b11317e74410f771bc75":{"kind":"REWORK","_number":13,"created":"2024-09-09 06:29:31.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/13"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-09 06:29:01.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2568b17573c3f1fb77f1b11317e74410f771bc75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2568b17573c3f1fb77f1b11317e74410f771bc75"}]},"branch":"refs/heads/master"},"f3f6f4e8158fdc3656f48a8d69525d8442c3f111":{"kind":"REWORK","_number":14,"created":"2024-09-22 07:16:11.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/14"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-22 07:15:20.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f3f6f4e8158fdc3656f48a8d69525d8442c3f111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f3f6f4e8158fdc3656f48a8d69525d8442c3f111"}]},"branch":"refs/heads/master"},"d05d405167a22054b082d363bc59642da91152bb":{"kind":"REWORK","_number":15,"created":"2024-09-22 07:27:09.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/15"}}},"commit":{"parents":[{"commit":"b65b7d4d3ee12578abfa1c461e4a66a8a7230c97","subject":"Merge \"Drop support for Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b65b7d4d3ee12578abfa1c461e4a66a8a7230c97"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-22 07:25:58.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d05d405167a22054b082d363bc59642da91152bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d05d405167a22054b082d363bc59642da91152bb"}]},"branch":"refs/heads/master"},"e53b1b7606b178d7d66ba0d46357a1c061ef8bb4":{"kind":"REWORK","_number":16,"created":"2024-10-05 07:19:12.000000000","uploader":{"_account_id":37209,"name":"Junwoo Park","display_name":"Junwoo Park","email":"junjunjun815@gmail.com","username":"JOOONU"},"ref":"refs/changes/54/927854/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/54/927854/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/54/927854/16"}}},"commit":{"parents":[{"commit":"c2470a21de72969812642ed4acf1c7c49b45dd6e","subject":"tox: Fix functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c2470a21de72969812642ed4acf1c7c49b45dd6e"}]}],"author":{"name":"Junwoo Park","email":"junjunjun815@gmail.com","date":"2024-09-03 16:48:25.000000000","tz":540},"committer":{"name":"JOONU","email":"junjunjun815@gmail.com","date":"2024-10-05 07:19:09.000000000","tz":540},"subject":"Add getting info about the Limit in a DNS SDK.","message":"Add getting info about the Limit in a DNS SDK.\n\nI added this because showing the info lists for a limit as specified in the DNS API documentation is not implemented in the SDK.\n\nChange-Id: I4d7b9a3c0f652f9a07fbf70df35e3c780462514c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e53b1b7606b178d7d66ba0d46357a1c061ef8bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e53b1b7606b178d7d66ba0d46357a1c061ef8bb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
